)]}'
{"id":"openstack%2Fheat~915949","triplet_id":"openstack%2Fheat~stable%2F2023.2~I8f6d3ed0aadd00e76710308fec7d3a1b1d5bb0e9","project":"openstack/heat","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"29474":{"account":{"_account_id":29474,"name":"Gheorghita BUTNARU","email":"gbutnaru@tuiasi.ro","username":"gbutnaru"},"last_update":"2025-04-30 15:40:44.000000000","reason":"Change was abandoned"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-04-17 01:23:10.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I8f6d3ed0aadd00e76710308fec7d3a1b1d5bb0e9","subject":"Remove logic for cloud-init \u003c 0.6.0","status":"ABANDONED","created":"2024-04-16 10:12:51.000000000","updated":"2025-04-30 15:40:44.000000000","total_comment_count":6,"unresolved_comment_count":2,"has_review_started":true,"cherry_pick_of_change":907010,"cherry_pick_of_patch_set":2,"meta_rev_id":"adc65070e26d6e8f37caa389f9fcc3ec2dc0ed1a","_number":915949,"virtual_id_number":915949,"owner":{"_account_id":29474,"name":"Gheorghita BUTNARU","email":"gbutnaru@tuiasi.ro","username":"gbutnaru"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-04-16 11:41:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-04-17 01:23:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-16 10:12:51.000000000","updated_by":{"_account_id":29474,"name":"Gheorghita BUTNARU","email":"gbutnaru@tuiasi.ro","username":"gbutnaru"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2024-04-16 10:14:38.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-04-16 10:27:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-16 11:41:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-30 15:40:44.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"bad9a64864c3f3f98d74f3a7594c7c1afa49c971","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29474,"name":"Gheorghita BUTNARU","email":"gbutnaru@tuiasi.ro","username":"gbutnaru"},"date":"2024-04-16 10:12:51.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"329c78ff584490a574701b8f7ac8db154646d6a1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-16 10:14:38.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8ab0e4208d7ae7cc853157a1ce7327d315fd3f20","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 10:27:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/954ad36c259e4a9890eed1ff1cf0ce0d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eb5c169b728b489fa015387b8ee2672d : SUCCESS in 11m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6616fd0c27784646bf2ec203fa87621f : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/239c57b1ccf94eaea439bb016cc7e23f : SUCCESS in 12m 22s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d61cb7d2ed8e4814b5510807c73cbf16 : SUCCESS in 7m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"03848bf1220e6a7de487516661a3d3d817d5992a","author":{"_account_id":29474,"name":"Gheorghita BUTNARU","email":"gbutnaru@tuiasi.ro","username":"gbutnaru"},"date":"2024-04-16 11:01:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c6b74d6f30a0c201596ae76f00a94fe994338370","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 11:41:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/88d7b083a4a64569b4b6ea010f055a0e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f54dbfbb1a914881aa2b8cd59b783d12 : SUCCESS in 11m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00ce1a7520844fdda8b99f279a21bcc8 : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5317ed8e4d13473995af7ad04a30b291 : SUCCESS in 8m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c78fdbbeaabe4d65be2c3df7d4df1f4c : SUCCESS in 8m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3656811237434da39f006d996f29244e : SUCCESS in 8m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5c3a548b62844392bc21b2ff17a056f3 : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de824466626449cbaee57ba5354431a9 : SUCCESS in 11m 33s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/bfc50ef0e04e40708156cab6aba3c16e : SUCCESS in 1h 15m 40s\n- heat-functional https://zuul.opendev.org/t/openstack/build/bc381ffc893c420e9b730203029a48cc : SUCCESS in 1h 04m 08s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/0679a59385734732b7b66c21c07d557c : FAILURE in 1h 17m 58s\n- heat-functional-centos-9-stream https://zuul.opendev.org/t/openstack/build/a558eb537d794570ac241ef7e4faac42 : SUCCESS in 1h 06m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"33629063649e02e3a0b2e926770bfc45a4121c9d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-17 01:23:10.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"856f3317e06ac3f52a36ee795adeabb699e7c1b8","author":{"_account_id":29474,"name":"Gheorghita BUTNARU","email":"gbutnaru@tuiasi.ro","username":"gbutnaru"},"date":"2024-04-17 07:53:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5eb6633f44985f6819609d352cba2db69531e67a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-17 11:12:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"96cea90c30f56358e29db34be4dee5b3849c0f7f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-17 11:13:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"adc65070e26d6e8f37caa389f9fcc3ec2dc0ed1a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-04-30 15:40:44.000000000","message":"Abandoned\n\nstable/2023.2 branch of openstack/heat transitioned to End of Life and is about to be deleted. To be able to do that, all open patches need to be abandoned.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9ee4b2fb95234320ac67bf998a95232700fa3092","revisions":{"9ee4b2fb95234320ac67bf998a95232700fa3092":{"kind":"REWORK","_number":1,"created":"2024-04-16 10:12:51.000000000","uploader":{"_account_id":29474,"name":"Gheorghita BUTNARU","email":"gbutnaru@tuiasi.ro","username":"gbutnaru"},"ref":"refs/changes/49/915949/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/49/915949/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/49/915949/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/49/915949/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/49/915949/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/49/915949/1"}}},"commit":{"parents":[{"commit":"4ca77ff93a6e5e058c52585197fbc7008718be70","subject":"Fix update of OS::Heat::UpdateWaitConditionHandle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/4ca77ff93a6e5e058c52585197fbc7008718be70"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-28 14:55:33.000000000","tz":540},"committer":{"name":"Gheorghita BUTNARU","email":"gbutnaru@tuiasi.ro","date":"2024-04-16 10:12:51.000000000","tz":0},"subject":"Remove logic for cloud-init \u003c 0.6.0","message":"Remove logic for cloud-init \u003c 0.6.0\n\ncloud-init was released on 2011-01-27. It may be reasonable to assume\nthat no one is using older versions.\n\nChange-Id: I8f6d3ed0aadd00e76710308fec7d3a1b1d5bb0e9\n(cherry picked from commit 19bcbc0390f24b0e9977a01d08380bf5a4fd39a4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/9ee4b2fb95234320ac67bf998a95232700fa3092"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/9ee4b2fb95234320ac67bf998a95232700fa3092"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
