)]}'
{"id":"openstack%2Fheat~919782","triplet_id":"openstack%2Fheat~stable%2F2024.1~I1e9c6b951c564b8a038737a1072a6df73aa925a6","project":"openstack/heat","branch":"stable/2024.1","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-05-16 04:02:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1e9c6b951c564b8a038737a1072a6df73aa925a6","subject":"Remove SQLAlchemy tips jobs","status":"MERGED","created":"2024-05-16 00:04:22.000000000","updated":"2024-05-16 04:07:23.000000000","submitted":"2024-05-16 04:02:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"919782-sqlalchemy-20","meta_rev_id":"e95a2915e549b2524add3fbeca030518ed5d63ae","_number":919782,"virtual_id_number":919782,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-16 04:02:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-05-16 00:05:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-05-16 00:05:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-16 00:04:22.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2024-05-16 00:13:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-16 01:56:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0e5013bb401f8ea34f16c4ec12dc7cacfd2d1480","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-16 00:04:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3bcdd0017d17c34ec6edf682ba7df40ef9de9802","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-16 00:05:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"35aa7047fc616a7f45b9e4a5b5c67e8d8a32bef9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 00:13:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee0f084ce540497bb5e8b2d794b2dd9b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bf5db00488724c09a988627e7f82b662 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1d72e282f2834ce39a47314fadb5d2e5 : SUCCESS in 8m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"30ea6a2f899c0f7c5a408adfe9a09f633a2d972c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 01:56:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4342c198f75d4fb59d9e5e555efa1972\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9d4a93385464622aebcfc893d9f216a : SUCCESS in 9m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6583f0cedec4224ab838109ef9c1ea5 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05a78d7a15d846a9877b8f9041674a8c : SUCCESS in 8m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b9648334626e4d949f339f2e65d0e57c : SUCCESS in 9m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc855544dacc48208dd958b2694447fc : SUCCESS in 10m 06s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/d98178f7a11d4f94ad6be176791fed23 : SUCCESS in 1h 45m 57s\n- grenade-heat-multinode-skip-level https://zuul.opendev.org/t/openstack/build/27d8e61214934b898ca30498a25f0c9a : SUCCESS in 1h 39m 54s\n- heat-functional https://zuul.opendev.org/t/openstack/build/37b2fce315f0440bb565c218b57ec28c : SUCCESS in 54m 54s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/3c4b0f60b0ab4d52a81aab4d87aaa242 : SUCCESS in 52m 41s\n- heat-functional-centos-9-stream https://zuul.opendev.org/t/openstack/build/d7025ccb6ea248ac90252222f62e86f0 : SUCCESS in 1h 05m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c46b67ae2ea2ba95e3a01fc3e73933de46d061fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 01:56:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b3ec9de5c8468ef9e699243000ace9be22aceaae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 04:02:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8158b3dc040d480ab5491f70d8670c0e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc4815846fd34abfb24bc875c11299a9 : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95e924a1ebdb4e9fa8ce2dc4fb20c643 : SUCCESS in 8m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b895d4bc8a734e648d7096dc1b2d8172 : SUCCESS in 10m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bee31f0a862a431e90dbf8701a29afdd : SUCCESS in 13m 26s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/7171119aec4b49abbe47a24f20ac44b5 : SUCCESS in 1h 47m 34s\n- grenade-heat-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b1e69e3b1dd74c65897472f48371bba7 : SUCCESS in 1h 39m 46s\n- heat-functional https://zuul.opendev.org/t/openstack/build/e074e350e21a475cb8c86d9819558f95 : SUCCESS in 1h 20m 15s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/ee81b4eb6ce149d5ba16fad67b8d0d23 : SUCCESS in 1h 09m 55s","accounts_in_message":[],"_revision_number":1},{"id":"2c025c0466e22729c457b8d167e996987b7b3789","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 04:02:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e95a2915e549b2524add3fbeca030518ed5d63ae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 04:07:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/160e9ee192324768b4224c0cc7da202d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1efca33b4b9e42f3ab2d90a91ac1b4a5 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3024a5f36ee1a671b530918554725f795fed695f","revisions":{"3024a5f36ee1a671b530918554725f795fed695f":{"kind":"REWORK","_number":1,"created":"2024-05-16 00:04:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/82/919782/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/82/919782/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/82/919782/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/82/919782/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/82/919782/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/82/919782/1"}}},"commit":{"parents":[{"commit":"23f0dc5e12eae5657ef4afa284a6308399cc4d35","subject":"Remove unused heat-functional-non-apache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/23f0dc5e12eae5657ef4afa284a6308399cc4d35"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-04-12 10:50:01.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-16 00:04:02.000000000","tz":540},"subject":"Remove SQLAlchemy tips jobs","message":"Remove SQLAlchemy tips jobs\n\nThe most recent SQLAlchemy and Alembic versions are now in\nupper-constraints. As a result, this job has served its purpose and can\nbe removed. For more information, see [1].\n\n[1] https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/RBHXHTO3GUOOXVSZXD4C2O3TKDOH2QSC/\n\nConflicts:\n\t.zuul.yaml\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I1e9c6b951c564b8a038737a1072a6df73aa925a6\n(cherry picked from commit 915ba0ceded80b0d9d12dd9100718da2f90cc7cc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/3024a5f36ee1a671b530918554725f795fed695f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/3024a5f36ee1a671b530918554725f795fed695f"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
