)]}'
{"id":"openstack%2Fheat~953724","triplet_id":"openstack%2Fheat~stable%2F2025.1~I91457df91bb2e6f6710af6f4fc19b11527d8e4c3","project":"openstack/heat","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-07-01 07:23:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I91457df91bb2e6f6710af6f4fc19b11527d8e4c3","subject":"Deprecate console scripts for api services","status":"MERGED","created":"2025-06-30 03:36:32.000000000","updated":"2025-07-01 07:24:14.000000000","submitted":"2025-07-01 07:23:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953724","cherry_pick_of_change":947668,"cherry_pick_of_patch_set":2,"meta_rev_id":"f2f075b452b62dffce51e3ac7cc6b409ef54af55","_number":953724,"virtual_id_number":953724,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-01 07:23:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":2,"date":"2025-06-30 04:05:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":1,"date":"2025-07-01 05:31:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-30 03:53:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-30 04:05:20.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"},{"updated":"2025-06-30 05:03:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"896c707343981d8e3300f46b19fecd41e7e71c99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-30 03:36:32.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c3c4245ba22137fd5415cab63b3449c1ce09fdd4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 03:53:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31e88819f4db48eabcd6e11c41201a72\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ff6c4b84edbb47ef8cbcc7d607420b69 : SUCCESS in 15m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f5183d93ba3d4643a33f64fb03550263 : SUCCESS in 13m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"64d3676ef95e43408157e2d3861ef44b16be2767","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2025-06-30 04:05:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"de07088863836d31ac1d6d311ebbd3cb6fb2c673","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 05:03:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa7ad193b7734e6f804322e6cfdb50d3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/90cae351573c4ad3a6981c6fa26e2afd : SUCCESS in 12m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/288fbe3eea5a4fe0bd4db3951437f5b5 : SUCCESS in 6m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b176dbd1923746d280226d036a37bc45 : SUCCESS in 9m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bc38dbf88de94f03a92cbedb2a2a0461 : SUCCESS in 9m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c41800133fd4bb9913a0bf7db6ded59 : SUCCESS in 14m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c467bb54cd0546719848305293d4d68b : SUCCESS in 10m 04s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/0ca07f3464f64be58da2a07ae098db09 : SUCCESS in 1h 25m 26s\n- grenade-heat-multinode-skip-level https://zuul.opendev.org/t/openstack/build/007ad54c609c41b7ae66b8af5922f1b2 : FAILURE in 1h 10m 28s\n- heat-functional https://zuul.opendev.org/t/openstack/build/126fc443fe184b4c947d7a7e0c4dd446 : SUCCESS in 1h 06m 20s\n- heat-functional-jammy https://zuul.opendev.org/t/openstack/build/a90828aa0e7049d6822c416f111a3eb3 : SUCCESS in 1h 07m 29s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/4c801a0cf8374baea14fc967ca1b254e : SUCCESS in 59m 47s\n- heat-functional-centos-9-stream https://zuul.opendev.org/t/openstack/build/e28416e078cf463baf11af413ff70b2a : FAILURE in 1h 19m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0812b8c16622d5945bab5c55fc96c4025025cfe1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-30 05:29:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2d0e9388be6a90fb4e81d6d9402aca61cec9ad4c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 05:45:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5dad93b6fed34526b556a67f88bcb8f0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0bb08df858554682850b83073c506f0c : SUCCESS in 13m 15s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a6e5c0371112463896101aa1f175410f : SUCCESS in 12m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"674d6444d6f065a213a4d7970c9cbb845086ba2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 07:11:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/98a8d26bad99470db3b92d173ed4f0a5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/90465e20be7a4beebb4b81f344838427 : SUCCESS in 9m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d70848928a2a467292ca8ca60c8e1df6 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c06b3da7137f49989d5fa91b4fff5707 : SUCCESS in 7m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/92b87be39bab4b68b385176c069ab007 : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88591ad0a9d94f38a9a5be26c78f3e8e : SUCCESS in 13m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29183bd504e244b6a9ec3f800aa0c213 : SUCCESS in 6m 58s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/979659ee9990476bb79b99ced0df0713 : SUCCESS in 1h 36m 31s\n- grenade-heat-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3da85390a2d347218b4809773edeb5d0 : FAILURE in 1h 20m 04s\n- heat-functional https://zuul.opendev.org/t/openstack/build/bb2632462ce04afca1ea2b03cb63de1e : SUCCESS in 1h 32m 58s\n- heat-functional-jammy https://zuul.opendev.org/t/openstack/build/531a9d72e7be477c87a81815d2f52f97 : SUCCESS in 1h 14m 51s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/adbda8baeba9460da6785b59921e8a35 : SUCCESS in 1h 12m 53s\n- heat-functional-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ab2e7ee0f2c4c2f868710a316aef75a : SUCCESS in 1h 11m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"66cf509253e7be25718239ef736acd871abc4a1f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-30 08:13:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a3264c36393768470f9160fe709d5f74d3d3367c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 08:27:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1159c402a9ad4b6e808b1f67856e1f36\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5ba11c8bc2b54234b216d8f60a6c0b17 : SUCCESS in 12m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/90e20e1a280e45b8a45d1fd8fb1cb61b : SUCCESS in 7m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1306fa50a69ac1fe613655bb4c43d85a09581040","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 09:45:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4f7525b22acd41ebade85bc834166d81\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47a4662b955f4ac2b41655e884c361dc : SUCCESS in 8m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c91409ca1d9e406287b20329bb2c1fc1 : SUCCESS in 2m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf48b419aa53419d9534f92d9581c0d3 : SUCCESS in 4m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e29ad19fc8f543658867ec823aaf9239 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fac4d04d22c4076aaee388f6647aa8f : SUCCESS in 10m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ecc505756e4949d3a235954e3c307e63 : SUCCESS in 5m 14s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/043e3780292442c9acf83f75eadaf5f9 : SUCCESS in 1h 31m 09s\n- grenade-heat-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1b976d85b4bd4632963b22e65cf8b126 : FAILURE in 1h 14m 28s\n- heat-functional https://zuul.opendev.org/t/openstack/build/b0cd18492f29490a9d4c6e282c3ef81e : SUCCESS in 1h 09m 46s\n- heat-functional-jammy https://zuul.opendev.org/t/openstack/build/903857877baa4a34bcfcf963b5fb06dd : FAILURE in 1h 25m 30s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/e7f0f713e0aa498ab61ac2d223e474a8 : SUCCESS in 1h 02m 33s\n- heat-functional-centos-9-stream https://zuul.opendev.org/t/openstack/build/bf61c36eda5640c39c23219501a1202b : FAILURE in 1h 04m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"62884066e0cf52c950bccbcfef65b1f9db523e0b","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2025-06-30 12:03:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"af736c23985b108e013b65e92e2985b8c74c4853","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 12:30:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71afc099d0104734b245b2ad56fc0a45\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1ed08b8c373d4fe8a77945335a10a3a5 : SUCCESS in 12m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ea12447a5c374531acad5f4a2be1cbfd : SUCCESS in 9m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9a0fc4de1bdc1091f60799b74cf7d809ca00f08f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 13:48:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb8e79523c2e41c38b4336bf4c78706f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/662f48b5e5da45f1b33009fbba42428c : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4514f86cc01a4b78963cb6f15db2c895 : SUCCESS in 2m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d8e32ae23cd94fd4b419491aeaf40dce : SUCCESS in 4m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2023afc96dd3462f8e6b98334118e356 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/964a88b8a56c4af49b380feefd93a770 : SUCCESS in 13m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c26e02503934949ae1dbc4069ba24c8 : SUCCESS in 4m 41s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/57309873e37e458fb58c0302cc932a87 : SUCCESS in 1h 37m 55s\n- grenade-heat-multinode-skip-level https://zuul.opendev.org/t/openstack/build/dbd7774e87484956b037746c8581670d : SUCCESS in 1h 34m 11s\n- heat-functional https://zuul.opendev.org/t/openstack/build/1fcf2375900c4b778d15eb62dd9ac1a9 : FAILURE in 1h 08m 49s\n- heat-functional-jammy https://zuul.opendev.org/t/openstack/build/78cb616cd4b147a8b823fd1d9134a839 : FAILURE in 1h 07m 54s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/24ee525a99694b0db9072741591e55ea : SUCCESS in 1h 14m 32s\n- heat-functional-centos-9-stream https://zuul.opendev.org/t/openstack/build/3fc8b35709d5427b9e95b1965dbe19b5 : FAILURE in 1h 09m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d6bf3314c913c7c3b5886d3c69a4c3cacbabb54c","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2025-07-01 03:42:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6f697236b0212a17da262b54f479aaeaa99a4e31","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 04:25:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cd2028816ff4389b766c8c89aa72ead\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f46c31ff227d4da794fc839fa8a36181 : SUCCESS in 17m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d4426cdb9afe436dbd20d1bc9c600dae : SUCCESS in 15m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e0c1e23682dc942776583a97a09777001307a47d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 05:16:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72d40ae0f33641b29eba27e41cc2f2a5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9cc3bf8d3fa44648353374553f89bc6 : SUCCESS in 9m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/585c160207cc4287852a684fc9e503d9 : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/493b1c17b1714e50842d617920ca41a5 : SUCCESS in 6m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/34f3d0797f4446b583991281efb497b7 : SUCCESS in 9m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0815c71f2c1b43c7926822d3d3631225 : SUCCESS in 12m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81230e74252f45ee9496320baec365ac : SUCCESS in 9m 07s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/e3164142842e421891b4e83da54771df : SUCCESS in 1h 29m 24s\n- grenade-heat-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4d06c4eb434d45ba8ba4a867f0af9650 : SUCCESS in 1h 24m 54s\n- heat-functional https://zuul.opendev.org/t/openstack/build/88ffe136a5304cb188f93b1a7eec5f5d : SUCCESS in 1h 16m 22s\n- heat-functional-jammy https://zuul.opendev.org/t/openstack/build/73695489d6fe45689c109dad8ac4f82e : SUCCESS in 1h 10m 06s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/b8fb13a117b24e91ba56b9a088f35457 : SUCCESS in 1h 10m 17s\n- heat-functional-centos-9-stream https://zuul.opendev.org/t/openstack/build/05bdcf00c7d94ee282e47e14b6c79a1c : FAILURE in 1h 18m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"58e333200ca8b4609c0fdb4a8bf9f902a969188f","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2025-07-01 05:31:59.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1009bc35a8430e7ee0c69557885f362afff7f728","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 05:32:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bda851959391110974e5d5ed7b301753fdc2af27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 07:23:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c60931e9408149d2bd29b2c732f9e5f1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1071ac56981499ebd63cf3dd0db8db9 : SUCCESS in 6m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a93fc25ecbb46f38968bbb679f60a78 : SUCCESS in 4m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a3125134038045f0b4ad70b97861c8fb : SUCCESS in 7m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cc9aa65255749cf91fc1b7230cb373e : SUCCESS in 13m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59bf6f3cbd0b40b0949d7f15c88935c7 : SUCCESS in 8m 29s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/50a990d5767d4500867774fb850a8ad4 : SUCCESS in 1h 38m 54s\n- grenade-heat-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e747a7297c454aa2ba5866ff448abd2e : SUCCESS in 1h 36m 35s\n- heat-functional https://zuul.opendev.org/t/openstack/build/04ec2d4e8d5c43d8a27b06d7ca7151da : SUCCESS in 1h 11m 15s\n- heat-functional-jammy https://zuul.opendev.org/t/openstack/build/d0e779421e7c4edfa428bd4f1e3214b0 : SUCCESS in 1h 09m 20s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/a4cd547b6f7743c792b2677605278595 : SUCCESS in 45m 20s","accounts_in_message":[],"_revision_number":1},{"id":"e18d0696d1fe7b4387711e4240695e4b26624bc3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 07:23:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f2f075b452b62dffce51e3ac7cc6b409ef54af55","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 07:24:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a34299ea1fac477fa0ed87378a0e42aa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ded5f392ba34eb3ac0472254a9bcb08 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2ed307f851a4f2c90ee7fde056aebe6 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"52b1a1710db07ead05249077cd307f268964d76c","revisions":{"52b1a1710db07ead05249077cd307f268964d76c":{"kind":"REWORK","_number":1,"created":"2025-06-30 03:36:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/24/953724/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/24/953724/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/24/953724/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/24/953724/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/24/953724/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/24/953724/1"}}},"commit":{"parents":[{"commit":"1ba2f085324fde3a66024e48659ecdf9fb151659","subject":"Set proper role name for the grenade job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/1ba2f085324fde3a66024e48659ecdf9fb151659"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-18 14:42:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-30 03:36:32.000000000","tz":0},"subject":"Deprecate console scripts for api services","message":"Deprecate console scripts for api services\n\nCurrent standalone api services heavily depend on WSGI server\nimplementation of eventlet and we aim to completely get rid of it as\npart of the effort to get rid of eventlet.\n\nIt is now officially recommended to use the wsgi scripts or\napplications and run these by a 3rd party mechanism such as uwsgi or\nhttpd + mod_wsgi.\n\nChange-Id: I91457df91bb2e6f6710af6f4fc19b11527d8e4c3\n(cherry picked from commit 9d6b898b52073260f291a6f224eba96a9ca84fd4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/52b1a1710db07ead05249077cd307f268964d76c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/52b1a1710db07ead05249077cd307f268964d76c"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
