)]}'
{"id":"openstack%2Fheat~968110","triplet_id":"openstack%2Fheat~stable%2F2025.1~Ib41f76c1ba56005b6c4233424cca5768657a7686","project":"openstack/heat","branch":"stable/2025.1","hashtags":[],"change_id":"Ib41f76c1ba56005b6c4233424cca5768657a7686","subject":"Use authentication with keystone ec2token api","status":"MERGED","created":"2025-11-22 09:53:56.000000000","updated":"2025-11-22 16:20:42.000000000","submitted":"2025-11-22 16:19:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"968110","cherry_pick_of_change":968109,"cherry_pick_of_patch_set":1,"meta_rev_id":"1898a6b93a130e86f0bd6d2f1df571ec0e1dd1b7","_number":968110,"virtual_id_number":968110,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-22 16:19:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2025-11-22 14:37:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2025-11-22 14:37:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"}],"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-22 09:53:56.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"state":"CC"},{"updated":"2025-11-22 10:09:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-22 11:36:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"64915464688ff0ff0bbdaf5b11a7a8bf9ef6aaf8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-22 09:53:56.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.2.","accounts_in_message":[],"_revision_number":1},{"id":"eb80f9c77ce9c4ee19388e239dd219e218f7e6e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-22 10:09:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9f94e55f4f84193a7b641ad3ced1a31\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/df8e0c79cb8d4ea5a842227621fca833 : SUCCESS in 13m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/680b47bae9be4d9a86b1eb1e05c6efc0 : SUCCESS in 13m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e193e87dbab867a114bb0ea973ac893044389159","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-22 11:36:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5606d785d5bf45b5a3efc135271e7b01\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f07405f71634e12940a5fa54f7786e6 : SUCCESS in 10m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e232cf126ecf4c35a722340fd340a28f : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e874bc6331eb4b709467d003b2c17d02 : SUCCESS in 4m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8ba9b27470e34378a4d35e0bad847d5d : SUCCESS in 7m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d15293d3eeab4ef88acef76f6fc653bd : SUCCESS in 10m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d17b40af1fc64e999ffcf5fe0b8c4310 : SUCCESS in 10m 42s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/b9319c855b05455c9326b548344c4a06 : SUCCESS in 1h 01m 19s\n- grenade-heat-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f5b3e5c5c1c44d6e886d9fb0d0abe3e3 : SUCCESS in 1h 36m 51s\n- heat-functional https://zuul.opendev.org/t/openstack/build/05389a1146db489982f690b3f2ad25ff : SUCCESS in 50m 16s\n- heat-functional-jammy https://zuul.opendev.org/t/openstack/build/0ac85f95c5d644bfa2d475b26e6d0556 : SUCCESS in 55m 29s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/5fb91415e38c43a5a8564e24b37dd2d7 : SUCCESS in 1h 17m 08s\n- heat-functional-centos-9-stream https://zuul.opendev.org/t/openstack/build/b89a06567ed54d669aebb4826e9bd0c7 : FAILURE in 18m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"065854857dba59a753828536013691c96519a892","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-22 14:37:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"310a53311d773119119011d1960acba54639f874","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-22 14:38:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0b19e7405c37dbc72d0e8eb51c9b2bc450fbeb70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-22 16:19:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/730f14e139f74a609b63f5029ee6291e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58ab0be57bb542438d1f2296743e5049 : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/11c16809696c4645b3ae9431d325d4e2 : SUCCESS in 7m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6522e47387754d51b158a3549a7011b1 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f84e91a35ce147c3a7fef565fef0eeeb : SUCCESS in 13m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0062d1cbbf8c443a87f942297bdb5df8 : SUCCESS in 7m 40s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/f3dbad4dbb5a4c5eb6a65d569647f57d : SUCCESS in 1h 00m 06s\n- grenade-heat-multinode-skip-level https://zuul.opendev.org/t/openstack/build/80c248183f9b422090e315534946c12b : SUCCESS in 1h 36m 35s\n- heat-functional https://zuul.opendev.org/t/openstack/build/54a3ae0953c1449cacc87617c28e6aa3 : SUCCESS in 48m 58s\n- heat-functional-jammy https://zuul.opendev.org/t/openstack/build/89b8c37bf02b4cd3af2e1430d5c4b295 : SUCCESS in 1h 22m 59s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/d01ad17b25cb4ea9a9096ff116b3aa45 : SUCCESS in 42m 45s","accounts_in_message":[],"_revision_number":1},{"id":"44f97a14b9e1ffcd36b8eeca21f658aa96ede9b1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-22 16:19:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1898a6b93a130e86f0bd6d2f1df571ec0e1dd1b7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-22 16:20:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/205408525fc846f0aad477449b734e80\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1078bc48289438eba2b68ff7ca458d0 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46fd8edaac1545af9ffb12fc07f7e7fe : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"006357801e76a5a30da592becd08f90c1c4deb44","revisions":{"006357801e76a5a30da592becd08f90c1c4deb44":{"kind":"REWORK","_number":1,"created":"2025-11-22 09:53:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/10/968110/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/10/968110/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/10/968110/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/10/968110/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/10/968110/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/10/968110/1"}}},"commit":{"parents":[{"commit":"46cbda240558f5b8a6e27ef6111b7b26e4b65ff1","subject":"Merge \"Fix stack update when reauthentication_auth_method\u003dtrusts\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/46cbda240558f5b8a6e27ef6111b7b26e4b65ff1"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2025-11-04 16:41:47.000000000","tz":0},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-22 09:53:56.000000000","tz":0},"subject":"Use authentication with keystone ec2token api","message":"Use authentication with keystone ec2token api\n\nAs a result of fixing OSSA-2025-002, ec2tokens API in Keystone\nnow by default requires authentication.\n\nThe config section `[ec2authtoken]` is now expected to have\nauth information for Heat to be able to use this API.\n\nIn multicloud configuration keystone auth credentials are\nrequired for each cloud. These can be configured using the new clouds\noption and the ``[ec2authtoken.{cloud}]`` sections.\n\nNOTE:\n Disables test_software_config.ParallelDeploymentsTest in\n grenade. Should follow up to re-enable once fix has been\n backported.\n\nConflicts:\n\t.pre-commit-config.yaml\n\nRelated-Bug: #2119646\nChange-Id: Ib41f76c1ba56005b6c4233424cca5768657a7686\nCo-Authored-By: Adrian Jarvis \u003cadrian.jarvis@catalystcloud.nz\u003e\nCo-Authored-By: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nSigned-off-by: Pavlo Shchelokovskyy \u003cshchelokovskyy@gmail.com\u003e\n(cherry picked from commit 0242c28d5f004591331bf7d005d7b305e555f398)\n(cherry picked from commit 76bf87551fdf3e1b9c82ae66c9ff22f82f70c01c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/006357801e76a5a30da592becd08f90c1c4deb44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/006357801e76a5a30da592becd08f90c1c4deb44"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
