)]}'
{"id":"openstack%2Fheat~968133","triplet_id":"openstack%2Fheat~stable%2F2025.1~I4d698e02c42e199f83ba337974d4b7e9c5725a9f","project":"openstack/heat","branch":"stable/2025.1","hashtags":[],"change_id":"I4d698e02c42e199f83ba337974d4b7e9c5725a9f","subject":"stable/2025.1 only: Drop SLURP job","status":"MERGED","created":"2025-11-22 20:24:44.000000000","updated":"2025-11-23 03:39:29.000000000","submitted":"2025-11-23 03:38:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"968133","meta_rev_id":"7f5ccdb74607045932ec00645839e09a4c677fcc","_number":968133,"virtual_id_number":968133,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-23 03:38:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2025-11-23 02:13:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2025-11-23 02:13:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-22 20:40:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-22 22:07:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"26c9dc5891e6c30639b40d802f8d4c5afe6a5347","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-22 20:24:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"59e77b45711e7602acce8960048fdcd4ee01c557","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-22 20:40:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72aa73f194aa44109531ccf318b29bb8\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b38a021cf33942c0869382e087fe1523 : SUCCESS in 10m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/edc470c8c1424cb0a893f8de76abc58b : SUCCESS in 11m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"da8fb139037bf10ba369ebb70c4d1a19265ff832","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-22 22:07:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44ccbaaa1e0846f7b3a1eb5d2865ca0c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9dc3980c097440482e079b8441b6073 : SUCCESS in 10m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/384c58e018d749ca926f9e614f4af556 : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0428efae62ef4f189d1c2254ce681c77 : SUCCESS in 6m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1ebe5a8118e0457dabfbce34faef768f : SUCCESS in 7m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37fe20503e2c463697ef7dc38fb7e01a : SUCCESS in 9m 32s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/e3db8b10877e4c449a22f7409193b0b4 : SUCCESS in 1h 34m 25s\n- heat-functional https://zuul.opendev.org/t/openstack/build/193fe339d3ae4384b77c977d883e8b58 : SUCCESS in 1h 16m 20s\n- heat-functional-jammy https://zuul.opendev.org/t/openstack/build/7955c44db43c438e8a380ea057d2f27f : SUCCESS in 48m 59s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/d7f6ecc577a1449d81abae8adbe8adde : SUCCESS in 1h 10m 37s\n- heat-functional-centos-9-stream https://zuul.opendev.org/t/openstack/build/1bc9f4b7198740b1a4df31f6de973cb4 : FAILURE in 35m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5cbee814ebd7d0162aaa9b8e9e4b64bf05854931","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-23 02:13:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d6b7ad274ece1b7749576fb145e750851bec5399","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-23 02:14:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"38118352813c9b312451565afeaeaf4f97c595ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-23 03:38:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/753f3aca827142ebbd3254a4acd7cbd4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/446e36ce30ed4a77a9c804a60524851a : SUCCESS in 7m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d3f4fe336ef41b7bc1f999b326e2cff : SUCCESS in 5m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a02a4b668979430aa3d7ef4db10c846e : SUCCESS in 7m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/800192248de245de8366297f98f43192 : SUCCESS in 6m 52s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/6872df1604ce4c31bc3bd81c437eea2a : SUCCESS in 1h 09m 31s\n- heat-functional https://zuul.opendev.org/t/openstack/build/3685767cf258435a95b5e5dd13f1626e : SUCCESS in 48m 02s\n- heat-functional-jammy https://zuul.opendev.org/t/openstack/build/1a655c901c0740e1b716f9bef53bb765 : SUCCESS in 53m 20s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/3c7000b92ba74781a2af694b9f13fc7a : SUCCESS in 1h 16m 32s","accounts_in_message":[],"_revision_number":1},{"id":"cde0ca6e681c40bb8c4bef0f8f74eebe8607df15","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-23 03:38:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7f5ccdb74607045932ec00645839e09a4c677fcc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-23 03:39:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41d8856d20c640bc80ea85556d370972\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2a98e871a7944f289b678be0cbb150c : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3b2ebd3074bd080e3172d45b672b9dd9555e6fb8","revisions":{"3b2ebd3074bd080e3172d45b672b9dd9555e6fb8":{"kind":"REWORK","_number":1,"created":"2025-11-22 20:24:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/968133/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/33/968133/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/33/968133/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/33/968133/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/33/968133/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/33/968133/1"}}},"commit":{"parents":[{"commit":"46cbda240558f5b8a6e27ef6111b7b26e4b65ff1","subject":"Merge \"Fix stack update when reauthentication_auth_method\u003dtrusts\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/46cbda240558f5b8a6e27ef6111b7b26e4b65ff1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-22 20:23:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-22 20:24:34.000000000","tz":540},"subject":"stable/2025.1 only: Drop SLURP job","message":"stable/2025.1 only: Drop SLURP job\n\nstable/2024.1 reached its end of maintenance, so we no longer need to\ntest this upgrade scenario.\n\nChange-Id: I4d698e02c42e199f83ba337974d4b7e9c5725a9f\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/3b2ebd3074bd080e3172d45b672b9dd9555e6fb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/3b2ebd3074bd080e3172d45b672b9dd9555e6fb8"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
