)]}'
{"id":"openstack%2Fheat~968137","triplet_id":"openstack%2Fheat~stable%2F2025.1~Id83c803b68640a403560703ed554e9aa627788fc","project":"openstack/heat","branch":"stable/2025.1","hashtags":[],"change_id":"Id83c803b68640a403560703ed554e9aa627788fc","subject":"Add back ParallelDeploymentsTest in upgrade test","status":"MERGED","created":"2025-11-23 04:00:39.000000000","updated":"2025-11-23 14:23:33.000000000","submitted":"2025-11-23 14:22:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"968137","cherry_pick_of_change":968134,"cherry_pick_of_patch_set":1,"meta_rev_id":"fc561fcd2f12afabdb585b19f724decf01edf42b","_number":968137,"virtual_id_number":968137,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-23 14:22:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2025-11-23 12:34:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2025-11-23 12:34:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-23 04:19:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-23 05:24:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fd9d731d37dd38f6efae328a54d9dcc2991e3f6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-23 04:00:39.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.2.","accounts_in_message":[],"_revision_number":1},{"id":"57da6c223f505c49dce0a71e1657c2d4c76a915e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-23 04:19:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/617cd17619a2429e9a47bc01110962ac\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e68e07908c8d4696a1ee401cb56e75c1 : SUCCESS in 15m 04s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2e5dbeffe72742db9365d8b3d343b8c2 : SUCCESS in 12m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b43d3bfcb2d8de502b3ccc5a800da9620f50c2dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-23 05:24:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51082615ffe8453f813261f702f9f3c2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60c9565a30384e3db767d9d0e4352a1e : SUCCESS in 10m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/371de92ba75846fe98bdb0b18cee0bcb : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5dcd960c0f1e4ef69036fb4093fd1038 : SUCCESS in 8m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fc48e32ece10414c96280c1ca5823700 : SUCCESS in 8m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb6d9cd7f7344069b554936ef41007ec : SUCCESS in 13m 03s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/4fe6739cc8dc4bf7b4698f44329fb927 : SUCCESS in 1h 07m 55s\n- heat-functional https://zuul.opendev.org/t/openstack/build/90f5b9e55f9e4567bddfcae8dfc82d15 : SUCCESS in 1h 16m 36s\n- heat-functional-jammy https://zuul.opendev.org/t/openstack/build/4d4b03c34e3d4c56bd5d8ea3091f6356 : SUCCESS in 1h 16m 56s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/0d4908358997497c90ed0647abc45c66 : SUCCESS in 50m 59s\n- heat-functional-centos-9-stream https://zuul.opendev.org/t/openstack/build/f16461c1d62f4ab4a85be8e2ae440c3c : FAILURE in 15m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"85cdcd3270dc64177e0553a5e448dead5842a892","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-23 12:34:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dadfc52a766309bace84bd577b756aa1d2c922d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-23 12:34:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eeb2c28212c8a191efe5d982d356754e009ea394","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-23 14:22:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/755ecf8a63e14c098fa586bdc71737d4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c47988b51d084f3cba052e6df505d211 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d70af3f9e0b406285cf95a50ad3021d : SUCCESS in 7m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/08390075de5248d6bc0d5d74d8d4230b : SUCCESS in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/938e2cae14d74bb4b590fe26ec25b7ad : SUCCESS in 10m 38s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/29cd146467754300b9f445f628865b6a : SUCCESS in 1h 42m 40s\n- heat-functional https://zuul.opendev.org/t/openstack/build/826f2c7e8c134fedb437edf2ead44c9a : SUCCESS in 1h 19m 50s\n- heat-functional-jammy https://zuul.opendev.org/t/openstack/build/18b27d9b14d941888cdb7deaa4124cb7 : SUCCESS in 56m 28s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/1f1c667b5e4441eab4e8f3a8c7475415 : SUCCESS in 1h 19m 59s","accounts_in_message":[],"_revision_number":1},{"id":"8eeec9d7b4eff3ee07832962e046edcbed619246","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-23 14:22:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fc561fcd2f12afabdb585b19f724decf01edf42b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-23 14:23:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c3d0d1681d64c4097505d3197c3ba6b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14c888b379cb4c8caeb3e26e770d33e8 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8bb75011745c20ad5aa21bbb23c8137cc1cd7f13","revisions":{"8bb75011745c20ad5aa21bbb23c8137cc1cd7f13":{"kind":"REWORK","_number":1,"created":"2025-11-23 04:00:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/37/968137/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/37/968137/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/37/968137/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/37/968137/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/37/968137/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/37/968137/1"}}},"commit":{"parents":[{"commit":"f5543343f343041bb832d973d4d82275fe45585e","subject":"Merge \"stable/2025.1 only: Drop SLURP job\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/f5543343f343041bb832d973d4d82275fe45585e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-22 16:10:27.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-23 04:00:39.000000000","tz":0},"subject":"Add back ParallelDeploymentsTest in upgrade test","message":"Add back ParallelDeploymentsTest in upgrade test\n\nChange-Id: Id83c803b68640a403560703ed554e9aa627788fc\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n(cherry picked from commit 71d2c41764cc9b89c729faca683dd7742b46cdfe)\n(cherry picked from commit f2db16ff6bfd17c76566d14c02b494b2d3db1c11)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/8bb75011745c20ad5aa21bbb23c8137cc1cd7f13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/8bb75011745c20ad5aa21bbb23c8137cc1cd7f13"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
