)]}'
{"id":"openstack%2Fheat~978246","triplet_id":"openstack%2Fheat~master~I4ee243eb79c1a100d6fc0a0b349f7fd1852822e4","project":"openstack/heat","branch":"master","topic":"story/2011685","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-02-28 05:01:26.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I4ee243eb79c1a100d6fc0a0b349f7fd1852822e4","subject":"Cap nova api microversion according to novaclient","status":"MERGED","created":"2026-02-27 18:10:46.000000000","updated":"2026-02-28 06:51:52.000000000","submitted":"2026-02-28 06:50:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978246-story/2011685","meta_rev_id":"71fe3149bfd91e215639989e0eb723afaebbcbd3","_number":978246,"virtual_id_number":978246,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-28 06:50:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-02-28 05:01:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-02-28 05:01:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-27 18:46:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-27 20:15:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bff7334c924cb1626ebd9b47fe0f1e7eb180a22c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-27 18:10:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6264dfe4bc9e003b19644fb5985ce489a8ff92c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 18:46:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdd60bee51fe40dd95fa5d5019ed6c37\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d860abc1124a4bf485e97f83b351afaa : SUCCESS in 7m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e81d5f0f62854ab4945fc8d4bbc47b62 : RETRY_LIMIT in 3m 49s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b5a8e3e09b8145bfaea9003728aa5757 : SUCCESS in 15m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"035d40406904c45727044e57cedfa35d3b4df911","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 20:15:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/02a1b1e0e8204ad4811dfa481672ace6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e18f2cfff1c4f08bf8ae12cc1d713ce : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1072d51dd3c04cfeaffd5a6f33bdab86 : SUCCESS in 3m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/63efa2ab2d2b4a1a8f83fc8902ead023 : SUCCESS in 4m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fa75315156fe49b5b029ac0033998cde : SUCCESS in 5m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/84e9734131c94781af6a1f0c38a42847 : SUCCESS in 10m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a6dda4d802842aaaef71d721d152093 : SUCCESS in 6m 24s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/609c226ac7a24ed8b0cae27958deb276 : FAILURE in 1h 49m 43s\n- grenade-heat-multinode-skip-level https://zuul.opendev.org/t/openstack/build/661453641c1b4edfa8a1a53a702ef988 : FAILURE in 1h 31m 15s\n- heat-functional https://zuul.opendev.org/t/openstack/build/b57337e7449749a5b7cef7d295d5761d : FAILURE in 49m 48s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/43d1b2ee00484b728a9cf50214daf89f : FAILURE in 1h 00m 37s","accounts_in_message":[],"_revision_number":1},{"id":"68b59c56b9ce62fb1555941925fa7b3498424be9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-28 01:06:33.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"4b4c16a8c4c52c6f30c5c0bb98d071ab0151eeae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 01:23:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4b2bbed43404c149e0cdefd289458dc\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/42f0cf937e9845788dd3710f88298d2c : SUCCESS in 10m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9b2de73b075f4778900d1ab687c69e98 : RETRY_LIMIT in 3m 10s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4f960be338df443d9b56b91c87e0b009 : SUCCESS in 11m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e33bd93fd7ea3e5eff07c043f7c490b828136116","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-28 02:33:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7aae37f53278a060c7b904fb3b96ea82abf14f52","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 04:26:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1d9082959e948e8a46efed49d5c3573\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7b5f5526951b4eaf92d9745e92e24334 : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/01e31cb908524f368f51c4ae3718e61d : RETRY_LIMIT in 3m 43s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/041c89f3f4354ea6995f5d9752e77396 : SUCCESS in 10m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"47f8290d875462e468a4ecbdca9c0c2358c43a5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 04:44:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e809e5c925848a7b45d62700d7ca2f6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aeeeb95f1b4a430c80c07df75b5de4a9 : SUCCESS in 10m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f79df652cecb472581845b081913f599 : SUCCESS in 6m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d87eea21e3cb48ecb1a05221cc998bf6 : SUCCESS in 7m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6d12cea86c124e18bb173217a86059f0 : SUCCESS in 6m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/edaefd18b29d48dc9b2edb6d581c195b : SUCCESS in 10m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7c4d4baa27f4a8a8da58a1e783fb026 : SUCCESS in 16m 33s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/b0884fd166de47c897fb0b095f0481f0 : SUCCESS in 1h 42m 27s\n- grenade-heat-multinode-skip-level https://zuul.opendev.org/t/openstack/build/5734efc0c46a471ab4b29ef22ec23e5d : SUCCESS in 1h 43m 17s\n- heat-functional https://zuul.opendev.org/t/openstack/build/5a399bdae2bf43ae9322000cc283722e : SUCCESS in 1h 26m 38s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/050217139bce4287bc0d3e5a1332f474 : SUCCESS in 1h 14m 30s","accounts_in_message":[],"_revision_number":3},{"id":"ff89eb49acf321d3814d86a29942533ced7b41ff","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-28 05:01:26.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3102ace885fdf6b7ca8b5385373a1e1b5fdb6ab6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 05:02:20.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0f1c2461ba69b712323bf0923303ab811050aadb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 06:50:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c0292d873ab45e590860f519e17bc65\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e47f9c131134e57a7d66d570002e300 : SUCCESS in 5m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3a2ecfe832234c75bbeabf047e0d699c : SUCCESS in 3m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5c7b7b50044c4186babd31abf69403c1 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/442a4cdb0cab4a3c9a97b9f01f53acf6 : SUCCESS in 6m 11s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/4ed3958fe39e4519ad575ccae2df5874 : SUCCESS in 1h 08m 20s\n- grenade-heat-multinode-skip-level https://zuul.opendev.org/t/openstack/build/96e02c3cacf74b30972cb50095000f30 : SUCCESS in 1h 41m 12s\n- heat-functional https://zuul.opendev.org/t/openstack/build/9d291455196c47d3900ae2413b48843e : SUCCESS in 1h 00m 17s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/ab1434d038404feeb1a1b925e1dd6256 : SUCCESS in 41m 18s","accounts_in_message":[],"_revision_number":3},{"id":"9f8d40a422c4dc38502577e91304906f5fd395c6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 06:50:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"71fe3149bfd91e215639989e0eb723afaebbcbd3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-28 06:51:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f729e40bdc714fc4aeb444d88e22a393\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06c4398414c547e09609fccf23eb7f85 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e6d1abe938ddfb30b3f63f439fcddd2d74e04985","revisions":{"2630fb623fd5986d8aa42ded4022adf99b9dd8a4":{"kind":"REWORK","_number":1,"created":"2026-02-27 18:10:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/46/978246/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/46/978246/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/46/978246/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/46/978246/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/46/978246/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/46/978246/1"}}},"commit":{"parents":[{"commit":"f110fc56492022f71b38a882c9a1ea60ba4da6d4","subject":"Merge \"Reject incomplete snapshot for restore operation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/f110fc56492022f71b38a882c9a1ea60ba4da6d4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-27 18:09:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-27 18:10:07.000000000","tz":540},"subject":"Cap nova api microversion according to novaclient","message":"Cap nova api microversion according to novaclient\n\nnovaclient has own maximum API microversion which it can supports.\nMake sure we don\u0027t use any version newer than this limit.\n\nDepends-on: https://review.opendev.org/978075\nChange-Id: I4ee243eb79c1a100d6fc0a0b349f7fd1852822e4\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/2630fb623fd5986d8aa42ded4022adf99b9dd8a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/2630fb623fd5986d8aa42ded4022adf99b9dd8a4"}]},"branch":"refs/heads/master"},"b4024adcc3db9f121782ee24f25baecc2af5c4fd":{"kind":"REWORK","_number":2,"created":"2026-02-28 01:06:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/46/978246/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/46/978246/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/46/978246/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/46/978246/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/46/978246/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/46/978246/2"}}},"commit":{"parents":[{"commit":"f110fc56492022f71b38a882c9a1ea60ba4da6d4","subject":"Merge \"Reject incomplete snapshot for restore operation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/f110fc56492022f71b38a882c9a1ea60ba4da6d4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-27 18:09:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-28 01:06:21.000000000","tz":540},"subject":"Cap nova api microversion according to novaclient","message":"Cap nova api microversion according to novaclient\n\nnovaclient has own maximum API microversion which it can support.\nMake sure we don\u0027t use any version newer than this limit.\n\nStory: 2011685\nTask: 53896\nDepends-on: https://review.opendev.org/978075\nChange-Id: I4ee243eb79c1a100d6fc0a0b349f7fd1852822e4\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/b4024adcc3db9f121782ee24f25baecc2af5c4fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/b4024adcc3db9f121782ee24f25baecc2af5c4fd"}]},"branch":"refs/heads/master"},"e6d1abe938ddfb30b3f63f439fcddd2d74e04985":{"kind":"REWORK","_number":3,"created":"2026-02-28 02:33:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/46/978246/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/46/978246/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/46/978246/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/46/978246/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/46/978246/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/46/978246/3"}}},"commit":{"parents":[{"commit":"f110fc56492022f71b38a882c9a1ea60ba4da6d4","subject":"Merge \"Reject incomplete snapshot for restore operation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/f110fc56492022f71b38a882c9a1ea60ba4da6d4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-27 18:09:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-28 02:31:54.000000000","tz":540},"subject":"Cap nova api microversion according to novaclient","message":"Cap nova api microversion according to novaclient\n\nnovaclient has own maximum API microversion which it can support.\nMake sure we don\u0027t use any version newer than this limit.\n\nStory: 2011685\nTask: 53896\nDepends-on: https://review.opendev.org/978075\nChange-Id: I4ee243eb79c1a100d6fc0a0b349f7fd1852822e4\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/e6d1abe938ddfb30b3f63f439fcddd2d74e04985"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/e6d1abe938ddfb30b3f63f439fcddd2d74e04985"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
