)]}'
{"id":"openstack%2Fheat~992639","triplet_id":"openstack%2Fheat~stable%2F2026.1~I47de8d474a44e0616b1053b8491938c58c9776f1","project":"openstack/heat","branch":"stable/2026.1","hashtags":[],"change_id":"I47de8d474a44e0616b1053b8491938c58c9776f1","subject":"Enable trunk plugin in CI","status":"MERGED","created":"2026-06-10 13:10:51.000000000","updated":"2026-06-10 17:31:32.000000000","submitted":"2026-06-10 17:29:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"992639","meta_rev_id":"cfe8c15cbd6d830ae2ad5e071431b85d526ce9b4","_number":992639,"virtual_id_number":992639,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-10 17:29:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-06-10 15:23:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-06-10 15:23:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-10 15:19:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"821f227d789ad544f0f27165d71eb7ad6876130c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-10 13:10:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cae38b70497a78998432f1f48973fc9695506966","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 15:19:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a26b0f95ba094ef6b7ed6aefb7ce3ec0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5d28a2e3ab74e0bbaaf1a6ae7833943 : SUCCESS in 8m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3acca128c6df4ea1955fa612763d5aa6 : SUCCESS in 5m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/777df724cc194c12904c67e25e40a32c : SUCCESS in 7m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/40ba16d51ce14984b18ad3edbd58af6a : SUCCESS in 6m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bd32cecd2fa042c8a095fa2229bdd07a : SUCCESS in 9m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f744dffb6b1342d89849290ec37079a9 : SUCCESS in 10m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0870df7b30743aba30826b975487257 : SUCCESS in 12m 55s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/74267e899ed04bd8bada27a04dbcd524 : SUCCESS in 1h 45m 57s\n- grenade-heat-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e6be202043e14e17a3f92f3b03398d5c : SUCCESS in 2h 00m 03s\n- heat-functional https://zuul.opendev.org/t/openstack/build/3a32489b03174f3989fd51f8ca399c3e : SUCCESS in 1h 24m 50s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/ae0641847ea7400dbd490c061b9bfa66 : SUCCESS in 1h 17m 12s","accounts_in_message":[],"_revision_number":1},{"id":"05b040829b2276e5324871dc4c5d815e13eaa25b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-10 15:23:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9a8df987c6a4274d6cc8f890ff3dafe5c5ec5d6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 15:24:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a2697e81f67eea854345b49ea540b7a1e9b72e0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 17:29:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43f1f6876e8b416aabcb91df0adc79f7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a70edfa5660f49988413fd0daa9007d0 : SUCCESS in 6m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/08f3494a789f4691908da931f4603d74 : SUCCESS in 7m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0fbf15d01bc64a87b5859a242ff288a0 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc763abaedc8407b8ae9db4c1598c0af : SUCCESS in 7m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0532316dd104003bede7e325534890a : SUCCESS in 6m 39s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/99ec4e40147e453f9beba5ca7a7d75ae : SUCCESS in 1h 51m 32s\n- grenade-heat-multinode-skip-level https://zuul.opendev.org/t/openstack/build/5c8facf01ca049dea1fc0062245e58b7 : SUCCESS in 1h 57m 57s\n- heat-functional https://zuul.opendev.org/t/openstack/build/b51e00886a78410f8c4929069ace5661 : SUCCESS in 1h 00m 23s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/515e67136a42490fb0a93696921349d7 : SUCCESS in 49m 47s","accounts_in_message":[],"_revision_number":1},{"id":"3bc0d52b6c679f06dc4724afc98922f2da904e42","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 17:29:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cfe8c15cbd6d830ae2ad5e071431b85d526ce9b4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 17:31:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e55cc4ca7a8f4c4a899fa88d554342bd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d20f0e24db464f84ba57aea5019f54e8 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ffc76501e5860cd8ae7d948f84ed2679f8c6509e","revisions":{"ffc76501e5860cd8ae7d948f84ed2679f8c6509e":{"kind":"REWORK","_number":1,"created":"2026-06-10 13:10:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/39/992639/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/39/992639/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/39/992639/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/39/992639/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/39/992639/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/39/992639/1"}}},"commit":{"parents":[{"commit":"8a31c6944947ae6cdeead44985f3adeebc5b4afb","subject":"Merge \"Preserve auto-populated binding:profile fields during update\" into stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/8a31c6944947ae6cdeead44985f3adeebc5b4afb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-10 06:50:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-10 13:10:36.000000000","tz":540},"subject":"Enable trunk plugin in CI","message":"Enable trunk plugin in CI\n\n... to run relevant tests.\n\nConflicts:\n\t.zuul.yaml\n\nChange-Id: I47de8d474a44e0616b1053b8491938c58c9776f1\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n(cherry picked from commit 61d052bdd2ba036049c387bc0d9bec87ac208d66)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/ffc76501e5860cd8ae7d948f84ed2679f8c6509e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/ffc76501e5860cd8ae7d948f84ed2679f8c6509e"}]},"branch":"refs/heads/stable/2026.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
