)]}'
{"id":"openstack%2Fheat~997882","triplet_id":"openstack%2Fheat~master~Ifb3442d151aa66fe3ec07234e7d52dc9425a6b57","project":"openstack/heat","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-07-20 11:43:33.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Ifb3442d151aa66fe3ec07234e7d52dc9425a6b57","subject":"devstack: Maintain hidden_stack_tag(s) option by plugin","status":"MERGED","created":"2026-07-19 16:27:13.000000000","updated":"2026-07-20 16:14:28.000000000","submitted":"2026-07-20 16:11:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"997882","meta_rev_id":"c33c0860cabc7dfe8992c0b2f78baf11613b0edf","_number":997882,"virtual_id_number":997882,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-20 16:11:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-07-19 18:01:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-07-19 18:01:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-19 16:45:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-19 17:48:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"96e738308b4d0530c8f85c151ea837ba23b1ab66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-19 16:27:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8102d954e937a8361e240e03dda5157430a4ec9d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-19 16:45:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0306851cd9e431fb9bfc1af0c8a4271\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1dae6d2f26a0488b8edd81f1814303b5 : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/27500d6dfa0d4b458682a368d38d5623 : SUCCESS in 9m 36s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7412652f469945d6a0fd7120f901f95c : SUCCESS in 15m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"08bfb83718d013dcdc1baa9a876fce441b16d2a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-19 17:48:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7db4abedaa63437cab93c3b9a3573795\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/123d65d60fc44f979a3520f645bece4c : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/030409e8ce2f48c384a89e83d55fce86 : SUCCESS in 3m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a9587a0b668c40369d115de706d19035 : SUCCESS in 8m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b7dbfd7945c9440c8baec9cd1c9428f2 : SUCCESS in 3m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/70724ff49d8b46439a94928205560ace : SUCCESS in 9m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0f71fbc4030470a9fe8c9a5a2a7a280 : SUCCESS in 7m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1870e53b7d642e1bb60dcde8eeb44db : SUCCESS in 14m 27s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/aa372d6364be4ef4b5596ee1ff00ced7 : SUCCESS in 59m 46s\n- heat-functional https://zuul.opendev.org/t/openstack/build/75f122ab874f4cb98a790d5d0eb91637 : SUCCESS in 1h 15m 37s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/d407ddcb401842ecb07d2c69614f3077 : SUCCESS in 54m 23s","accounts_in_message":[],"_revision_number":1},{"id":"293995c4a01f1e70553956ab4075cd6b0850ecb6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-19 18:01:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3e2d80b2a2d500e53d47d1778d0cec12d314bd5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-19 18:01:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b09c92555ad2d8b62bbd3bdfd104ca69b05d7841","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-19 19:41:35.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f9b5445771794cb5882dc976cce7a8ac\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b66992f2b2a94d20bb51721c38794c29 : SUCCESS in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ea0236aa57b94023971eeeb88bd0c42b : SUCCESS in 7m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5eef99de07ac4a6f9ab54f70ce08ef7d : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26df4d028df1466b81d5abada105bbd4 : SUCCESS in 13m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a3add2d6b2c459cba0be8293efedacc : SUCCESS in 12m 05s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/60b88af876624d6ebf0cfca0856586ad : SUCCESS in 1h 29m 01s\n- heat-functional https://zuul.opendev.org/t/openstack/build/1419b79153534ba29166ccb9cb06a8db : SUCCESS in 44m 04s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/bf130c98e9244911aa94b9eb52cc0015 : FAILURE in 1h 08m 07s","accounts_in_message":[],"_revision_number":1},{"id":"8e96dab2d3f30bce8abb821f1b1fb18c89aef26e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-20 00:54:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ae5666b9051704ffee9be8f5f7c3335e7a144364","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 01:13:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5464ed58eb164390879b1f12ba7fcd08\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/57734cf13c054eaaa3d09f04153f6923 : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a9e5b1f8233040c29cfac7365de990e0 : SUCCESS in 9m 59s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7bb80886258a4fb3ba214bed2e6e6795 : SUCCESS in 16m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d3049d880b1185ee87001690b12a5844bd9046ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 02:50:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/940ebe59b2e04c7fa75a26662ccefbd0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81c3d9ccd0f64331a9d99def483f4868 : SUCCESS in 9m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f4643b2c1d54bb1be4f12ec30acab27 : SUCCESS in 6m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3d49d869684845529ec39b979328680a : SUCCESS in 6m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/302c32096ecb4c9795782b13632e9e2f : SUCCESS in 5m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dc2ec02471304e189c7ad4ce22d90848 : SUCCESS in 10m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73899bdb314c4034a4e5dc1f09cc8666 : SUCCESS in 12m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0867fb0292424d538e551ce04f47a313 : SUCCESS in 12m 06s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/dffb4bcbda284e7b9784eb58b6b45a86 : SUCCESS in 1h 49m 18s\n- heat-functional https://zuul.opendev.org/t/openstack/build/d7ea9f732aec4ccb8b7bf3f712abebab : FAILURE in 1h 33m 07s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/65ee4a78bea24fd092432d47d6e5ce57 : SUCCESS in 1h 13m 25s","accounts_in_message":[],"_revision_number":1},{"id":"cf4c9a19d36334d4482ed85dc1da6088562cfeef","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-20 04:29:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f9feda284173b5b2eb9cccf8f1a4a3e0300c68cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 04:51:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/502d8a5749154c469907a02af0aecbab\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6638cccafe894578a456e1b062b2fc7c : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e19b06c26e854cf3a72f09619f86ff6d : SUCCESS in 12m 04s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/02e06fdc552d495e82bb266a2de139b0 : SUCCESS in 19m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fdd3f65c4cd2845b3dd65354e4a2bd4ed46c7d35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 05:54:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a38f07a9db647448deebea089c63c90\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af600690224a4f269d5ba9ff3c34cee5 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/069018726e844779b08ea37778067b59 : SUCCESS in 3m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9b05adcde96e4cbeb8a72cc8536f15de : SUCCESS in 3m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e69426beba9a415f91eb935ad4850156 : SUCCESS in 3m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2a5c195e774f4e9caf0223d02bbb634d : SUCCESS in 10m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e3732b320fd4db69afda8e65b64cf45 : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4f3e246350247238de815786914c3a8 : SUCCESS in 6m 04s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/c96f43fbc9b14152815d3f6718f1a0aa : SUCCESS in 1h 24m 26s\n- heat-functional https://zuul.opendev.org/t/openstack/build/3b553a9b26f944978177f102072bc889 : FAILURE in 55m 01s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/0e1f2b712927475e878c80d6baedc70e : SUCCESS in 38m 22s","accounts_in_message":[],"_revision_number":1},{"id":"e68ef6c6bd9089062bfb2b27f32b653e378313d2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-20 07:13:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"58c9f5bcd1dfa7e6b1808a4fb94af919b0368b34","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-20 07:13:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b6086001816a7ceed630b08efff22a355f2be5da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 07:32:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69f5f4d84d1c43e2a86b764cc42b3f94\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a517a35e2a0044128dbe041f5bdb4f43 : SUCCESS in 10m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/330ce9e0cd244773910a322e2d9b1d35 : SUCCESS in 9m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8c28889d72564fdc9eeba935fd59b034 : SUCCESS in 10m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7a83f796c2db99340aa07693f07ba3b58100e79b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 08:45:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c4d76c10d0c4c989efe59725479840f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12e5c7982aa74377b5e6b11c36b8f2ac : SUCCESS in 9m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00bccdd927c845b8a6cbfd745dee5c06 : SUCCESS in 5m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c8480880a5314d0ca5c38d349f0a8d22 : SUCCESS in 6m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f7eaac84ce3640149ff2ac044009f263 : SUCCESS in 6m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7648470c27c24157aa3ba4ff42eb4200 : SUCCESS in 10m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/995ead47504f4e089ade757052906e80 : SUCCESS in 14m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42f72ccd4cbb4e318205f9b3e7cea06d : SUCCESS in 11m 55s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/1760db838350414a8b4c8ce199dfeafb : SUCCESS in 1h 26m 15s\n- heat-functional https://zuul.opendev.org/t/openstack/build/1abe80184cdf4e40a2f5567f747bbf0d : FAILURE in 1h 12m 39s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/f053f69382d54b15823178028700b9f2 : SUCCESS in 1h 03m 38s","accounts_in_message":[],"_revision_number":1},{"id":"968191a3006587be35bf822f28924f3b165c4544","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-20 09:56:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"186fdd6f2d0b5729c45d5c26d3bfcf3216d3ddf4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 10:27:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c7876f0b433470596d24a2b642f8a5e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7eab262535d54f329e3a173434292ef6 : SUCCESS in 11m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c6a7b453c3d748a2b2c04573e69f7808 : SUCCESS in 9m 49s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8b355689f90142acb67a26054cfd4f33 : SUCCESS in 9m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f13dd6df39883256538a9abbd12f14d8914fedc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 11:40:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/350d31fa7dc34da2bb52a93d7f8bdd73\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33ad5cedcccf4c05a7290559a8d78eea : SUCCESS in 11m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d4bc986b89b4f94b936756fa2f5c40e : SUCCESS in 5m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d2dc2a52af7a4260affa8bc7f32b4297 : SUCCESS in 6m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1751db95a4b1409a9ffaccf136d06078 : SUCCESS in 5m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e7d4bfbff56d4273a8df7cc5e0a35f6b : SUCCESS in 8m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d07e0262a1c24253a8bc55e518c9b7de : SUCCESS in 12m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c60ed56603004639bff8d0d648e60e5e : SUCCESS in 11m 57s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/1aeadfd8966645dfaa36530318e38a70 : FAILURE in 1h 37m 36s\n- heat-functional https://zuul.opendev.org/t/openstack/build/fc9dab3d07a24778ae8d3bba0fad0faf : SUCCESS in 1h 17m 35s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/e430a90651354eedab050bd2af9e5827 : SUCCESS in 46m 53s","accounts_in_message":[],"_revision_number":1},{"id":"bc12c4e8eb3a5a71769945675821f5a675ddc3a6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-20 11:43:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"38ad8eead04b5783693a55838d7a164ef317d140","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 12:15:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fedf37cc701b49c5896c058d3171f6be\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7bf4721b668e477c85e61f6185ed45fb : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0d434034348e4aa696e956e234550b48 : SUCCESS in 12m 01s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8df86fded6e844278f651ce4eed72a58 : SUCCESS in 18m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f6483dbd571d58feb4c578795e5532925155997d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 13:26:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3200b9dd34684ca9bcd3ba8e98f5f94b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a932425e0d0d4b82b149ac9922c38b45 : SUCCESS in 10m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a261bd91ec8476db4d099f0fecd7962 : SUCCESS in 5m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/17b8eb0e8a3c4ce486ee07821b69a47c : SUCCESS in 5m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b4a3ad29a0ad483ab0d4a0915d5aa2bd : SUCCESS in 7m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e7e2105203534b9582301d48e8488095 : SUCCESS in 10m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25f365eed38440c5b8a10a6c564277ae : SUCCESS in 17m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d237ec10a90848bdaf343241274b520d : SUCCESS in 13m 52s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/4c18914c5ddd4590bab8750385354200 : SUCCESS in 1h 35m 18s\n- heat-functional https://zuul.opendev.org/t/openstack/build/91dc257ac03d4d9388028e6701218a87 : SUCCESS in 1h 19m 52s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/b39cf2c7cdc94c1bacdce47f253fb1d7 : SUCCESS in 1h 27m 27s","accounts_in_message":[],"_revision_number":1},{"id":"7072a8968092518bcb85f44ba72235547e9669aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 13:27:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1a4cd2372217d9637347cb40ce01dbc9c2bb2e77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 16:11:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c556ab13af874a3daba7d0f889634bfc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ab9ed2e8bea49ff874bdcb10c3b822b : SUCCESS in 5m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/69fc80eec3c0451386846472793b9968 : SUCCESS in 6m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/db1b0b3f8094479db64a1df9ec50d077 : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5919359ae9d646158db90ca64ff2cdac : SUCCESS in 12m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36435b172973420fa08c5d36864589b4 : SUCCESS in 14m 14s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/6b19ba3c25d34b4cb51cc30db5c27623 : SUCCESS in 1h 37m 19s\n- heat-functional https://zuul.opendev.org/t/openstack/build/891e34e021a3498aab372a33918ad69e : SUCCESS in 1h 09m 24s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/0b1141c1bbf14e55afe14a21259825be : SUCCESS in 1h 12m 08s","accounts_in_message":[],"_revision_number":1},{"id":"7ee7a52fade8dd87fd1c6aede25c1fbec48ba71a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 16:11:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c33c0860cabc7dfe8992c0b2f78baf11613b0edf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 16:14:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f84e0d8195c94aabb867b54946035878\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b955f64f6fab48ca8bf88d1ad35075fe : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"91badd5a41ae0ead36a9fc9502d16188d3f43bcf","revisions":{"91badd5a41ae0ead36a9fc9502d16188d3f43bcf":{"kind":"REWORK","_number":1,"created":"2026-07-19 16:27:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/82/997882/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/82/997882/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/82/997882/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/82/997882/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/82/997882/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/82/997882/1"}}},"commit":{"parents":[{"commit":"0dce00d42b3dd7c50c8715d3a6467eabe46e96f3","subject":"Remove vitrage integration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/0dce00d42b3dd7c50c8715d3a6467eabe46e96f3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-19 16:26:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-19 16:26:22.000000000","tz":540},"subject":"devstack: Maintain hidden_stack_tag(s) option by plugin","message":"devstack: Maintain hidden_stack_tag(s) option by plugin\n\nThis option should be explicitly set so that tempest-plugin works.\n\nChange-Id: Ifb3442d151aa66fe3ec07234e7d52dc9425a6b57\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/91badd5a41ae0ead36a9fc9502d16188d3f43bcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/91badd5a41ae0ead36a9fc9502d16188d3f43bcf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
