)]}'
{"id":"openstack%2Fheat~998536","triplet_id":"openstack%2Fheat~stable%2F2025.1~I729c6b8b44adb274f5e073e280b349d0668dc089","project":"openstack/heat","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-07-24 00:10:54.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I729c6b8b44adb274f5e073e280b349d0668dc089","subject":"stable/2025.1 only: Disable Jammy job","status":"MERGED","created":"2026-07-23 16:57:00.000000000","updated":"2026-07-24 02:13:45.000000000","submitted":"2026-07-24 02:02:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"998536","meta_rev_id":"e75f5e2dcdceb69bf247e89272ca3328800c0a56","_number":998536,"virtual_id_number":998536,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-24 02:02:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-07-24 00:10:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-07-24 00:10:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-23 18:21:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3f085417dfc372f8b0a1e9045c6f7ffdb321debc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-23 16:57:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b328b399452d237ad0352f13e3203855094be444","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 18:21:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/99ec33c71c744af2835bb1168a04aab0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b63442005084d6dbaa4907771520ae9 : SUCCESS in 13m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bd3b101431a4769a41a418bb8c96159 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67da4c5ec14e45ce924742aaf3d429bb : SUCCESS in 8m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f76b90d1e05e46429cb4a864f1b78060 : SUCCESS in 9m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5942a0afb5704a9cb27001ae2b2edced : SUCCESS in 12m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9281bcb530804601a64a267f0d56274e : SUCCESS in 15m 55s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/70d600984df14093b15057b3c59e7165 : FAILURE in 41m 41s\n- heat-functional https://zuul.opendev.org/t/openstack/build/b3b57a3d24bf496eb821292116bcd7aa : SUCCESS in 1h 16m 24s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/dd199d1739da43788da0925a651219ea : SUCCESS in 1h 07m 43s\n- heat-functional-centos-9-stream https://zuul.opendev.org/t/openstack/build/c720ecdbabad47b29e973d5df7f5897e : FAILURE in 22m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"92131f10c0a0e5f64e539a641883b4a4e8ee0c6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-23 18:23:56.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"4ceac84a6059d19ee9a8b9e69f5d98bb2243c3fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 19:38:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2576202401f47828fee4559036cc1b3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e2167fdceed46edbccbcbcf9ee038f6 : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f293497d6f5441fa63d3a6e5af10e05 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c40d768810a4d5c9ef58571d9188c50 : SUCCESS in 10m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a68e3d0697994a108b615ba784ec22fb : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddff47b88529461a9af2f3d681e45b88 : SUCCESS in 6m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/586411fc3d104bd4a421479cb0d9ed85 : SUCCESS in 4m 55s\n- heat-functional https://zuul.opendev.org/t/openstack/build/d32aa404e4154b428ffbd359413d2d7d : SUCCESS in 1h 11m 41s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/142de346ff6043dd91ca5d56c3a1218a : SUCCESS in 43m 24s\n- heat-functional-centos-9-stream https://zuul.opendev.org/t/openstack/build/b6066bcebcbd4b04b2a3b51abe5be8d4 : FAILURE in 17m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"17a0eb1ec50b9e5d01054707ffa2eb38a452aa8e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-24 00:10:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a1274b10401b7987ab70757dd0b9f603d0bff106","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 00:11:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"82d1234493ab8a26c231f041d555210d13525f5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 02:02:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fb1c1825ad24544846a5cbb126f0dc5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b91bed1465b4596868bb6fd28709322 : SUCCESS in 2m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/abe7f423bd9d4e628db96b11c6606469 : SUCCESS in 10m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b364f3fd736f4a828795493136f3cccd : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec1ed9fac7eb44bd9e55a272504ba7af : SUCCESS in 13m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2888476d47a46e58d37b17a1b4d0b74 : SUCCESS in 8m 53s\n- heat-functional https://zuul.opendev.org/t/openstack/build/f3b994d01e3e4fd0b642ac7a3ebb0d3e : SUCCESS in 1h 16m 53s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/a90a24444f9c44b58199d2b33e49bb8b : SUCCESS in 52m 15s","accounts_in_message":[],"_revision_number":2},{"id":"15f842d85859f5f363d5ca6c22c1ae19286f3558","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 02:02:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e75f5e2dcdceb69bf247e89272ca3328800c0a56","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 02:13:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/360073b1ce7e4839a7d93326b89d2e7d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a45bce2e41a34a3fb2b405727d270f5f : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a132308615b460c38903a2ba26eaab7c460817fa","revisions":{"98bf7a6d35f4cf0f7e271e9dd8161ead36d199de":{"kind":"REWORK","_number":1,"created":"2026-07-23 16:57:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/36/998536/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/36/998536/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/36/998536/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/36/998536/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/36/998536/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/36/998536/1"}}},"commit":{"parents":[{"commit":"5e56c980e8f20ecee20c906314214d23c1ee61ce","subject":"Add `sata` to OS::Nova::Server disk_bus constraint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/5e56c980e8f20ecee20c906314214d23c1ee61ce"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-23 16:56:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-23 16:56:40.000000000","tz":540},"subject":"stable/2025.1 only: Disable Jammy job","message":"stable/2025.1 only: Disable Jammy job\n\nThe latest tempest no longer runs on Jammy because Python 3.10 support\nwas removed. Disable the Jammy job to unblock CI. We may pin tempest\nand plugins later to restore it.\n\nChange-Id: I729c6b8b44adb274f5e073e280b349d0668dc089\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/98bf7a6d35f4cf0f7e271e9dd8161ead36d199de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/98bf7a6d35f4cf0f7e271e9dd8161ead36d199de"}]},"branch":"refs/heads/stable/2025.1"},"a132308615b460c38903a2ba26eaab7c460817fa":{"kind":"REWORK","_number":2,"created":"2026-07-23 18:23:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/36/998536/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/36/998536/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/36/998536/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/36/998536/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/36/998536/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/36/998536/2"}}},"commit":{"parents":[{"commit":"5e56c980e8f20ecee20c906314214d23c1ee61ce","subject":"Add `sata` to OS::Nova::Server disk_bus constraint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/5e56c980e8f20ecee20c906314214d23c1ee61ce"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-23 16:56:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-23 18:23:25.000000000","tz":540},"subject":"stable/2025.1 only: Disable Jammy job","message":"stable/2025.1 only: Disable Jammy job\n\nThe latest tempest no longer runs on Jammy because Python 3.10 support\nwas removed. Disable the Jammy job to unblock CI. We may pin tempest\nand plugins later to restore it.\n\nAlso disable grenade job because 2024.2 already reached its EOL.\n\nChange-Id: I729c6b8b44adb274f5e073e280b349d0668dc089\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/a132308615b460c38903a2ba26eaab7c460817fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/a132308615b460c38903a2ba26eaab7c460817fa"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
