)]}'
{"id":"openstack%2Fheat~999633","triplet_id":"openstack%2Fheat~stable%2F2026.1~I27ee7511417ac2548729e2c91c93c756f9300668","project":"openstack/heat","branch":"stable/2026.1","hashtags":[],"change_id":"I27ee7511417ac2548729e2c91c93c756f9300668","subject":"devstack: Set image_ref and instance_type","status":"MERGED","created":"2026-08-03 17:03:27.000000000","updated":"2026-08-04 07:59:03.000000000","submitted":"2026-08-04 07:59:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"999633","meta_rev_id":"2dfd6b2a59838537727e9b487370cd867d40f07e","_number":999633,"virtual_id_number":999633,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-04 07:59:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-08-04 05:57:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-08-04 05:57:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-03 19:03:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ea4821a285771931a5fbfe6eca45cbb464bd7edc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-03 17:03:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a771fd20a879bf0362b02bd4e2738c2935c838f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-03 19:03:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0436d09ab67a4ec482a9ce6fc3c7ee5a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4602923359e6456da7d312c4a0c75104 : SUCCESS in 9m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cf76726b3fe45538d5a8f8a47112fdd : SUCCESS in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f2624ce49439487bbc7f9506e6c47d88 : SUCCESS in 6m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bebe448339af4e46ac1492d80f755790 : SUCCESS in 5m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3f5397dbc67c45e3a32c1f185166f95f : SUCCESS in 10m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b447481c8e1542da9d524f49c2bf8d92 : SUCCESS in 9m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dac7f6f07da646fe8dd5e62bbd09cb46 : SUCCESS in 12m 44s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/a4013ce99f5249c1af91e108b655fddf : SUCCESS in 1h 14m 08s\n- grenade-heat-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4ef966eb7da24c78a453f6eedcc02267 : SUCCESS in 1h 51m 38s\n- heat-functional https://zuul.opendev.org/t/openstack/build/b43fbcfe594a454d968a63bf9b111a30 : SUCCESS in 1h 26m 12s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/c46f44508c17414f9a26795089b41590 : SUCCESS in 51m 33s","accounts_in_message":[],"_revision_number":1},{"id":"befbab1d5e17b50ef5f077d71569c8eb26c91f40","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-04 05:57:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e5dce42795a30ab190b8715da845d82685d9556f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 05:58:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"00a4c71723e8594040bd05f232ad8176eec53bf0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 07:59:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b127c67bcb044adb0500f97bffe8bf4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f4a6fd9a4ea4377ad0237289f7afbe1 : SUCCESS in 5m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/72e9cee3e8734530ac271ae4a148d601 : SUCCESS in 6m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0dd92090fc49408da50da2acd51fecb4 : SUCCESS in 7m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5b06b216b55428d83819721cc866036 : SUCCESS in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ee805a14bcb4fb88ff73e3b7a9ef888 : SUCCESS in 11m 20s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/8feaed5942bf4e0f880c1f4390663bd3 : SUCCESS in 1h 24m 23s\n- grenade-heat-multinode-skip-level https://zuul.opendev.org/t/openstack/build/eeda29da329845acb81c6fb4ec355af4 : SUCCESS in 1h 01m 57s\n- heat-functional https://zuul.opendev.org/t/openstack/build/5e2db5dbae664b4c9b3eba5b16d9e796 : SUCCESS in 1h 18m 22s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/279bfcf1648645d8a4b1afa950eb91ee : SUCCESS in 1h 18m 14s","accounts_in_message":[],"_revision_number":1},{"id":"2dfd6b2a59838537727e9b487370cd867d40f07e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 07:59:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"418b90bac0b94e8cad3cc779f081ba4ab2038057","revisions":{"418b90bac0b94e8cad3cc779f081ba4ab2038057":{"kind":"REWORK","_number":1,"created":"2026-08-03 17:03:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/999633/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/33/999633/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/33/999633/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/33/999633/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/33/999633/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/33/999633/1"}}},"commit":{"parents":[{"commit":"5aaeba874afb3cb2614898174389cb2ecb978f41","subject":"devstack: Maintain hidden_stack_tag(s) option by plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/5aaeba874afb3cb2614898174389cb2ecb978f41"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-22 02:57:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-03 17:03:00.000000000","tz":540},"subject":"devstack: Set image_ref and instance_type","message":"devstack: Set image_ref and instance_type\n\nThe required resources are created by the plugin so we should set\nthe options to use these resources.\n\nChange-Id: I27ee7511417ac2548729e2c91c93c756f9300668\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n(cherry picked from commit 5a781a9c01e5ddfd152e6816dbfdd916e376036b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/418b90bac0b94e8cad3cc779f081ba4ab2038057"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/418b90bac0b94e8cad3cc779f081ba4ab2038057"}]},"branch":"refs/heads/stable/2026.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
