)]}'
{"id":"openstack%2Fhorizon~1005821","triplet_id":"openstack%2Fhorizon~master~Iaa21253d769168c68302d9f8456d30c107f2fed7","project":"openstack/horizon","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"Iaa21253d769168c68302d9f8456d30c107f2fed7","subject":"Imported Translations from Zanata","status":"MERGED","created":"2026-09-16 03:52:20.000000000","updated":"2026-09-16 21:41:41.000000000","submitted":"2026-09-16 21:37:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005821-zanata/translations","meta_rev_id":"e38efbb841212937d0d1e1ff6bab6143c8291e50","_number":1005821,"virtual_id_number":1005821,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-16 21:37:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-16 20:23:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-16 20:23:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-16 04:16:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-16 20:23:53.000000000","updated_by":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"reviewer":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"state":"REVIEWER"}],"messages":[{"id":"f84b68f34d97c4c2896bf73e7a6e59d59e858477","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-09-16 03:52:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a68e8343e4d087d54693fcd248e267a5f2d9d51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 04:16:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae20fabaf09145798a01e23d7050422d\n\n- horizon-cross-designate-dashboard-python https://zuul.opendev.org/t/openstack/build/4f42e583d29e485ba0ed13ed65f56a41 : SUCCESS in 2m 34s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/e4bed46331b448319e876b1feab1b693 : SUCCESS in 2m 16s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/05b5511c7eba41e1bc8a063e54306fbd : SUCCESS in 3m 49s (non-voting)\n- horizon-cross-octavia-dashboard-python https://zuul.opendev.org/t/openstack/build/c1441e4e30214769a4b7cdbeeab332a9 : SUCCESS in 2m 41s (non-voting)\n- horizon-cross-magnum-ui-python https://zuul.opendev.org/t/openstack/build/e7a88c0553ac46f19b9a51f3490c4d17 : SUCCESS in 2m 28s (non-voting)\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/ed769f0632a649d1ab485a768da77b36 : SUCCESS in 4m 20s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/7cef52d6615242fd928220e37e5e4347 : SUCCESS in 5m 56s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/e20315ce23f941a8a02b4fd4c4fcb1cb : SUCCESS in 4m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b1d4ef9efa1e4428a85661ab9ba71e25 : SUCCESS in 18m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f4a8f3c5a9c475e9313f3b4dbaac9c2 : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/15a0b3b37ea345399710db67802ecabc : SUCCESS in 8m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ec461d84b868432fb9a07c25f609022c : SUCCESS in 11m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/125e61464b6346e1a86053fc2deb1e03 : SUCCESS in 12m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2fa7bd61b450496b865cb00b123b966c : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":1},{"id":"ec4e980f9117e58522c0dcab0299a0efe90b013a","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2026-09-16 20:23:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a78d3ce8800938e1b2c184d17620fd09f4cefdf2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 20:24:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"65e626e019a57062ab9c37dbe2eb49722cf401db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 21:37:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19ab303e8b2b426d9338ab2fe06eb7a8\n\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/75ff57e78b0f475a97d4031ac5fe75d0 : SUCCESS in 2m 48s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/172b4fa2ead348c88452d6c9a9c5f2a8 : SUCCESS in 13m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/53b2bec8996e4f1692872400a4809223 : SUCCESS in 18m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6db057b1120419e849489ef75111334 : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2bacac8eb01c49e9b0525b053d7c58af : SUCCESS in 9m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4696f2d96311416b86d29047c071f4a6 : SUCCESS in 14m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b2a6a0e79be40269a549c0cf65c3cc4 : SUCCESS in 5m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d0f8bae9132d47609a9af76f18a60dea : SUCCESS in 10m 45s","accounts_in_message":[],"_revision_number":1},{"id":"86a8582ffff5714372d2d6347af3e86bc81eaca5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 21:37:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e38efbb841212937d0d1e1ff6bab6143c8291e50","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 21:41:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9dcf690859445769000ff2b2560b873\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/017604edeba1444fbc511ad44c1973ed : SUCCESS in 1m 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"269fc85be484609a3d9ccc2ba616e8779edf4d16","revisions":{"269fc85be484609a3d9ccc2ba616e8779edf4d16":{"kind":"REWORK","_number":1,"created":"2026-09-16 03:52:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/21/1005821/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/21/1005821/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/21/1005821/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/21/1005821/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/21/1005821/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/21/1005821/1"}}},"commit":{"parents":[{"commit":"f2543d02815e3afa00109be373e2ef2b1816ee62","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/f2543d02815e3afa00109be373e2ef2b1816ee62"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-16 03:52:18.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-16 03:52:18.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Iaa21253d769168c68302d9f8456d30c107f2fed7\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/openstack-zuul-jobs:roles/prepare-zanata-client/files/common_translation_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/269fc85be484609a3d9ccc2ba616e8779edf4d16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/269fc85be484609a3d9ccc2ba616e8779edf4d16"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
