)]}'
{"id":"openstack%2Fhorizon~1007143","triplet_id":"openstack%2Fhorizon~master~I0429219dca2a998dd02fdc16a9ec79e50ecd5caa","project":"openstack/horizon","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-09-29 15:53:17.000000000","reason":"Change was submitted"},"32432":{"account":{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},"last_update":"2026-09-29 15:53:17.000000000","reason":"Change was submitted"},"6914":{"account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"last_update":"2026-09-29 14:13:59.000000000","reason":"\u003cGERRIT_ACCOUNT_6914\u003e replied on the change","reason_account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"}},"35133":{"account":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"last_update":"2026-09-25 11:19:05.000000000","reason":"\u003cGERRIT_ACCOUNT_35133\u003e replied on the change","reason_account":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}}},"hashtags":[],"change_id":"I0429219dca2a998dd02fdc16a9ec79e50ecd5caa","subject":"Fix truncated labels in key pair details row","status":"MERGED","created":"2026-09-24 12:22:38.000000000","updated":"2026-09-29 15:57:00.000000000","submitted":"2026-09-29 15:53:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1007143","meta_rev_id":"e4f212b8b12f8aa232757358b91353b8253b442d","_number":1007143,"virtual_id_number":1007143,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-29 15:53:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"all":[{"value":2,"date":"2026-09-29 14:13:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-25 11:19:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},{"value":1,"date":"2026-09-25 02:47:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"all":[{"value":1,"date":"2026-09-29 14:13:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"}],"reviewers":{"REVIEWER":[{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-24 13:59:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-25 02:47:09.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-09-25 02:47:09.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"state":"REVIEWER"},{"updated":"2026-09-25 02:47:09.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"state":"REVIEWER"},{"updated":"2026-09-25 02:47:09.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},"state":"REVIEWER"}],"messages":[{"id":"0848e35308273e1d490706c87871626768def78a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-09-24 12:22:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ff1e5c100228941f10342dfb0b7a3c40fa6b725","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-09-24 12:24:50.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"916a2e73011f9e7dac2b59c278c2041ea7fc05f4","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-09-24 12:24:52.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"55840e5e7a4602fb22d9137b863ad3c978dc7a75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 13:59:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a174b628633247efb4ea3c85185f258f\n\n- horizon-cross-designate-dashboard-python https://zuul.opendev.org/t/openstack/build/38c28da93241419aaf22999ea4ec76ae : SUCCESS in 3m 32s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/0f4032ff461e423f9805d12c7c3b15eb : SUCCESS in 5m 29s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/2d6188704e07465c8569ce414ff06f3a : SUCCESS in 4m 55s (non-voting)\n- horizon-cross-octavia-dashboard-python https://zuul.opendev.org/t/openstack/build/31c166c2e2244f06ba980924981751fb : SUCCESS in 3m 34s (non-voting)\n- horizon-cross-magnum-ui-python https://zuul.opendev.org/t/openstack/build/a0f6bee4e37c475ea973224df6540105 : SUCCESS in 3m 24s (non-voting)\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/a44c5c82142f4511b8bf2223e1a12565 : SUCCESS in 4m 07s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/6806a75418cf4a8ea5c71cba77e4ac9a : SUCCESS in 6m 06s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/b1c52b9781164d21985679e2319d4c40 : SUCCESS in 7m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d5c42bebad3e4922b1d98b2216fd1935 : SUCCESS in 18m 59s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/2038b30a3d4647d5a34f3ae014e480ac : SUCCESS in 8m 59s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/6570e4ecc2cc47ceaec450888b5e22c2 : SUCCESS in 18m 11s\n- horizon-tox-python3-django60 https://zuul.opendev.org/t/openstack/build/9c8bce00e38c4335b80a06155e46ee04 : SUCCESS in 20m 01s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e8f083bc2cf42ec9aae9f0e7b195a63 : SUCCESS in 4m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/edcefad5f9774d9587e2e20a75269d96 : SUCCESS in 9m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bb48a006313b41c89c117d97506fe6b2 : SUCCESS in 23m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/311884a92efe4b4ebd47bb450b5c21b9 : SUCCESS in 9m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e933f62359254b51968a6405b8afb26b : SUCCESS in 13m 26s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/e7186d2fb5014eccb55068de074ce2db : SUCCESS in 6m 21s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/1d6ce5ea0c62421698626088c0a47641 : SUCCESS in 1h 20m 28s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/139b1fdeb9144cfd901b0bd4e718b7fb : SUCCESS in 31m 32s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/99085eea0d2c41f1b1c10f0da043e9a2 : SUCCESS in 23m 00s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b47f462a46114378b7281f420a8d4ca2 : SUCCESS in 4m 16s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/68cececf4bdb4bbca02973a16d1e5e45 : SUCCESS in 37m 24s","accounts_in_message":[],"_revision_number":2},{"id":"a3c0d36fd43e29271d0656eb89dfca23133f905a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-25 02:47:09.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"054645d38084eb052e15164ebb3ddb5e473ca893","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-25 03:22:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7e32501e44616f048e80f947772aaf9f6f3f2c2a","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2026-09-25 11:19:05.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b750dd91d161d2d06f5d398bc2415f1543fcf5eb","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2026-09-29 14:13:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8b65afbb9fa37e6c1e98ae915200bdb8f2503094","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 14:14:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d49092ecf0361f8bdd204cf683aa52a1ece9ab14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 15:53:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d14b4a503c3447d9b6f7b14b62112050\n\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/c8092f6c9c064568bd0f87ac5ddf808d : SUCCESS in 6m 38s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/8b4d51fdec9d49f0b12a921863d6664f : SUCCESS in 9m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a26428567e90497aae57e65fd6a195fb : SUCCESS in 20m 25s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/d589da5112074cf9a280e5e1a6115857 : SUCCESS in 15m 42s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/ab1d878bae66414e8b74c2c39020fc2b : SUCCESS in 19m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd8337a16e9e4c61b1b412e56524d5b3 : SUCCESS in 5m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2a242553e1bb43bcb1dd3220d8f3bc53 : SUCCESS in 19m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3ae94495efac4f76a3fce4afd2fd9174 : SUCCESS in 23m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e18818b5955642a4b9f02de231cb350d : SUCCESS in 17m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c56413f48e346cbae9f1629605de233 : SUCCESS in 13m 06s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/bdbead20139142b9af9751228bd5ff6e : SUCCESS in 29m 55s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/e05b962148524dbfa3890a7607b2daf8 : SUCCESS in 6m 14s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/82540b1fad9a449094f6a8601199b347 : SUCCESS in 1h 29m 00s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/b8e2d30b87f54b31a5db1ebb3c0f410f : SUCCESS in 30m 57s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/e3a3d175e6544ece9c1f02eb086def84 : SUCCESS in 38m 15s","accounts_in_message":[],"_revision_number":2},{"id":"7154c943c1f93dfa159c751fa6daee57a3118f9e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 15:53:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e4f212b8b12f8aa232757358b91353b8253b442d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 15:57:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66490eabf7684c3c8fe8aa324fa6c0da\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a01614e3d50642cdb6dec3fbbb68290e : SUCCESS in 1m 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ec968048a669b4532a9f9cad66d023c7a2c36142","revisions":{"12680bb5cb0cd20ed91539ea5053b0a0d70ca47a":{"kind":"REWORK","_number":1,"created":"2026-09-24 12:22:38.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/43/1007143/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/43/1007143/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/43/1007143/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/43/1007143/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/43/1007143/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/43/1007143/1"}}},"commit":{"parents":[{"commit":"5131ca09edb17fe238d836982038bc334d4a47d2","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/5131ca09edb17fe238d836982038bc334d4a47d2"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-09-24 12:22:38.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-09-24 12:22:38.000000000","tz":0},"subject":"Fix truncated labels in key pair details row","message":"Fix truncated labels in key pair details row\n\nIn some localizations, long translated labels such as\n \"Public Key\" were cut off with an ellipsis in the\nexpanded key pair row.\n\nChange-Id: I0429219dca2a998dd02fdc16a9ec79e50ecd5caa\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/12680bb5cb0cd20ed91539ea5053b0a0d70ca47a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/12680bb5cb0cd20ed91539ea5053b0a0d70ca47a"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"ec968048a669b4532a9f9cad66d023c7a2c36142":{"kind":"REWORK","_number":2,"created":"2026-09-24 12:24:50.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/43/1007143/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/43/1007143/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/43/1007143/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/43/1007143/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/43/1007143/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/43/1007143/2"}}},"commit":{"parents":[{"commit":"5131ca09edb17fe238d836982038bc334d4a47d2","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/5131ca09edb17fe238d836982038bc334d4a47d2"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-09-24 12:22:38.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-09-24 12:24:38.000000000","tz":0},"subject":"Fix truncated labels in key pair details row","message":"Fix truncated labels in key pair details row\n\nIn some localizations, long translated labels such as\n \"Public Key\" were cut off with an ellipsis in the\nexpanded key pair row.\n\nChange-Id: I0429219dca2a998dd02fdc16a9ec79e50ecd5caa\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/ec968048a669b4532a9f9cad66d023c7a2c36142"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/ec968048a669b4532a9f9cad66d023c7a2c36142"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
