)]}'
{"id":"openstack%2Fhorizon~56821","triplet_id":"openstack%2Fhorizon~master~If901008171d964bf93941c0b6de2c66e20f0aa27","project":"openstack/horizon","branch":"master","topic":"1237377_hm_display_name","hashtags":[],"change_id":"If901008171d964bf93941c0b6de2c66e20f0aa27","subject":"Give HealthMonitor a proper display name","status":"MERGED","created":"2013-11-17 16:18:26.000000000","updated":"2013-11-25 09:35:55.000000000","submitted":"2013-11-25 09:35:55.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fff34a15c1a7d6ec512e8f65c7cda1fe29283302","_number":56821,"virtual_id_number":56821,"owner":{"_account_id":6447,"name":"Avishay Balderman","email":"avishayb@radware.com","username":"avishayb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":4978,"name":"Julie Pichon","email":"jpichon@redhat.com","username":"jpichon"},{"value":2,"date":"2013-11-25 09:35:55.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":6447,"name":"Avishay Balderman","email":"avishayb@radware.com","username":"avishayb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":0,"_account_id":4978,"name":"Julie Pichon","email":"jpichon@redhat.com","username":"jpichon"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2013-11-25 09:04:47.000000000","_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":2,"date":"2013-11-25 08:59:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"date":"2013-11-19 13:04:42.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":6447,"name":"Avishay Balderman","email":"avishayb@radware.com","username":"avishayb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":0,"date":"2013-11-20 14:20:01.000000000","_account_id":4978,"name":"Julie Pichon","email":"jpichon@redhat.com","username":"jpichon"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2013-11-25 09:04:47.000000000","_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"date":"2013-11-25 08:59:26.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":6447,"name":"Avishay Balderman","email":"avishayb@radware.com","username":"avishayb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4978,"name":"Julie Pichon","email":"jpichon@redhat.com","username":"jpichon"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":6447,"name":"Avishay Balderman","email":"avishayb@radware.com","username":"avishayb"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":4978,"name":"Julie Pichon","email":"jpichon@redhat.com","username":"jpichon"},{"_account_id":6447,"name":"Avishay Balderman","email":"avishayb@radware.com","username":"avishayb"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-11-20 14:20:01.000000000","updated_by":{"_account_id":4978,"name":"Julie Pichon","email":"jpichon@redhat.com","username":"jpichon"},"reviewer":{"_account_id":4978,"name":"Julie Pichon","email":"jpichon@redhat.com","username":"jpichon"},"state":"REVIEWER"},{"updated":"2013-11-25 08:59:26.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2013-11-25 09:04:47.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2013-11-25 09:35:55.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"95b83f2593db7e453930f883f1042bf91e413561","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-17 17:06:11.000000000","message":"Patch Set 1: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-horizon-pep8 http://logs.openstack.org/21/56821/1/check/gate-horizon-pep8/5ce6a7e : SUCCESS in 2m 02s\n- gate-horizon-docs http://docs-draft.openstack.org/21/56821/1/check/gate-horizon-docs/5ed6d4c/doc/build/html/ : SUCCESS in 8m 17s\n- gate-horizon-python26 http://logs.openstack.org/21/56821/1/check/gate-horizon-python26/b0454de : FAILURE in 4m 59s\n- gate-horizon-python27 http://logs.openstack.org/21/56821/1/check/gate-horizon-python27/06bd199 : FAILURE in 3m 58s\n- gate-horizon-python27-django14 http://logs.openstack.org/21/56821/1/check/gate-horizon-python27-django14/4dd1b20 : FAILURE in 4m 54s\n- gate-horizon-selenium http://logs.openstack.org/21/56821/1/check/gate-horizon-selenium/e34ae5e : SUCCESS in 2m 25s\n- check-tempest-devstack-vm-full http://logs.openstack.org/21/56821/1/check/check-tempest-devstack-vm-full/949955a : SUCCESS in 47m 37s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/21/56821/1/check/check-tempest-devstack-vm-postgres-full/39ee33c : SUCCESS in 33m 53s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/21/56821/1/check/check-tempest-devstack-vm-neutron/5bc963a : SUCCESS in 25m 42s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/21/56821/1/check/gate-tempest-devstack-vm-large-ops/1cb6d56 : SUCCESS in 13m 29s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/21/56821/1/check/gate-tempest-devstack-vm-neutron-large-ops/77f3dcd : SUCCESS in 17m 09s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/21/56821/1/check/check-grenade-devstack-vm/a0f370c : SUCCESS in 40m 18s\n","accounts_in_message":[],"_revision_number":1},{"id":"c93d1df54f0ea4fc5febc94af35f06ab8bc7942c","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2013-11-17 19:16:37.000000000","message":"Patch Set 1: I would prefer that you didn\u0027t merge this\n\n(4 inline comments)\n\n","accounts_in_message":[],"_revision_number":1},{"id":"6203c5ca5534caad39de74447dcf1057a59bf067","author":{"_account_id":6447,"name":"Avishay Balderman","email":"avishayb@radware.com","username":"avishayb"},"date":"2013-11-18 09:06:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"66490d9162ea7b1cbbe957a2d94ef5d214f4c15a","author":{"_account_id":6447,"name":"Avishay Balderman","email":"avishayb@radware.com","username":"avishayb"},"date":"2013-11-18 09:06:36.000000000","message":"Patch Set 1: (4 inline comments)\n\n","accounts_in_message":[],"_revision_number":1},{"id":"6826e36ccbb073904eb5c3de778e36eeb2df5ef7","author":{"_account_id":6447,"name":"Avishay Balderman","email":"avishayb@radware.com","username":"avishayb"},"date":"2013-11-18 09:21:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5928fa6113013679a7155f63a732609b1911c59d","author":{"_account_id":5494,"name":"Trivial Rebase","email":"trivial-rebase@review.openstack.org","username":"trivial-rebase"},"date":"2013-11-18 09:21:30.000000000","message":"Patch Set 3:\n\nNew patchset patch-id matches previous patchset, but commit message has changed.","accounts_in_message":[],"_revision_number":3},{"id":"ec8f97875b62671a29438faf29168fdc9ca04fe5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-18 12:52:40.000000000","message":"Patch Set 3: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-horizon-pep8 http://logs.openstack.org/21/56821/3/check/gate-horizon-pep8/77e9de7 : FAILURE in 1m 49s\n- gate-horizon-docs http://docs-draft.openstack.org/21/56821/3/check/gate-horizon-docs/96db77a/doc/build/html/ : SUCCESS in 9m 32s\n- gate-horizon-python26 http://logs.openstack.org/21/56821/3/check/gate-horizon-python26/dab06ae : FAILURE in 4m 07s\n- gate-horizon-python27 http://logs.openstack.org/21/56821/3/check/gate-horizon-python27/e1493ad : FAILURE in 5m 15s\n- gate-horizon-python27-django14 http://logs.openstack.org/21/56821/3/check/gate-horizon-python27-django14/59282a5 : FAILURE in 4m 02s\n- gate-horizon-selenium http://logs.openstack.org/21/56821/3/check/gate-horizon-selenium/50bec32 : SUCCESS in 2m 28s\n- check-tempest-devstack-vm-full http://logs.openstack.org/21/56821/3/check/check-tempest-devstack-vm-full/e5b0392 : FAILURE in 1h 02m 16s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/21/56821/3/check/check-tempest-devstack-vm-postgres-full/4ae5ee4 : SUCCESS in 48m 26s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/21/56821/3/check/check-tempest-devstack-vm-neutron/318ad7b : SUCCESS in 50m 28s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/21/56821/3/check/gate-tempest-devstack-vm-large-ops/dd9edfd : SUCCESS in 26m 36s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/21/56821/3/check/gate-tempest-devstack-vm-neutron-large-ops/b4fb57e : SUCCESS in 30m 45s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/21/56821/3/check/check-grenade-devstack-vm/dc64d87 : FAILURE in 51m 11s\n","accounts_in_message":[],"_revision_number":3},{"id":"ec4ddac5518c9ecd4fd756c071204b340d584570","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2013-11-19 11:47:50.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"22f5792cf75956799d74a159c62a41786fdc182c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-19 13:02:01.000000000","message":"Patch Set 4: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-horizon-pep8 http://logs.openstack.org/21/56821/4/check/gate-horizon-pep8/dac4cb6 : SUCCESS in 2m 04s\n- gate-horizon-docs http://docs-draft.openstack.org/21/56821/4/check/gate-horizon-docs/f3f0e6b/doc/build/html/ : SUCCESS in 10m 43s\n- gate-horizon-python26 http://logs.openstack.org/21/56821/4/check/gate-horizon-python26/d10b27d : SUCCESS in 4m 12s\n- gate-horizon-python27 http://logs.openstack.org/21/56821/4/check/gate-horizon-python27/7d6b1e0 : SUCCESS in 3m 28s\n- gate-horizon-python27-django14 http://logs.openstack.org/21/56821/4/check/gate-horizon-python27-django14/6e591f3 : SUCCESS in 4m 23s\n- gate-horizon-selenium http://logs.openstack.org/21/56821/4/check/gate-horizon-selenium/8da9cda : SUCCESS in 3m 38s\n- check-tempest-devstack-vm-full http://logs.openstack.org/21/56821/4/check/check-tempest-devstack-vm-full/37b8543 : SUCCESS in 39m 02s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/21/56821/4/check/check-tempest-devstack-vm-postgres-full/2434a45 : FAILURE in 43m 41s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/21/56821/4/check/check-tempest-devstack-vm-neutron/66b8399 : SUCCESS in 32m 30s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/21/56821/4/check/gate-tempest-devstack-vm-large-ops/733e305 : SUCCESS in 22m 08s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/21/56821/4/check/gate-tempest-devstack-vm-neutron-large-ops/0189b46 : SUCCESS in 21m 41s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/21/56821/4/check/check-grenade-devstack-vm/b83ab36 : FAILURE in 36m 27s\n","accounts_in_message":[],"_revision_number":4},{"id":"af04badff7f495a358d09f460084b4ce7a826c1a","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2013-11-19 13:02:40.000000000","message":"Patch Set 4:\n\nI noticed tempest failed, I think you hit bug(s): https://bugs.launchpad.net/bugs/1251920","accounts_in_message":[],"_revision_number":4},{"id":"887047d84f82e4d292a465c484b1e5defc6f5918","author":{"_account_id":6447,"name":"Avishay Balderman","email":"avishayb@radware.com","username":"avishayb"},"date":"2013-11-19 13:04:42.000000000","message":"Patch Set 4:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":4},{"id":"b81000b1a5501e0cc9c465b0ca6f6d63cb79b57b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-19 14:39:18.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-horizon-pep8 http://logs.openstack.org/21/56821/4/check/gate-horizon-pep8/0e1b233 : SUCCESS in 5m 02s\n- gate-horizon-docs http://docs-draft.openstack.org/21/56821/4/check/gate-horizon-docs/22dfe39/doc/build/html/ : SUCCESS in 9m 51s\n- gate-horizon-python26 http://logs.openstack.org/21/56821/4/check/gate-horizon-python26/b69f6ee : SUCCESS in 3m 20s\n- gate-horizon-python27 http://logs.openstack.org/21/56821/4/check/gate-horizon-python27/f42091d : SUCCESS in 4m 49s\n- gate-horizon-python27-django14 http://logs.openstack.org/21/56821/4/check/gate-horizon-python27-django14/fc9185e : SUCCESS in 5m 16s\n- gate-horizon-selenium http://logs.openstack.org/21/56821/4/check/gate-horizon-selenium/fdcacac : SUCCESS in 3m 13s\n- check-tempest-devstack-vm-full http://logs.openstack.org/21/56821/4/check/check-tempest-devstack-vm-full/718b9bb : SUCCESS in 1h 00m 18s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/21/56821/4/check/check-tempest-devstack-vm-postgres-full/42b41b2 : FAILURE in 55m 47s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/21/56821/4/check/check-tempest-devstack-vm-neutron/608e2da : SUCCESS in 44m 54s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/21/56821/4/check/gate-tempest-devstack-vm-large-ops/5c37cd5 : SUCCESS in 18m 38s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/21/56821/4/check/gate-tempest-devstack-vm-neutron-large-ops/fd044c3 : SUCCESS in 21m 33s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/21/56821/4/check/check-grenade-devstack-vm/b15f292 : FAILURE in 42m 59s\n","accounts_in_message":[],"_revision_number":4},{"id":"d127000d718d9db8af699bc3c2e92fe2849b53c7","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2013-11-19 14:39:42.000000000","message":"Patch Set 4:\n\nI noticed tempest failed, I think you hit bug(s): https://bugs.launchpad.net/bugs/1251784 and https://bugs.launchpad.net/bugs/1251920","accounts_in_message":[],"_revision_number":4},{"id":"9d9074a0369017c0c6181ad8b2b1d169ce3469b4","author":{"_account_id":6447,"name":"Avishay Balderman","email":"avishayb@radware.com","username":"avishayb"},"date":"2013-11-19 15:15:24.000000000","message":"Patch Set 4:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":4},{"id":"b081a0c87b80715ed604e659599d3c881cd6df0b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-19 16:53:27.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-horizon-pep8 http://logs.openstack.org/21/56821/4/check/gate-horizon-pep8/deb2034 : SUCCESS in 2m 36s\n- gate-horizon-docs http://docs-draft.openstack.org/21/56821/4/check/gate-horizon-docs/d09c35d/doc/build/html/ : SUCCESS in 8m 35s\n- gate-horizon-python26 http://logs.openstack.org/21/56821/4/check/gate-horizon-python26/b343f1b : SUCCESS in 4m 01s\n- gate-horizon-python27 http://logs.openstack.org/21/56821/4/check/gate-horizon-python27/91c1876 : SUCCESS in 3m 41s\n- gate-horizon-python27-django14 http://logs.openstack.org/21/56821/4/check/gate-horizon-python27-django14/81d1a38 : SUCCESS in 3m 57s\n- gate-horizon-selenium http://logs.openstack.org/21/56821/4/check/gate-horizon-selenium/34d29ec : SUCCESS in 3m 09s\n- check-tempest-devstack-vm-full http://logs.openstack.org/21/56821/4/check/check-tempest-devstack-vm-full/585ed34 : SUCCESS in 53m 02s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/21/56821/4/check/check-tempest-devstack-vm-postgres-full/d10bb2a : SUCCESS in 49m 46s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/21/56821/4/check/check-tempest-devstack-vm-neutron/20fa3c6 : FAILURE in 55m 24s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/21/56821/4/check/gate-tempest-devstack-vm-large-ops/9eb7b3c : SUCCESS in 30m 16s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/21/56821/4/check/gate-tempest-devstack-vm-neutron-large-ops/360fcf5 : SUCCESS in 37m 14s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/21/56821/4/check/check-grenade-devstack-vm/d29ab03 : FAILURE in 52m 14s\n","accounts_in_message":[],"_revision_number":4},{"id":"db881717d010c61e2a7af090e54964e4d187ee86","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2013-11-19 16:53:55.000000000","message":"Patch Set 4:\n\nI noticed tempest failed, I think you hit bug(s): https://bugs.launchpad.net/bugs/1251784","accounts_in_message":[],"_revision_number":4},{"id":"bf7f4ffd3d3a21e269f245af70d914ccc73e035e","author":{"_account_id":6447,"name":"Avishay Balderman","email":"avishayb@radware.com","username":"avishayb"},"date":"2013-11-20 08:37:09.000000000","message":"Patch Set 4:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":4},{"id":"36ea90c09a573b32e8bacca02e55cf04d7bc9d60","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-20 10:21:12.000000000","message":"Patch Set 4: Works for me\n\nBuild succeeded.\n\n- gate-horizon-pep8 http://logs.openstack.org/21/56821/4/check/gate-horizon-pep8/93a1c75 : SUCCESS in 2m 03s\n- gate-horizon-docs http://docs-draft.openstack.org/21/56821/4/check/gate-horizon-docs/8ea3019/doc/build/html/ : SUCCESS in 7m 56s\n- gate-horizon-python26 http://logs.openstack.org/21/56821/4/check/gate-horizon-python26/019376a : SUCCESS in 4m 16s\n- gate-horizon-python27 http://logs.openstack.org/21/56821/4/check/gate-horizon-python27/9c0fc4f : SUCCESS in 4m 08s\n- gate-horizon-python27-django14 http://logs.openstack.org/21/56821/4/check/gate-horizon-python27-django14/6aaf350 : SUCCESS in 2m 58s\n- gate-horizon-selenium http://logs.openstack.org/21/56821/4/check/gate-horizon-selenium/4f90f15 : SUCCESS in 2m 41s\n- check-tempest-devstack-vm-full http://logs.openstack.org/21/56821/4/check/check-tempest-devstack-vm-full/536f07e : SUCCESS in 59m 54s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/21/56821/4/check/check-tempest-devstack-vm-postgres-full/263955b : SUCCESS in 48m 11s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/21/56821/4/check/check-tempest-devstack-vm-neutron/53a5092 : SUCCESS in 41m 25s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/21/56821/4/check/gate-tempest-devstack-vm-large-ops/b43db7b : SUCCESS in 28m 01s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/21/56821/4/check/gate-tempest-devstack-vm-neutron-large-ops/b5eb46d : SUCCESS in 29m 37s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/21/56821/4/check/check-grenade-devstack-vm/114b9d6 : SUCCESS in 49m 17s\n","accounts_in_message":[],"_revision_number":4},{"id":"dddc7a280cdf269fb630be2e00e871ace1444cb2","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2013-11-25 08:59:26.000000000","message":"Patch Set 4: Looks good to me (core reviewer)\n\n","accounts_in_message":[],"_revision_number":4},{"id":"7302f74c24d23327e1b5bd66dd86bb96f5198238","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-25 09:04:47.000000000","message":"Patch Set 4: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":4},{"id":"0b9768fea7b1c887f5b3814ecf107a456651e781","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2013-11-25 09:04:47.000000000","message":"Patch Set 4: Looks good to me (core reviewer); Approved\n\n","accounts_in_message":[],"_revision_number":4},{"id":"0f0cc19003edd9c5f8537e8c24e695c3ba3bbcad","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-25 09:35:55.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":4},{"id":"2578c02e60f7546480e1da581c1ace8838013828","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-11-25 09:35:55.000000000","message":"Patch Set 4: Verified\n\nBuild succeeded.\n\n- gate-horizon-docs http://docs-draft.openstack.org/21/56821/4/gate/gate-horizon-docs/c1b4042/doc/build/html/ : SUCCESS in 9m 11s\n- gate-horizon-pep8 http://logs.openstack.org/21/56821/4/gate/gate-horizon-pep8/564ce99 : SUCCESS in 2m 08s\n- gate-horizon-python26 http://logs.openstack.org/21/56821/4/gate/gate-horizon-python26/4e71b0b : SUCCESS in 4m 57s\n- gate-horizon-python27 http://logs.openstack.org/21/56821/4/gate/gate-horizon-python27/9434161 : SUCCESS in 2m 38s\n- gate-horizon-python27-django14 http://logs.openstack.org/21/56821/4/gate/gate-horizon-python27-django14/18b099f : SUCCESS in 3m 44s\n- gate-horizon-selenium http://logs.openstack.org/21/56821/4/gate/gate-horizon-selenium/44e8abe : SUCCESS in 2m 17s\n- gate-tempest-devstack-vm-full http://logs.openstack.org/21/56821/4/gate/gate-tempest-devstack-vm-full/617cc69 : SUCCESS in 30m 32s\n- gate-tempest-devstack-vm-postgres-full http://logs.openstack.org/21/56821/4/gate/gate-tempest-devstack-vm-postgres-full/6c0ed0d : SUCCESS in 31m 02s\n- gate-tempest-devstack-vm-neutron http://logs.openstack.org/21/56821/4/gate/gate-tempest-devstack-vm-neutron/6bbbfd9 : SUCCESS in 27m 19s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/21/56821/4/gate/gate-tempest-devstack-vm-large-ops/13a2827 : SUCCESS in 13m 41s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/21/56821/4/gate/gate-tempest-devstack-vm-neutron-large-ops/98ed9d0 : SUCCESS in 15m 56s\n- gate-grenade-devstack-vm http://logs.openstack.org/21/56821/4/gate/gate-grenade-devstack-vm/b74693c : SUCCESS in 23m 48s\n","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3a28423cd96596edd5e30a1a53956fa6c93f439f","revisions":{"c3b3b3e426bb399d6b6cc1da848225859a33df64":{"kind":"REWORK","_number":1,"created":"2013-11-17 16:18:26.000000000","uploader":{"_account_id":6447,"name":"Avishay Balderman","email":"avishayb@radware.com","username":"avishayb"},"ref":"refs/changes/21/56821/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/21/56821/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/21/56821/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/21/56821/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/21/56821/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/21/56821/1"}}},"commit":{"parents":[{"commit":"ce1ef5c4413d6119bd5fbd17624204ec2153351f","subject":"Merge \"fix help text in \"Create An image\" window\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/ce1ef5c4413d6119bd5fbd17624204ec2153351f"}]}],"author":{"name":"Avishay Balderman","email":"avishayb@radware.com","date":"2013-11-17 16:16:45.000000000","tz":120},"committer":{"name":"Avishay Balderman","email":"avishayb@radware.com","date":"2013-11-17 16:18:27.000000000","tz":120},"subject":"Give HealthMonitor a proper display name Closes-Bug: #1237377","message":"Give HealthMonitor a proper display name\nCloses-Bug: #1237377\n\nChange-Id: If901008171d964bf93941c0b6de2c66e20f0aa27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/c3b3b3e426bb399d6b6cc1da848225859a33df64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/c3b3b3e426bb399d6b6cc1da848225859a33df64"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"ce1ef5c4413d6119bd5fbd17624204ec2153351f","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"5a637c78009c12f33a0bcb014c195272e0e74117":{"kind":"REWORK","_number":2,"created":"2013-11-18 09:06:11.000000000","uploader":{"_account_id":6447,"name":"Avishay Balderman","email":"avishayb@radware.com","username":"avishayb"},"ref":"refs/changes/21/56821/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/21/56821/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/21/56821/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/21/56821/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/21/56821/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/21/56821/2"}}},"commit":{"parents":[{"commit":"1dd9ec17ed81873057c3423b8632d13d159da64a","subject":"Merge \"Documenting OPENSTACK_SSL_CACERT\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/1dd9ec17ed81873057c3423b8632d13d159da64a"}]}],"author":{"name":"Avishay Balderman","email":"avishayb@radware.com","date":"2013-11-17 16:16:45.000000000","tz":120},"committer":{"name":"Avishay Balderman","email":"avishayb@radware.com","date":"2013-11-18 09:06:10.000000000","tz":120},"subject":"Give HealthMonitor a proper display name Closes-Bug: #1237377","message":"Give HealthMonitor a proper display name\nCloses-Bug: #1237377\n\nChange-Id: If901008171d964bf93941c0b6de2c66e20f0aa27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/5a637c78009c12f33a0bcb014c195272e0e74117"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/5a637c78009c12f33a0bcb014c195272e0e74117"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1dd9ec17ed81873057c3423b8632d13d159da64a","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"58425055f21215a6f92a7526c6afc25d042f075f":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2013-11-18 09:21:23.000000000","uploader":{"_account_id":6447,"name":"Avishay Balderman","email":"avishayb@radware.com","username":"avishayb"},"ref":"refs/changes/21/56821/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/21/56821/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/21/56821/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/21/56821/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/21/56821/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/21/56821/3"}}},"commit":{"parents":[{"commit":"1dd9ec17ed81873057c3423b8632d13d159da64a","subject":"Merge \"Documenting OPENSTACK_SSL_CACERT\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/1dd9ec17ed81873057c3423b8632d13d159da64a"}]}],"author":{"name":"Avishay Balderman","email":"avishayb@radware.com","date":"2013-11-17 16:16:45.000000000","tz":120},"committer":{"name":"Avishay Balderman","email":"avishayb@radware.com","date":"2013-11-18 09:21:01.000000000","tz":120},"subject":"Give HealthMonitor a proper display name","message":"Give HealthMonitor a proper display name\n\nCloses-Bug: #1237377\n\nChange-Id: If901008171d964bf93941c0b6de2c66e20f0aa27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/58425055f21215a6f92a7526c6afc25d042f075f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/58425055f21215a6f92a7526c6afc25d042f075f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1dd9ec17ed81873057c3423b8632d13d159da64a","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"3a28423cd96596edd5e30a1a53956fa6c93f439f":{"kind":"REWORK","_number":4,"created":"2013-11-19 11:47:50.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/21/56821/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/21/56821/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/21/56821/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/21/56821/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/21/56821/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/21/56821/4"}}},"commit":{"parents":[{"commit":"1dd9ec17ed81873057c3423b8632d13d159da64a","subject":"Merge \"Documenting OPENSTACK_SSL_CACERT\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/1dd9ec17ed81873057c3423b8632d13d159da64a"}]}],"author":{"name":"Avishay Balderman","email":"avishayb@radware.com","date":"2013-11-17 16:16:45.000000000","tz":120},"committer":{"name":"Akihiro MOTOKI","email":"motoki@da.jp.nec.com","date":"2013-11-19 11:47:30.000000000","tz":540},"subject":"Give HealthMonitor a proper display name","message":"Give HealthMonitor a proper display name\n\nCloses-Bug: #1237377\n\nChange-Id: If901008171d964bf93941c0b6de2c66e20f0aa27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/3a28423cd96596edd5e30a1a53956fa6c93f439f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/3a28423cd96596edd5e30a1a53956fa6c93f439f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1dd9ec17ed81873057c3423b8632d13d159da64a","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
