)]}'
{"id":"openstack%2Fhorizon~653339","triplet_id":"openstack%2Fhorizon~master~I4607b3b6d90ce28ba1b63808fd2028755039dcde","project":"openstack/horizon","branch":"master","topic":"bug/1825102","hashtags":[],"change_id":"I4607b3b6d90ce28ba1b63808fd2028755039dcde","subject":"Disable textbox on create image form when submitting","status":"MERGED","created":"2019-04-17 07:38:35.000000000","updated":"2019-05-29 12:39:19.000000000","submitted":"2019-05-29 12:39:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"653339-1559133559594-e94487ec","meta_rev_id":"dcdd2b4a58ee6ec60d1fb3e2a6d2e649f8942fa7","_number":653339,"virtual_id_number":653339,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":2,"date":"2019-05-29 12:39:19.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27679,"name":"gaoruyue","email":"gaory@awcloud.com","username":"gaoruyue"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":2,"date":"2019-05-29 10:34:42.000000000","_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":2,"date":"2019-05-29 11:38:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-29 10:35:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27679,"name":"gaoruyue","email":"gaory@awcloud.com","username":"gaoruyue"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":1,"date":"2019-05-29 11:38:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27679,"name":"gaoruyue","email":"gaory@awcloud.com","username":"gaoruyue"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27679,"name":"gaoruyue","email":"gaory@awcloud.com","username":"gaoruyue"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27679,"name":"gaoruyue","email":"gaory@awcloud.com","username":"gaoruyue"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-29 10:34:42.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2019-05-29 10:35:00.000000000","updated_by":{"_account_id":27679,"name":"gaoruyue","email":"gaory@awcloud.com","username":"gaoruyue"},"reviewer":{"_account_id":27679,"name":"gaoruyue","email":"gaory@awcloud.com","username":"gaoruyue"},"state":"REVIEWER"},{"updated":"2019-05-29 11:38:28.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-05-29 12:39:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c0f2e64872a403c57f9bea5c6779a6640d111e17","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-04-17 07:38:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f454c3285d7ab2be9e842c2fb7f71ae4001910c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 09:07:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint http://logs.openstack.org/39/653339/1/check/nodejs10-npm-run-lint/3b15b4b/ : SUCCESS in 5m 47s\n- nodejs10-npm-run-test http://logs.openstack.org/39/653339/1/check/nodejs10-npm-run-test/e8c68e4/ : SUCCESS in 7m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/39/653339/1/check/openstack-tox-lower-constraints/b4a1e8b/ : SUCCESS in 10m 03s\n- openstack-tox-pep8 http://logs.openstack.org/39/653339/1/check/openstack-tox-pep8/7544c02/ : SUCCESS in 6m 27s\n- openstack-tox-py27 http://logs.openstack.org/39/653339/1/check/openstack-tox-py27/327157d/ : SUCCESS in 11m 34s\n- openstack-tox-py36 http://logs.openstack.org/39/653339/1/check/openstack-tox-py36/3778adc/ : SUCCESS in 10m 15s\n- openstack-tox-py37 http://logs.openstack.org/39/653339/1/check/openstack-tox-py37/027ad72/ : SUCCESS in 10m 36s\n- openstack-tox-docs http://logs.openstack.org/39/653339/1/check/openstack-tox-docs/1a13931/html/ : SUCCESS in 4m 19s\n- horizon-openstack-tox-python3-django111 http://logs.openstack.org/39/653339/1/check/horizon-openstack-tox-python3-django111/f046ff0/ : SUCCESS in 17m 24s\n- horizon-selenium-headless http://logs.openstack.org/39/653339/1/check/horizon-selenium-headless/ce78e4e/ : SUCCESS in 8m 19s\n- horizon-integration-tests http://logs.openstack.org/39/653339/1/check/horizon-integration-tests/57df24d/ : SUCCESS in 1h 00m 10s\n- horizon-dsvm-tempest-plugin http://logs.openstack.org/39/653339/1/check/horizon-dsvm-tempest-plugin/d4754c2/ : SUCCESS in 37m 29s\n- horizon-dsvm-tempest-plugin-py27 http://logs.openstack.org/39/653339/1/check/horizon-dsvm-tempest-plugin-py27/7b9c695/ : SUCCESS in 38m 08s\n- horizon-tox-bandit-baseline http://logs.openstack.org/39/653339/1/check/horizon-tox-bandit-baseline/bab5d8a/ : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":1},{"id":"225c18e9396c467809f8b2acceb0532932611786","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-05-29 10:34:42.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"15bc82cf53d0a0c910d4acbb09ba6a252da87ded","author":{"_account_id":27679,"name":"gaoruyue","email":"gaory@awcloud.com","username":"gaoruyue"},"date":"2019-05-29 10:35:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"50e88e2534278d420def2fe2a7e850e0012afcb2","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-05-29 11:38:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3a79875e019ae65823a9b5a87eac5721c3ddc391","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 11:38:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1735367f30a13452cf31382d34aca1287461e205","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 12:39:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e32b063e2bc9fea770b44380be10476d91917c6e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 12:39:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- nodejs10-npm-run-lint http://logs.openstack.org/39/653339/1/gate/nodejs10-npm-run-lint/508ca18/ : SUCCESS in 3m 36s\n- nodejs10-npm-run-test http://logs.openstack.org/39/653339/1/gate/nodejs10-npm-run-test/3353f90/ : SUCCESS in 5m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/39/653339/1/gate/openstack-tox-lower-constraints/11f85fd/ : SUCCESS in 10m 38s\n- openstack-tox-pep8 http://logs.openstack.org/39/653339/1/gate/openstack-tox-pep8/cea88f2/ : SUCCESS in 5m 22s\n- openstack-tox-py27 http://logs.openstack.org/39/653339/1/gate/openstack-tox-py27/5beeb36/ : SUCCESS in 10m 12s\n- openstack-tox-py36 http://logs.openstack.org/39/653339/1/gate/openstack-tox-py36/ed55022/ : SUCCESS in 9m 18s\n- openstack-tox-py37 http://logs.openstack.org/39/653339/1/gate/openstack-tox-py37/ca9db27/ : SUCCESS in 9m 17s\n- openstack-tox-docs http://logs.openstack.org/39/653339/1/gate/openstack-tox-docs/965e042/html/ : SUCCESS in 4m 24s\n- horizon-openstack-tox-python3-django111 http://logs.openstack.org/39/653339/1/gate/horizon-openstack-tox-python3-django111/5eba0f4/ : SUCCESS in 9m 07s\n- horizon-dsvm-tempest-plugin http://logs.openstack.org/39/653339/1/gate/horizon-dsvm-tempest-plugin/e45ad82/ : SUCCESS in 33m 20s\n- horizon-dsvm-tempest-plugin-py27 http://logs.openstack.org/39/653339/1/gate/horizon-dsvm-tempest-plugin-py27/0d7542a/ : SUCCESS in 33m 32s\n- horizon-selenium-headless http://logs.openstack.org/39/653339/1/gate/horizon-selenium-headless/2e05aff/ : SUCCESS in 5m 08s\n- horizon-integration-tests http://logs.openstack.org/39/653339/1/gate/horizon-integration-tests/68f8f17/ : SUCCESS in 58m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f8e5c4ef8e1eb1c55aa380864b1a023f3c4b22ce","revisions":{"f8e5c4ef8e1eb1c55aa380864b1a023f3c4b22ce":{"kind":"REWORK","_number":1,"created":"2019-04-17 07:38:35.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/39/653339/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/39/653339/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/39/653339/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/39/653339/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/39/653339/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/39/653339/1"}}},"commit":{"parents":[{"commit":"afd04dc92b4f0d910f43b95e6246abfee5a07ba4","subject":"Merge \"Drop nodejs4 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/afd04dc92b4f0d910f43b95e6246abfee5a07ba4"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-04-17 07:16:44.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-04-17 07:16:53.000000000","tz":480},"subject":"Disable textbox on create image form when submitting","message":"Disable textbox on create image form when submitting\n\nIn the process of creating the image,\ncan still edit the name, description,\nSource Type and file.\nThis patch is Disable textbox when submitting\n\nChange-Id: I4607b3b6d90ce28ba1b63808fd2028755039dcde\nCloses-Bug: #1825102\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/f8e5c4ef8e1eb1c55aa380864b1a023f3c4b22ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/f8e5c4ef8e1eb1c55aa380864b1a023f3c4b22ce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
