)]}'
{"id":"openstack%2Fhorizon~657183","triplet_id":"openstack%2Fhorizon~master~Ib075efd46af6c8f0d0d61b4ac76160d8de08b77d","project":"openstack/horizon","branch":"master","hashtags":[],"change_id":"Ib075efd46af6c8f0d0d61b4ac76160d8de08b77d","subject":"Add tooltips for \"Admin State\" in related forms","status":"MERGED","created":"2019-05-05 01:47:09.000000000","updated":"2019-05-28 08:25:32.000000000","submitted":"2019-05-28 08:25:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657183-1559031932580-c755068b","meta_rev_id":"5c0824b7bcbe0bc71b42114042d700790697bcf4","_number":657183,"virtual_id_number":657183,"owner":{"_account_id":27979,"name":"steven","email":"gaoyl@rc.inesa.com","username":"stevenGao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":27979,"name":"steven","email":"gaoyl@rc.inesa.com","username":"stevenGao"},{"value":2,"date":"2019-05-28 08:25:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":2,"date":"2019-05-28 04:45:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":27979,"name":"steven","email":"gaoyl@rc.inesa.com","username":"stevenGao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":1,"date":"2019-05-28 07:18:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":27979,"name":"steven","email":"gaoyl@rc.inesa.com","username":"stevenGao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":27979,"name":"steven","email":"gaoyl@rc.inesa.com","username":"stevenGao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27979,"name":"steven","email":"gaoyl@rc.inesa.com","username":"stevenGao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-05 03:51:21.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2019-05-28 07:18:24.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-05-28 08:25:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2bf4e5bd5a69380bb0a66171893002cca710979c","author":{"_account_id":27979,"name":"steven","email":"gaoyl@rc.inesa.com","username":"stevenGao"},"date":"2019-05-05 01:47:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"05354e48d3281c9bc49c799e7fb5f74b8d3e6877","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 02:47:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- nodejs10-npm-run-lint http://logs.openstack.org/83/657183/1/check/nodejs10-npm-run-lint/0a0ec4c/ : SUCCESS in 4m 00s\n- nodejs10-npm-run-test http://logs.openstack.org/83/657183/1/check/nodejs10-npm-run-test/0239ce8/ : SUCCESS in 6m 13s\n- openstack-tox-lower-constraints http://logs.openstack.org/83/657183/1/check/openstack-tox-lower-constraints/2edf79a/ : SUCCESS in 9m 13s\n- openstack-tox-pep8 http://logs.openstack.org/83/657183/1/check/openstack-tox-pep8/10eba71/ : FAILURE in 4m 28s\n- openstack-tox-py27 http://logs.openstack.org/83/657183/1/check/openstack-tox-py27/458994f/ : SUCCESS in 9m 16s\n- openstack-tox-py36 http://logs.openstack.org/83/657183/1/check/openstack-tox-py36/daa5a40/ : SUCCESS in 7m 42s\n- openstack-tox-py37 http://logs.openstack.org/83/657183/1/check/openstack-tox-py37/9f51015/ : SUCCESS in 8m 09s\n- openstack-tox-docs http://logs.openstack.org/83/657183/1/check/openstack-tox-docs/827f63b/html/ : SUCCESS in 3m 12s\n- horizon-openstack-tox-python3-django111 http://logs.openstack.org/83/657183/1/check/horizon-openstack-tox-python3-django111/e343ea2/ : SUCCESS in 7m 42s\n- horizon-selenium-headless http://logs.openstack.org/83/657183/1/check/horizon-selenium-headless/280c5d5/ : SUCCESS in 7m 22s\n- horizon-integration-tests http://logs.openstack.org/83/657183/1/check/horizon-integration-tests/0f8cde8/ : SUCCESS in 57m 13s\n- horizon-dsvm-tempest-plugin http://logs.openstack.org/83/657183/1/check/horizon-dsvm-tempest-plugin/0b86bc4/ : SUCCESS in 29m 22s\n- horizon-dsvm-tempest-plugin-py27 http://logs.openstack.org/83/657183/1/check/horizon-dsvm-tempest-plugin-py27/c0dd590/ : SUCCESS in 35m 29s\n- horizon-tox-bandit-baseline http://logs.openstack.org/83/657183/1/check/horizon-tox-bandit-baseline/6dafe94/ : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":1},{"id":"d42083110185d2b03563f7926f570f07802fc5b7","author":{"_account_id":27979,"name":"steven","email":"gaoyl@rc.inesa.com","username":"stevenGao"},"date":"2019-05-05 03:07:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9b9e8eb6ce0fb60632d36a987cf6863195b50f8c","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-05-05 03:51:21.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3a05c9902b3f42b8ee955b6d3ba26cb859f77fa4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 04:08:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint http://logs.openstack.org/83/657183/2/check/nodejs10-npm-run-lint/a06b69e/ : SUCCESS in 4m 40s\n- nodejs10-npm-run-test http://logs.openstack.org/83/657183/2/check/nodejs10-npm-run-test/85c0286/ : SUCCESS in 6m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/83/657183/2/check/openstack-tox-lower-constraints/bef90c6/ : SUCCESS in 11m 20s\n- openstack-tox-pep8 http://logs.openstack.org/83/657183/2/check/openstack-tox-pep8/9ea48f4/ : SUCCESS in 6m 19s\n- openstack-tox-py27 http://logs.openstack.org/83/657183/2/check/openstack-tox-py27/e513c74/ : SUCCESS in 10m 42s\n- openstack-tox-py36 http://logs.openstack.org/83/657183/2/check/openstack-tox-py36/98e2b86/ : SUCCESS in 9m 16s\n- openstack-tox-py37 http://logs.openstack.org/83/657183/2/check/openstack-tox-py37/0a6bd6f/ : SUCCESS in 8m 47s\n- openstack-tox-docs http://logs.openstack.org/83/657183/2/check/openstack-tox-docs/6e1f7d7/html/ : SUCCESS in 4m 38s\n- horizon-openstack-tox-python3-django111 http://logs.openstack.org/83/657183/2/check/horizon-openstack-tox-python3-django111/14ee942/ : SUCCESS in 8m 27s\n- horizon-selenium-headless http://logs.openstack.org/83/657183/2/check/horizon-selenium-headless/0956e51/ : SUCCESS in 6m 49s\n- horizon-integration-tests http://logs.openstack.org/83/657183/2/check/horizon-integration-tests/6b92e87/ : SUCCESS in 57m 34s\n- horizon-dsvm-tempest-plugin http://logs.openstack.org/83/657183/2/check/horizon-dsvm-tempest-plugin/53f60de/ : SUCCESS in 41m 43s\n- horizon-dsvm-tempest-plugin-py27 http://logs.openstack.org/83/657183/2/check/horizon-dsvm-tempest-plugin-py27/52fde66/ : SUCCESS in 33m 24s\n- horizon-tox-bandit-baseline http://logs.openstack.org/83/657183/2/check/horizon-tox-bandit-baseline/b76c716/ : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":2},{"id":"f0fc076b29ef0736e2ef3d37f6b290b977137890","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-05-05 18:00:27.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4d3fa40a9fabdf7580376996b0ac121a77d1a73e","author":{"_account_id":27979,"name":"steven","email":"gaoyl@rc.inesa.com","username":"stevenGao"},"date":"2019-05-06 02:27:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8bfd5907ad532c330c5c7aa705d78334a1453853","author":{"_account_id":27979,"name":"steven","email":"gaoyl@rc.inesa.com","username":"stevenGao"},"date":"2019-05-06 02:27:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b9d41c3382259ff22971118bb45a4541ddf39f0b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 03:24:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint http://logs.openstack.org/83/657183/3/check/nodejs10-npm-run-lint/4bf13d9/ : SUCCESS in 3m 24s\n- nodejs10-npm-run-test http://logs.openstack.org/83/657183/3/check/nodejs10-npm-run-test/74d589d/ : SUCCESS in 7m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/83/657183/3/check/openstack-tox-lower-constraints/426c943/ : SUCCESS in 10m 05s\n- openstack-tox-pep8 http://logs.openstack.org/83/657183/3/check/openstack-tox-pep8/96b0c34/ : SUCCESS in 7m 29s\n- openstack-tox-py27 http://logs.openstack.org/83/657183/3/check/openstack-tox-py27/c306b57/ : SUCCESS in 10m 32s\n- openstack-tox-py36 http://logs.openstack.org/83/657183/3/check/openstack-tox-py36/fcc2363/ : SUCCESS in 8m 05s\n- openstack-tox-py37 http://logs.openstack.org/83/657183/3/check/openstack-tox-py37/2d561d5/ : SUCCESS in 8m 30s\n- openstack-tox-docs http://logs.openstack.org/83/657183/3/check/openstack-tox-docs/480e73f/html/ : SUCCESS in 3m 55s\n- horizon-openstack-tox-python3-django111 http://logs.openstack.org/83/657183/3/check/horizon-openstack-tox-python3-django111/f9c58a0/ : SUCCESS in 8m 12s\n- horizon-selenium-headless http://logs.openstack.org/83/657183/3/check/horizon-selenium-headless/805146d/ : SUCCESS in 5m 30s\n- horizon-integration-tests http://logs.openstack.org/83/657183/3/check/horizon-integration-tests/ec5b071/ : SUCCESS in 52m 24s\n- horizon-dsvm-tempest-plugin http://logs.openstack.org/83/657183/3/check/horizon-dsvm-tempest-plugin/060fe08/ : SUCCESS in 29m 45s\n- horizon-dsvm-tempest-plugin-py27 http://logs.openstack.org/83/657183/3/check/horizon-dsvm-tempest-plugin-py27/43eb608/ : SUCCESS in 35m 00s\n- horizon-tox-bandit-baseline http://logs.openstack.org/83/657183/3/check/horizon-tox-bandit-baseline/1a3d30c/ : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":3},{"id":"9695e025d5c7423179591e4e44639c489525b360","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-05-08 09:16:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"37954e50d156c4bc58b3e88abf09f3387e70f215","author":{"_account_id":27979,"name":"steven","email":"gaoyl@rc.inesa.com","username":"stevenGao"},"date":"2019-05-09 01:25:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"54d8d1807d668891ef84aa5fbe1e8437fa668b82","author":{"_account_id":27979,"name":"steven","email":"gaoyl@rc.inesa.com","username":"stevenGao"},"date":"2019-05-09 01:31:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d179cb5b58cfa52468acf446b04b138440512995","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 02:21:10.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint http://logs.openstack.org/83/657183/4/check/nodejs10-npm-run-lint/5e9f20d/ : SUCCESS in 5m 47s\n- nodejs10-npm-run-test http://logs.openstack.org/83/657183/4/check/nodejs10-npm-run-test/b6228c6/ : SUCCESS in 7m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/83/657183/4/check/openstack-tox-lower-constraints/58dd564/ : SUCCESS in 10m 41s\n- openstack-tox-pep8 http://logs.openstack.org/83/657183/4/check/openstack-tox-pep8/514731a/ : SUCCESS in 6m 17s\n- openstack-tox-py27 http://logs.openstack.org/83/657183/4/check/openstack-tox-py27/bba2dce/ : SUCCESS in 9m 36s\n- openstack-tox-py36 http://logs.openstack.org/83/657183/4/check/openstack-tox-py36/46f849e/ : SUCCESS in 8m 58s\n- openstack-tox-py37 http://logs.openstack.org/83/657183/4/check/openstack-tox-py37/95f8eb5/ : SUCCESS in 7m 35s\n- openstack-tox-docs http://logs.openstack.org/83/657183/4/check/openstack-tox-docs/710094d/html/ : SUCCESS in 4m 25s\n- horizon-openstack-tox-python3-django111 http://logs.openstack.org/83/657183/4/check/horizon-openstack-tox-python3-django111/391a948/ : SUCCESS in 7m 18s\n- horizon-selenium-headless http://logs.openstack.org/83/657183/4/check/horizon-selenium-headless/2a64782/ : SUCCESS in 6m 30s\n- horizon-integration-tests http://logs.openstack.org/83/657183/4/check/horizon-integration-tests/ac1e81d/ : SUCCESS in 53m 19s\n- horizon-dsvm-tempest-plugin http://logs.openstack.org/83/657183/4/check/horizon-dsvm-tempest-plugin/9c5b45e/ : SUCCESS in 30m 38s\n- horizon-dsvm-tempest-plugin-py27 http://logs.openstack.org/83/657183/4/check/horizon-dsvm-tempest-plugin-py27/bb5e0d6/ : SUCCESS in 30m 50s\n- horizon-tox-bandit-baseline http://logs.openstack.org/83/657183/4/check/horizon-tox-bandit-baseline/4ceb0f2/ : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":4},{"id":"34a6cdbc5dafc736e258312f117049d32667dc16","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-05-12 09:29:19.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"4a632f56d2e7ac7554306e3c4eda2463c10955ad","author":{"_account_id":27979,"name":"steven","email":"gaoyl@rc.inesa.com","username":"stevenGao"},"date":"2019-05-13 02:15:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8b2278da58982700155c90875afdfca34f95a270","author":{"_account_id":27979,"name":"steven","email":"gaoyl@rc.inesa.com","username":"stevenGao"},"date":"2019-05-13 02:25:49.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c1677f50f0f12675a9f9b6f6028937e4303242cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 03:28:17.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint http://logs.openstack.org/83/657183/5/check/nodejs10-npm-run-lint/2225a2e/ : SUCCESS in 4m 16s\n- nodejs10-npm-run-test http://logs.openstack.org/83/657183/5/check/nodejs10-npm-run-test/d9f017a/ : SUCCESS in 5m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/83/657183/5/check/openstack-tox-lower-constraints/886da03/ : SUCCESS in 10m 24s\n- openstack-tox-pep8 http://logs.openstack.org/83/657183/5/check/openstack-tox-pep8/aa933e3/ : SUCCESS in 5m 18s\n- openstack-tox-py27 http://logs.openstack.org/83/657183/5/check/openstack-tox-py27/5205c93/ : SUCCESS in 9m 52s\n- openstack-tox-py36 http://logs.openstack.org/83/657183/5/check/openstack-tox-py36/45db5f6/ : SUCCESS in 8m 42s\n- openstack-tox-py37 http://logs.openstack.org/83/657183/5/check/openstack-tox-py37/eb0a711/ : SUCCESS in 9m 19s\n- openstack-tox-docs http://logs.openstack.org/83/657183/5/check/openstack-tox-docs/c5cdec6/html/ : SUCCESS in 3m 20s\n- horizon-openstack-tox-python3-django111 http://logs.openstack.org/83/657183/5/check/horizon-openstack-tox-python3-django111/486e318/ : SUCCESS in 8m 50s\n- horizon-selenium-headless http://logs.openstack.org/83/657183/5/check/horizon-selenium-headless/00e485e/ : SUCCESS in 4m 52s\n- horizon-integration-tests http://logs.openstack.org/83/657183/5/check/horizon-integration-tests/2bb5c05/ : SUCCESS in 56m 56s\n- horizon-dsvm-tempest-plugin http://logs.openstack.org/83/657183/5/check/horizon-dsvm-tempest-plugin/52c8438/ : SUCCESS in 35m 24s\n- horizon-dsvm-tempest-plugin-py27 http://logs.openstack.org/83/657183/5/check/horizon-dsvm-tempest-plugin-py27/19c2157/ : SUCCESS in 35m 13s\n- horizon-tox-bandit-baseline http://logs.openstack.org/83/657183/5/check/horizon-tox-bandit-baseline/59f8c1d/ : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":5},{"id":"b296d330ba1b4babd35bc6ee4e4f34bc213bf07e","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-05-22 04:01:42.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7c55680e0ed603d1e962fa5e0d58c550c411a77b","author":{"_account_id":27979,"name":"steven","email":"gaoyl@rc.inesa.com","username":"stevenGao"},"date":"2019-05-27 05:22:07.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"aeb7dcc08f852f7df2baae8bfa3d5923482e1045","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 06:20:35.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint http://logs.openstack.org/83/657183/6/check/nodejs10-npm-run-lint/f404300/ : SUCCESS in 3m 25s\n- nodejs10-npm-run-test http://logs.openstack.org/83/657183/6/check/nodejs10-npm-run-test/95f0d97/ : SUCCESS in 5m 22s\n- openstack-tox-lower-constraints http://logs.openstack.org/83/657183/6/check/openstack-tox-lower-constraints/8a2ccbb/ : SUCCESS in 8m 21s\n- openstack-tox-pep8 http://logs.openstack.org/83/657183/6/check/openstack-tox-pep8/70dd256/ : SUCCESS in 4m 50s\n- openstack-tox-py27 http://logs.openstack.org/83/657183/6/check/openstack-tox-py27/6d2fd39/ : SUCCESS in 8m 46s\n- openstack-tox-py36 http://logs.openstack.org/83/657183/6/check/openstack-tox-py36/6982e62/ : SUCCESS in 8m 42s\n- openstack-tox-py37 http://logs.openstack.org/83/657183/6/check/openstack-tox-py37/a33d4fe/ : SUCCESS in 8m 47s\n- openstack-tox-docs http://logs.openstack.org/83/657183/6/check/openstack-tox-docs/5e1c9c4/html/ : SUCCESS in 3m 25s\n- horizon-openstack-tox-python3-django111 http://logs.openstack.org/83/657183/6/check/horizon-openstack-tox-python3-django111/564df26/ : SUCCESS in 8m 48s\n- horizon-selenium-headless http://logs.openstack.org/83/657183/6/check/horizon-selenium-headless/d6c39a6/ : SUCCESS in 5m 18s\n- horizon-integration-tests http://logs.openstack.org/83/657183/6/check/horizon-integration-tests/5287ce6/ : SUCCESS in 53m 15s\n- horizon-dsvm-tempest-plugin http://logs.openstack.org/83/657183/6/check/horizon-dsvm-tempest-plugin/eaf438a/ : SUCCESS in 33m 42s\n- horizon-dsvm-tempest-plugin-py27 http://logs.openstack.org/83/657183/6/check/horizon-dsvm-tempest-plugin-py27/a8ad994/ : SUCCESS in 33m 42s\n- horizon-tox-bandit-baseline http://logs.openstack.org/83/657183/6/check/horizon-tox-bandit-baseline/1810f8f/ : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":6},{"id":"fa840f158402b35fcaa2224932c967496c7ec565","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-05-28 04:45:11.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"501157496a0fa69e864d9d216a47783877e8c434","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-05-28 07:18:24.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"293d55960e5aacd8fd5188012e8f7baac85ee83f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 07:18:39.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"a41e7eab0e6dec88e32ffc0f76a9bf2a4c303b8d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 08:25:32.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- nodejs10-npm-run-lint http://logs.openstack.org/83/657183/6/gate/nodejs10-npm-run-lint/e6f4623/ : SUCCESS in 4m 09s\n- nodejs10-npm-run-test http://logs.openstack.org/83/657183/6/gate/nodejs10-npm-run-test/10c4d61/ : SUCCESS in 5m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/83/657183/6/gate/openstack-tox-lower-constraints/1cb9e1b/ : SUCCESS in 8m 56s\n- openstack-tox-pep8 http://logs.openstack.org/83/657183/6/gate/openstack-tox-pep8/9e88de6/ : SUCCESS in 4m 37s\n- openstack-tox-py27 http://logs.openstack.org/83/657183/6/gate/openstack-tox-py27/be47f15/ : SUCCESS in 10m 09s\n- openstack-tox-py36 http://logs.openstack.org/83/657183/6/gate/openstack-tox-py36/a288431/ : SUCCESS in 10m 42s\n- openstack-tox-py37 http://logs.openstack.org/83/657183/6/gate/openstack-tox-py37/f6cf95c/ : SUCCESS in 8m 03s\n- openstack-tox-docs http://logs.openstack.org/83/657183/6/gate/openstack-tox-docs/9b9c307/html/ : SUCCESS in 3m 07s\n- horizon-openstack-tox-python3-django111 http://logs.openstack.org/83/657183/6/gate/horizon-openstack-tox-python3-django111/f301a11/ : SUCCESS in 7m 48s\n- horizon-dsvm-tempest-plugin http://logs.openstack.org/83/657183/6/gate/horizon-dsvm-tempest-plugin/80f4160/ : SUCCESS in 33m 28s\n- horizon-dsvm-tempest-plugin-py27 http://logs.openstack.org/83/657183/6/gate/horizon-dsvm-tempest-plugin-py27/9d54d5c/ : SUCCESS in 41m 49s\n- horizon-selenium-headless http://logs.openstack.org/83/657183/6/gate/horizon-selenium-headless/369699d/ : SUCCESS in 5m 49s\n- horizon-integration-tests http://logs.openstack.org/83/657183/6/gate/horizon-integration-tests/88b004b/ : SUCCESS in 1h 02m 56s","accounts_in_message":[],"_revision_number":6},{"id":"5c0824b7bcbe0bc71b42114042d700790697bcf4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 08:25:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"a473887aaa719bc95b6c1a65773d9c0bc6c70f3a","revisions":{"ade9c7af5540e9a91fbb8b6d96d0ca7cfd5dd250":{"kind":"REWORK","_number":1,"created":"2019-05-05 01:47:09.000000000","uploader":{"_account_id":27979,"name":"steven","email":"gaoyl@rc.inesa.com","username":"stevenGao"},"ref":"refs/changes/83/657183/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/83/657183/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/657183/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/657183/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/657183/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/83/657183/1"}}},"commit":{"parents":[{"commit":"4ab7f806b822bf1770de12511445bbfdf78a3335","subject":"Merge \"Fix KeyError when image was deleted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/4ab7f806b822bf1770de12511445bbfdf78a3335"}]}],"author":{"name":"gaoyl","email":"gaoyl@rc.inesa.com","date":"2019-05-05 01:27:11.000000000","tz":480},"committer":{"name":"gaoyl","email":"gaoyl@rc.inesa.com","date":"2019-05-05 01:34:50.000000000","tz":480},"subject":"Add tooltips for \"Admin State\" in related forms","message":"Add tooltips for \"Admin State\" in related forms\n\nNow only workflows.py in project/networks/\nhas the tooltip for \"admin state\".\nSince \"admin state\" is confusing, there should\nbe tooltips for wherever it appears.\n\nChange-Id: Ib075efd46af6c8f0d0d61b4ac76160d8de08b77d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/ade9c7af5540e9a91fbb8b6d96d0ca7cfd5dd250"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/ade9c7af5540e9a91fbb8b6d96d0ca7cfd5dd250"}]},"branch":"refs/heads/master"},"f9bc221f00b30442968abd4cf3088e9db07c0609":{"kind":"REWORK","_number":2,"created":"2019-05-05 03:07:35.000000000","uploader":{"_account_id":27979,"name":"steven","email":"gaoyl@rc.inesa.com","username":"stevenGao"},"ref":"refs/changes/83/657183/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/83/657183/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/657183/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/657183/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/657183/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/83/657183/2"}}},"commit":{"parents":[{"commit":"4ab7f806b822bf1770de12511445bbfdf78a3335","subject":"Merge \"Fix KeyError when image was deleted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/4ab7f806b822bf1770de12511445bbfdf78a3335"}]}],"author":{"name":"gaoyl","email":"gaoyl@rc.inesa.com","date":"2019-05-05 01:27:11.000000000","tz":480},"committer":{"name":"gaoyl","email":"gaoyl@rc.inesa.com","date":"2019-05-05 03:06:03.000000000","tz":480},"subject":"Add tooltips for \"Admin State\" in related forms","message":"Add tooltips for \"Admin State\" in related forms\n\nNow only workflows.py in project/networks/\nhas the tooltip for \"admin state\".\nSince \"admin state\" is confusing, there should\nbe tooltips for wherever it appears.\n\nChange-Id: Ib075efd46af6c8f0d0d61b4ac76160d8de08b77d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/f9bc221f00b30442968abd4cf3088e9db07c0609"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/f9bc221f00b30442968abd4cf3088e9db07c0609"}]},"branch":"refs/heads/master"},"7cd5317bf434281a081e67526eb4fd4dd9657e91":{"kind":"REWORK","_number":3,"created":"2019-05-06 02:27:18.000000000","uploader":{"_account_id":27979,"name":"steven","email":"gaoyl@rc.inesa.com","username":"stevenGao"},"ref":"refs/changes/83/657183/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/83/657183/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/657183/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/657183/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/657183/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/83/657183/3"}}},"commit":{"parents":[{"commit":"4ab7f806b822bf1770de12511445bbfdf78a3335","subject":"Merge \"Fix KeyError when image was deleted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/4ab7f806b822bf1770de12511445bbfdf78a3335"}]}],"author":{"name":"gaoyl","email":"gaoyl@rc.inesa.com","date":"2019-05-05 01:27:11.000000000","tz":480},"committer":{"name":"gaoyl","email":"gaoyl@rc.inesa.com","date":"2019-05-06 02:26:02.000000000","tz":480},"subject":"Add tooltips for \"Admin State\" in related forms","message":"Add tooltips for \"Admin State\" in related forms\n\nNow only workflows.py in project/networks/\nhas the tooltip for \"admin state\".\nSince \"admin state\" is confusing, there should\nbe tooltips for wherever it appears.\n\nChange-Id: Ib075efd46af6c8f0d0d61b4ac76160d8de08b77d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/7cd5317bf434281a081e67526eb4fd4dd9657e91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/7cd5317bf434281a081e67526eb4fd4dd9657e91"}]},"branch":"refs/heads/master"},"8557c04ffb915d3e177cc98107938ba970d4b477":{"kind":"REWORK","_number":4,"created":"2019-05-09 01:25:49.000000000","uploader":{"_account_id":27979,"name":"steven","email":"gaoyl@rc.inesa.com","username":"stevenGao"},"ref":"refs/changes/83/657183/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/83/657183/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/657183/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/657183/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/657183/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/83/657183/4"}}},"commit":{"parents":[{"commit":"4ab7f806b822bf1770de12511445bbfdf78a3335","subject":"Merge \"Fix KeyError when image was deleted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/4ab7f806b822bf1770de12511445bbfdf78a3335"}]}],"author":{"name":"gaoyl","email":"gaoyl@rc.inesa.com","date":"2019-05-05 01:27:11.000000000","tz":480},"committer":{"name":"gaoyl","email":"gaoyl@rc.inesa.com","date":"2019-05-09 01:24:28.000000000","tz":480},"subject":"Add tooltips for \"Admin State\" in related forms","message":"Add tooltips for \"Admin State\" in related forms\n\nNow only workflows.py in project/networks/\nhas the tooltip for \"admin state\".\nSince \"admin state\" is confusing, there should\nbe tooltips for wherever it appears.\n\nChange-Id: Ib075efd46af6c8f0d0d61b4ac76160d8de08b77d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/8557c04ffb915d3e177cc98107938ba970d4b477"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/8557c04ffb915d3e177cc98107938ba970d4b477"}]},"branch":"refs/heads/master"},"9141dfc403cefe5d5b19cd5c0850b47e55188fcb":{"kind":"REWORK","_number":5,"created":"2019-05-13 02:25:49.000000000","uploader":{"_account_id":27979,"name":"steven","email":"gaoyl@rc.inesa.com","username":"stevenGao"},"ref":"refs/changes/83/657183/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/83/657183/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/657183/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/657183/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/657183/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/83/657183/5"}}},"commit":{"parents":[{"commit":"4ab7f806b822bf1770de12511445bbfdf78a3335","subject":"Merge \"Fix KeyError when image was deleted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/4ab7f806b822bf1770de12511445bbfdf78a3335"}]}],"author":{"name":"gaoyl","email":"gaoyl@rc.inesa.com","date":"2019-05-05 01:27:11.000000000","tz":480},"committer":{"name":"gaoyl","email":"gaoyl@rc.inesa.com","date":"2019-05-13 02:22:27.000000000","tz":480},"subject":"Add tooltips for \"Admin State\" in related forms","message":"Add tooltips for \"Admin State\" in related forms\n\nNow only workflows.py in project/networks/\nhas the tooltip for \"admin state\".\nSince \"admin state\" is confusing, there should\nbe tooltips for wherever it appears.\n\nChange-Id: Ib075efd46af6c8f0d0d61b4ac76160d8de08b77d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/9141dfc403cefe5d5b19cd5c0850b47e55188fcb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/9141dfc403cefe5d5b19cd5c0850b47e55188fcb"}]},"branch":"refs/heads/master"},"a473887aaa719bc95b6c1a65773d9c0bc6c70f3a":{"kind":"REWORK","_number":6,"created":"2019-05-27 05:22:07.000000000","uploader":{"_account_id":27979,"name":"steven","email":"gaoyl@rc.inesa.com","username":"stevenGao"},"ref":"refs/changes/83/657183/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/83/657183/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/657183/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/657183/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/657183/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/83/657183/6"}}},"commit":{"parents":[{"commit":"4ab7f806b822bf1770de12511445bbfdf78a3335","subject":"Merge \"Fix KeyError when image was deleted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/4ab7f806b822bf1770de12511445bbfdf78a3335"}]}],"author":{"name":"gaoyl","email":"gaoyl@rc.inesa.com","date":"2019-05-05 01:27:11.000000000","tz":480},"committer":{"name":"gaoyl","email":"gaoyl@rc.inesa.com","date":"2019-05-27 05:20:03.000000000","tz":480},"subject":"Add tooltips for \"Admin State\" in related forms","message":"Add tooltips for \"Admin State\" in related forms\n\nNow only workflows.py in project/networks/\nhas the tooltip for \"admin state\".\nSince \"admin state\" is confusing, there should\nbe tooltips for wherever it appears.\n\nChange-Id: Ib075efd46af6c8f0d0d61b4ac76160d8de08b77d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/a473887aaa719bc95b6c1a65773d9c0bc6c70f3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/a473887aaa719bc95b6c1a65773d9c0bc6c70f3a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
