)]}'
{"id":"openstack%2Fhorizon~657367","triplet_id":"openstack%2Fhorizon~master~I7d380b6e12099f3a7c4380903e9259233a833612","project":"openstack/horizon","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"I7d380b6e12099f3a7c4380903e9259233a833612","subject":"Imported Translations from Zanata","status":"MERGED","created":"2019-05-06 09:53:43.000000000","updated":"2019-05-07 23:19:29.000000000","submitted":"2019-05-07 23:19:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657367-1557271169931-86dd5be3","meta_rev_id":"298dd65aa230ce488238de306ade379ef57074de","_number":657367,"virtual_id_number":657367,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":2,"date":"2019-05-07 23:19:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":2,"date":"2019-05-07 11:33:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2019-05-07 22:18:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":1,"date":"2019-05-07 22:18:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-07 11:33:47.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2019-05-07 22:18:36.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-05-07 23:19:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3c4bf9e3b110d6062fa57ffb9242d9d599f85ea5","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-06 09:53:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"603e64dd49b49a8138dd7e97ad6797d5eeb24c02","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-05-06 10:30:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a01f73e1cbf4b15fd26ba204f576abfd16001658","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 10:49:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint http://logs.openstack.org/67/657367/1/check/nodejs10-npm-run-lint/3d218b9/ : SUCCESS in 5m 27s\n- nodejs10-npm-run-test http://logs.openstack.org/67/657367/1/check/nodejs10-npm-run-test/7539010/ : SUCCESS in 5m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/67/657367/1/check/openstack-tox-lower-constraints/066aa54/ : SUCCESS in 9m 42s\n- openstack-tox-pep8 http://logs.openstack.org/67/657367/1/check/openstack-tox-pep8/2b74fbb/ : SUCCESS in 5m 11s\n- openstack-tox-py27 http://logs.openstack.org/67/657367/1/check/openstack-tox-py27/9493871/ : SUCCESS in 9m 10s\n- openstack-tox-py36 http://logs.openstack.org/67/657367/1/check/openstack-tox-py36/4b21d06/ : SUCCESS in 7m 46s\n- openstack-tox-py37 http://logs.openstack.org/67/657367/1/check/openstack-tox-py37/88809d2/ : SUCCESS in 9m 13s\n- openstack-tox-docs http://logs.openstack.org/67/657367/1/check/openstack-tox-docs/45e5803/html/ : SUCCESS in 3m 32s\n- horizon-integration-tests http://logs.openstack.org/67/657367/1/check/horizon-integration-tests/f47eba9/ : SUCCESS in 54m 49s","accounts_in_message":[],"_revision_number":1},{"id":"7063362e647ffbd587eedb3a3e8447fe0e99535b","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-07 10:01:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8d5a7d497e652fdd546913ec1c0fbc1abe5a9c89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 11:03:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint http://logs.openstack.org/67/657367/2/check/nodejs10-npm-run-lint/8f571eb/ : SUCCESS in 4m 23s\n- nodejs10-npm-run-test http://logs.openstack.org/67/657367/2/check/nodejs10-npm-run-test/5166ff5/ : SUCCESS in 5m 29s\n- openstack-tox-lower-constraints http://logs.openstack.org/67/657367/2/check/openstack-tox-lower-constraints/03287b3/ : SUCCESS in 10m 32s\n- openstack-tox-pep8 http://logs.openstack.org/67/657367/2/check/openstack-tox-pep8/db0d17d/ : SUCCESS in 6m 06s\n- openstack-tox-py27 http://logs.openstack.org/67/657367/2/check/openstack-tox-py27/f181e05/ : SUCCESS in 11m 53s\n- openstack-tox-py36 http://logs.openstack.org/67/657367/2/check/openstack-tox-py36/8bbe005/ : SUCCESS in 12m 58s\n- openstack-tox-py37 http://logs.openstack.org/67/657367/2/check/openstack-tox-py37/c58df1f/ : SUCCESS in 10m 48s\n- openstack-tox-docs http://logs.openstack.org/67/657367/2/check/openstack-tox-docs/c62325b/html/ : SUCCESS in 3m 23s\n- horizon-integration-tests http://logs.openstack.org/67/657367/2/check/horizon-integration-tests/cd76990/ : SUCCESS in 54m 58s","accounts_in_message":[],"_revision_number":2},{"id":"56857a817a3d8d6f6e27df8752c7ab1384b71955","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-05-07 11:33:47.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b14b13ddc320ba937c5e32acc3d343c78e002cfa","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-05-07 22:18:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8384b64c2fe3f9e86cb44ea4a81e2c9fade8677f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 22:18:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f3fd06866bc39ca85435a7c57c53ab8571697d69","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 23:19:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f0f7e9bc83e01ace2ae949f1f378126f35c7f122","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 23:19:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- nodejs10-npm-run-lint http://logs.openstack.org/67/657367/2/gate/nodejs10-npm-run-lint/f0ddbd2/ : SUCCESS in 4m 57s\n- nodejs10-npm-run-test http://logs.openstack.org/67/657367/2/gate/nodejs10-npm-run-test/313334a/ : SUCCESS in 5m 53s\n- openstack-tox-lower-constraints http://logs.openstack.org/67/657367/2/gate/openstack-tox-lower-constraints/c5ee141/ : SUCCESS in 9m 04s\n- openstack-tox-pep8 http://logs.openstack.org/67/657367/2/gate/openstack-tox-pep8/f47b9f7/ : SUCCESS in 5m 47s\n- openstack-tox-py27 http://logs.openstack.org/67/657367/2/gate/openstack-tox-py27/5792df1/ : SUCCESS in 8m 59s\n- openstack-tox-py36 http://logs.openstack.org/67/657367/2/gate/openstack-tox-py36/4836636/ : SUCCESS in 8m 03s\n- openstack-tox-py37 http://logs.openstack.org/67/657367/2/gate/openstack-tox-py37/d49a876/ : SUCCESS in 8m 08s\n- openstack-tox-docs http://logs.openstack.org/67/657367/2/gate/openstack-tox-docs/0c8fcf4/html/ : SUCCESS in 4m 25s\n- horizon-integration-tests http://logs.openstack.org/67/657367/2/gate/horizon-integration-tests/2b21425/ : SUCCESS in 57m 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a1d1756ea0f561cfc109928219110dffe9677306","revisions":{"47b4aa656196d5592acd8560f6adac1483009e6f":{"kind":"REWORK","_number":1,"created":"2019-05-06 09:53:43.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/67/657367/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/67/657367/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/67/657367/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/67/657367/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/67/657367/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/67/657367/1"}}},"commit":{"parents":[{"commit":"4ab7f806b822bf1770de12511445bbfdf78a3335","subject":"Merge \"Fix KeyError when image was deleted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/4ab7f806b822bf1770de12511445bbfdf78a3335"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-06 09:53:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-06 09:53:39.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I7d380b6e12099f3a7c4380903e9259233a833612\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/47b4aa656196d5592acd8560f6adac1483009e6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/47b4aa656196d5592acd8560f6adac1483009e6f"}]},"branch":"refs/heads/master"},"a1d1756ea0f561cfc109928219110dffe9677306":{"kind":"REWORK","_number":2,"created":"2019-05-07 10:01:44.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/67/657367/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/67/657367/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/67/657367/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/67/657367/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/67/657367/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/67/657367/2"}}},"commit":{"parents":[{"commit":"4ab7f806b822bf1770de12511445bbfdf78a3335","subject":"Merge \"Fix KeyError when image was deleted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/4ab7f806b822bf1770de12511445bbfdf78a3335"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-07 10:01:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-07 10:01:41.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I7d380b6e12099f3a7c4380903e9259233a833612\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/a1d1756ea0f561cfc109928219110dffe9677306"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/a1d1756ea0f561cfc109928219110dffe9677306"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
