)]}'
{"id":"openstack%2Fhorizon~676313","triplet_id":"openstack%2Fhorizon~master~I6672ab5970178bc5526012f557f0f59a712f0a07","project":"openstack/horizon","branch":"master","topic":"bug/1830301","hashtags":[],"change_id":"I6672ab5970178bc5526012f557f0f59a712f0a07","subject":"Restore IMAGE_RESERVED_CUSTOM_PROPERTIES behaviour","status":"ABANDONED","created":"2019-08-14 03:10:52.000000000","updated":"2020-01-17 18:13:14.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7054babc78c202f54b7dc0bc9f773ab75d3ca1b9","_number":676313,"virtual_id_number":676313,"owner":{"_account_id":25089,"name":"MarkMielke","email":"mark.mielke@gmail.com","username":"MarkMielke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-10-25 10:46:20.000000000","_account_id":25089,"name":"MarkMielke","email":"mark.mielke@gmail.com","username":"MarkMielke"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":1,"date":"2019-08-14 04:28:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25089,"name":"MarkMielke","email":"mark.mielke@gmail.com","username":"MarkMielke"},{"value":-1,"date":"2019-10-25 10:23:52.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":25089,"name":"MarkMielke","email":"mark.mielke@gmail.com","username":"MarkMielke"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":25089,"name":"MarkMielke","email":"mark.mielke@gmail.com","username":"MarkMielke"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25089,"name":"MarkMielke","email":"mark.mielke@gmail.com","username":"MarkMielke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-14 04:28:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-25 10:23:52.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"}],"messages":[{"id":"17dddbf0c495da06626bfb6a3ab87fed3c55f6d9","author":{"_account_id":25089,"name":"MarkMielke","email":"mark.mielke@gmail.com","username":"MarkMielke"},"date":"2019-08-14 03:10:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b4629c40249fc8b3b8f182fd22ba3a4c37c29177","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 04:28:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/06a2c56d7c5a4b4588647b6f37dd207e : SUCCESS in 6m 33s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/0a892b6a286c4e5fb5e824cda5f82e86 : SUCCESS in 5m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/74cb0dca25cd40fbb07293d35a3d755f : SUCCESS in 10m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e00ce79bedd4556ae956e6de456516e : SUCCESS in 5m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b686ce556ef941f39e8dcb1ccad26e73 : SUCCESS in 9m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d770da3e6d5c4853836932653bbcce4a : SUCCESS in 9m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/073c7cef18244396adbd4382fdf8fff5 : SUCCESS in 8m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/167f7f4c2bed414eb0ea5b1dd6bcad97 : SUCCESS in 3m 41s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/11ba29882f694fb29bc915b61826b893 : SUCCESS in 9m 06s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/deb9964da42848319e755122d707ddc4 : FAILURE in 8m 05s (non-voting)\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/7775ea5e383d42fba94994c4fbe9b4ba : SUCCESS in 8m 32s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/95edaf2974ff4623810c5d72e52bc217 : SUCCESS in 1h 11m 50s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/9bed333ca43f4d228d99392bd9fbbd7e : SUCCESS in 35m 18s\n- horizon-dsvm-tempest-plugin-py27 https://zuul.opendev.org/t/openstack/build/f11999e3018847de90a56f4376e276e9 : SUCCESS in 32m 31s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8ec26f35caa546e3af88a920fe1c0bc3 : SUCCESS in 5m 45s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/f616c9a78c0944b5b53ac3a19460712e : SUCCESS in 36m 27s","accounts_in_message":[],"_revision_number":1},{"id":"7ba7b691c2f8f24fa211cecb4742b6c778b8ae73","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-10-25 09:59:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"95ad1b6f559d6a8f0b8572a5413feda3f7c51271","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-10-25 10:23:52.000000000","message":"Patch Set 1: Code-Review-1\n\nPlease, as unit tests to prevent this issue in the future","accounts_in_message":[],"_revision_number":1},{"id":"61da3bf6f35261eb21ef557ce16135aee93edff4","author":{"_account_id":25089,"name":"MarkMielke","email":"mark.mielke@gmail.com","username":"MarkMielke"},"date":"2019-10-25 10:46:20.000000000","message":"Patch Set 1:\n\n\u003e Please, as unit tests to prevent this issue in the future\n\nI don\u0027t disagree. However, I\u0027m not familiar with how to do this for these projects and I\u0027m not making things worse.\n\nHorizon breaks with Rocky without this. I would be willing to try to add a test case, but gating it on this seems like it will prevent it from seeing the light of day for a longer period of time. :-)","accounts_in_message":[],"_revision_number":1},{"id":"6eb0db04b6a190222fe5ec7058f3fe61c9c6f37d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 11:15:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/03f0b6a019e74af6af03879a1d825d02 : SUCCESS in 5m 43s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/0e36a09f7a004b70b79f973087934967 : SUCCESS in 6m 53s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/4c781c88e2e0444f9a63e3856bf448b1 : SUCCESS in 8m 52s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/01397795ad614cb7ba3bca2d92551d2c : SUCCESS in 10m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f383dcc8da343e59b6f40edf98dfb23 : SUCCESS in 11m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afd48c649f0e465f81b2d2df73bd13f7 : SUCCESS in 6m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3bbaa6960c3a489e8dcd8738a93708e0 : SUCCESS in 11m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6012440c75c043a9bcd2643ce3b21461 : SUCCESS in 9m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/30af910660b944749da25a5256d82ad6 : SUCCESS in 8m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84fa96a03d334df5ba8a0bfaf30397e0 : SUCCESS in 7m 46s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/78a3ec05071a49e791fa66f2af504676 : SUCCESS in 6m 41s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/c8b909fcc2104be2905e23009b49d058 : SUCCESS in 1h 10m 56s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/42ef9d46005e47b8853a335078af517d : SUCCESS in 32m 07s\n- horizon-dsvm-tempest-plugin-py27 https://zuul.opendev.org/t/openstack/build/3d4f5f9a403b416db5756bd8c4a84844 : SUCCESS in 41m 04s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/01b6fa19b62f4ca78fcbec2aeb0306d0 : SUCCESS in 4m 56s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/548a57b6ec984411b65ce3dc7a69858a : SUCCESS in 36m 42s","accounts_in_message":[],"_revision_number":1},{"id":"74d664bc25dc1c9d5330893f444cbf269b97f984","author":{"_account_id":25089,"name":"MarkMielke","email":"mark.mielke@gmail.com","username":"MarkMielke"},"date":"2020-01-17 18:13:14.000000000","message":"Abandoned\n\nIt looks like Train has removed documentation for this capability, and it looks like this fix isn\u0027t needed in Train for the original problem I was having (\u0027os_hash_algo\u0027, \u0027os_hash_value\u0027).\n\nI still think this feature was allowed to break, and then the fix was not allowed to be submitted. But, I no longer need this feature, so I will abandon this review and let somebody else take it up if they need this feature returned.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4d41902dae1ff848424a25fdae280cdbd09de49c","revisions":{"4d41902dae1ff848424a25fdae280cdbd09de49c":{"kind":"REWORK","_number":1,"created":"2019-08-14 03:10:52.000000000","uploader":{"_account_id":25089,"name":"MarkMielke","email":"mark.mielke@gmail.com","username":"MarkMielke"},"ref":"refs/changes/13/676313/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/13/676313/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/676313/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/676313/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/676313/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/13/676313/1"}}},"commit":{"parents":[{"commit":"1dc33d0f58cd1f3b36a1a7006c62a578ed6bf81c","subject":"Merge \"Selenium tests fail when response is slow.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/1dc33d0f58cd1f3b36a1a7006c62a578ed6bf81c"}]}],"author":{"name":"Mark Mielke","email":"mark.mielke@gmail.com","date":"2019-08-14 00:54:01.000000000","tz":-240},"committer":{"name":"Mark Mielke","email":"mark.mielke@gmail.com","date":"2019-08-14 01:24:01.000000000","tz":-240},"subject":"Restore IMAGE_RESERVED_CUSTOM_PROPERTIES behaviour","message":"Restore IMAGE_RESERVED_CUSTOM_PROPERTIES behaviour\n\nAs a result of commit 3c2ce866 (\"Angularize metadata update modals\"),\nthe capability to define image custom properties that should not be\ndisplayed in the Image Custom Properties table was lost. The setting\nremained in the sample file and documentation, but it no longer had\nany effect.\n\nThis change restores the original behaviour. Image properties are\nfiltered before they are returned to the client.\n\nChange-Id: I6672ab5970178bc5526012f557f0f59a712f0a07\nPartial-Bug: 1830301\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/4d41902dae1ff848424a25fdae280cdbd09de49c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/4d41902dae1ff848424a25fdae280cdbd09de49c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
