)]}'
{"id":"openstack%2Fhorizon~677149","triplet_id":"openstack%2Fhorizon~master~I34ff8674495e2ba0c7fbbec48bb017365baec6ea","project":"openstack/horizon","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"I34ff8674495e2ba0c7fbbec48bb017365baec6ea","subject":"Imported Translations from Zanata","status":"MERGED","created":"2019-08-19 08:47:47.000000000","updated":"2019-08-19 11:18:41.000000000","submitted":"2019-08-19 11:18:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677149-1566213521301-28760403","meta_rev_id":"ccf4b44e2719e6144a82b7dd97e00710c5ab954d","_number":677149,"virtual_id_number":677149,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":2,"date":"2019-08-19 11:18:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":2,"date":"2019-08-19 10:06:46.000000000","_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":1,"date":"2019-08-19 10:06:46.000000000","_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-19 10:06:46.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2019-08-19 11:18:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7e16d65d09541f51f6094a76faa2c165aa8f188e","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-08-19 08:47:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"312db31e8eaf72b614185ba4aa380e49ce7d9aa3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 09:52:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/4e9962a5632d4b57a54ec3744f2dba28 : SUCCESS in 5m 15s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/8cc4a0192d70471fb9166bc972ffd958 : SUCCESS in 5m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/686137a67dd34c958de763f518709035 : SUCCESS in 10m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46cee7f3140249d193fbb57ddda7ff28 : SUCCESS in 6m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fcaaf35ad89d4d518d10091dd163eb9c : SUCCESS in 12m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9025d5733dd84456ae7abc8df04f3a43 : SUCCESS in 9m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1b0729db620b485bbb648f1bb0699953 : SUCCESS in 9m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5be172f5d99a44bdaff253d443f0caba : SUCCESS in 4m 43s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/60bd35bb09774ca0b4056f86b28e823c : SUCCESS in 59m 20s","accounts_in_message":[],"_revision_number":1},{"id":"1c5eac7f66a0c7fbe080aaac16902852ba119fb9","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-08-19 10:06:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cff477f7611e2f8e31bc6d38b9fcea5c377b5147","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 10:07:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c64867219e4ee58ec21fa55f3583bca2c27c5643","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 11:18:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a65c9171d133a320c83e297b343dfd763f54025d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 11:18:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/5782521046c4499c828c1090743c67e9 : SUCCESS in 8m 24s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/2ff808027e5d45bd81d4aa1fea275cb9 : SUCCESS in 7m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36a0b5876c7b4f60a944507ff18235ec : SUCCESS in 11m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f61e56748b8e41ed93100f8aeada3597 : SUCCESS in 6m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/746214c378c64c1fa7984c1c113e077c : SUCCESS in 12m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d01d58866b2344d3af84841a40d99ba2 : SUCCESS in 11m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/356c7f53aeae435e8626f5114d97fd7c : SUCCESS in 8m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9aae43a26a5c4bd69703fc1bc02ee071 : SUCCESS in 6m 48s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/18ece2d8d5174168a045b467fd3c6091 : SUCCESS in 1h 04m 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fffdb8cc9d2cad35335a9a1ed8488e9c09d9f6e7","revisions":{"fffdb8cc9d2cad35335a9a1ed8488e9c09d9f6e7":{"kind":"REWORK","_number":1,"created":"2019-08-19 08:47:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/49/677149/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/49/677149/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/49/677149/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/49/677149/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/49/677149/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/49/677149/1"}}},"commit":{"parents":[{"commit":"1dc33d0f58cd1f3b36a1a7006c62a578ed6bf81c","subject":"Merge \"Selenium tests fail when response is slow.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/1dc33d0f58cd1f3b36a1a7006c62a578ed6bf81c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-19 08:47:44.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-19 08:47:44.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I34ff8674495e2ba0c7fbbec48bb017365baec6ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/fffdb8cc9d2cad35335a9a1ed8488e9c09d9f6e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/fffdb8cc9d2cad35335a9a1ed8488e9c09d9f6e7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
