)]}'
{"id":"openstack%2Fhorizon~679513","triplet_id":"openstack%2Fhorizon~master~If093f4ecf1e344792b347c5c338b441b1ab42a04","project":"openstack/horizon","branch":"master","topic":"rmcheck","hashtags":[],"change_id":"If093f4ecf1e344792b347c5c338b441b1ab42a04","subject":"Remove the check which causes plugin\u0027s quotas update failure","status":"MERGED","created":"2019-08-30 21:00:04.000000000","updated":"2019-10-01 17:06:01.000000000","submitted":"2019-10-01 06:57:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679513-1569913047898-33733760","meta_rev_id":"53829ee4caa0d2d218c65b480f04a33d47dd708b","_number":679513,"virtual_id_number":679513,"owner":{"_account_id":11880,"name":"Gloria Gu","email":"gloriafanggu@gmail.com","username":"gugl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":11880,"name":"Gloria Gu","email":"gloriafanggu@gmail.com","username":"gugl"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":6825,"name":"Gary W. Smith","email":"garysmith123@gmail.com","username":"gary-w-smith"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-01 06:57:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},{"value":0,"_account_id":5623,"name":"David Lyle","email":"dklyle0@gmail.com","username":"david-lyle"},{"value":0,"_account_id":27897,"name":"Chris DeVita","email":"cdevita@suse.com","username":"cdev3"},{"value":0,"_account_id":12156,"name":"Jeremy Moffitt","email":"jmoffitt@suse.com","username":"jmoffitt"},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":11880,"name":"Gloria Gu","email":"gloriafanggu@gmail.com","username":"gugl"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":2,"date":"2019-09-26 14:07:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":1,"date":"2019-09-05 15:06:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6825,"name":"Gary W. Smith","email":"garysmith123@gmail.com","username":"gary-w-smith"},{"value":2,"date":"2019-10-01 05:48:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-10 03:39:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":1,"date":"2019-09-30 03:33:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},{"value":0,"_account_id":5623,"name":"David Lyle","email":"dklyle0@gmail.com","username":"david-lyle"},{"value":0,"_account_id":27897,"name":"Chris DeVita","email":"cdevita@suse.com","username":"cdev3"},{"value":1,"date":"2019-09-03 20:26:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12156,"name":"Jeremy Moffitt","email":"jmoffitt@suse.com","username":"jmoffitt"},{"value":1,"date":"2019-09-23 07:10:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"date":"2019-09-23 07:10:11.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":11880,"name":"Gloria Gu","email":"gloriafanggu@gmail.com","username":"gugl"},{"value":0,"date":"2019-09-23 07:10:11.000000000","_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":6825,"name":"Gary W. Smith","email":"garysmith123@gmail.com","username":"gary-w-smith"},{"value":1,"date":"2019-10-01 05:48:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},{"value":0,"date":"2019-09-23 07:10:11.000000000","_account_id":5623,"name":"David Lyle","email":"dklyle0@gmail.com","username":"david-lyle"},{"value":0,"date":"2019-09-17 18:22:43.000000000","_account_id":27897,"name":"Chris DeVita","email":"cdevita@suse.com","username":"cdev3"},{"value":0,"_account_id":12156,"name":"Jeremy Moffitt","email":"jmoffitt@suse.com","username":"jmoffitt"},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":11880,"name":"Gloria Gu","email":"gloriafanggu@gmail.com","username":"gugl"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":6825,"name":"Gary W. Smith","email":"garysmith123@gmail.com","username":"gary-w-smith"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},{"value":0,"_account_id":5623,"name":"David Lyle","email":"dklyle0@gmail.com","username":"david-lyle"},{"value":0,"_account_id":27897,"name":"Chris DeVita","email":"cdevita@suse.com","username":"cdev3"},{"value":0,"_account_id":12156,"name":"Jeremy Moffitt","email":"jmoffitt@suse.com","username":"jmoffitt"},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":5623,"name":"David Lyle","email":"dklyle0@gmail.com","username":"david-lyle"},{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":6825,"name":"Gary W. Smith","email":"garysmith123@gmail.com","username":"gary-w-smith"},{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":11880,"name":"Gloria Gu","email":"gloriafanggu@gmail.com","username":"gugl"},{"_account_id":12156,"name":"Jeremy Moffitt","email":"jmoffitt@suse.com","username":"jmoffitt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},{"_account_id":27897,"name":"Chris DeVita","email":"cdevita@suse.com","username":"cdev3"},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-03 20:26:19.000000000","updated_by":{"_account_id":12156,"name":"Jeremy Moffitt","email":"jmoffitt@suse.com","username":"jmoffitt"},"reviewer":{"_account_id":12156,"name":"Jeremy Moffitt","email":"jmoffitt@suse.com","username":"jmoffitt"},"state":"REVIEWER"},{"updated":"2019-09-05 15:06:57.000000000","updated_by":{"_account_id":6825,"name":"Gary W. Smith","email":"garysmith123@gmail.com","username":"gary-w-smith"},"reviewer":{"_account_id":6825,"name":"Gary W. Smith","email":"garysmith123@gmail.com","username":"gary-w-smith"},"state":"REVIEWER"},{"updated":"2019-09-10 03:39:21.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2019-09-17 18:22:43.000000000","updated_by":{"_account_id":27897,"name":"Chris DeVita","email":"cdevita@suse.com","username":"cdev3"},"reviewer":{"_account_id":27897,"name":"Chris DeVita","email":"cdevita@suse.com","username":"cdev3"},"state":"REVIEWER"},{"updated":"2019-09-23 07:10:11.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2019-09-23 07:10:11.000000000","updated_by":{"_account_id":5623,"name":"David Lyle","email":"dklyle0@gmail.com","username":"david-lyle"},"reviewer":{"_account_id":5623,"name":"David Lyle","email":"dklyle0@gmail.com","username":"david-lyle"},"state":"REVIEWER"},{"updated":"2019-09-23 07:10:11.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2019-09-23 07:10:23.000000000","updated_by":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"reviewer":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"state":"REVIEWER"},{"updated":"2019-09-26 14:07:48.000000000","updated_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"REVIEWER"},{"updated":"2019-09-30 03:33:33.000000000","updated_by":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"reviewer":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"state":"REVIEWER"},{"updated":"2019-10-01 05:48:14.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-10-01 06:57:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f53a638aec3ed48ca748674257d4646c04aba5a3","author":{"_account_id":11880,"name":"Gloria Gu","email":"gloriafanggu@gmail.com","username":"gugl"},"date":"2019-08-30 21:00:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dad70964838b6823b8cee1dd21e86a6cc5fa0012","author":{"_account_id":12156,"name":"Jeremy Moffitt","email":"jmoffitt@suse.com","username":"jmoffitt"},"date":"2019-08-30 21:44:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f475f10be721cb5c6f74379f629460edc56abe00","author":{"_account_id":11880,"name":"Gloria Gu","email":"gloriafanggu@gmail.com","username":"gugl"},"date":"2019-08-30 22:00:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"082090c5267a64e815bb76d705c92bf467655828","author":{"_account_id":11880,"name":"Gloria Gu","email":"gloriafanggu@gmail.com","username":"gugl"},"date":"2019-08-30 22:01:53.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e2c3cd2079637e31ad4dee3092635f2da959453c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 23:12:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/6bdebe5533e843a5b4900b4ca5a0a84d : SUCCESS in 8m 01s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/8a6af42359ab41ac908ba7843855fed2 : SUCCESS in 9m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b1c3d9bd5c9046a688ac857b942fd220 : SUCCESS in 10m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd5b5314b5574af28b30f8695cfc7e26 : SUCCESS in 8m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7022339d5b304324b7a5cd343f6ee0d6 : SUCCESS in 11m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d4e98526d194fc2addd60f92be7f258 : SUCCESS in 10m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/948be1d27d8d47cb83226f549b0e4c13 : SUCCESS in 9m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3e14d09bc864ce2bf49f57b9d94e814 : SUCCESS in 5m 27s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/9f8afaf47f4c4e65847f6b672492acb4 : SUCCESS in 9m 59s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/3f0566db8b6841ebba7be46d17f9c097 : FAILURE in 11m 25s (non-voting)\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/498fc401247a451788b4e5c0870313e3 : SUCCESS in 9m 08s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/34d929d498b940e6ad962352b47c7eaf : SUCCESS in 1h 02m 34s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/dff736971def47b5b7c93ade3f272d7c : SUCCESS in 35m 26s\n- horizon-dsvm-tempest-plugin-py27 https://zuul.opendev.org/t/openstack/build/96d034df6ead46c7bb7a939d5cb3878e : SUCCESS in 39m 39s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6e1fc2ed2e204402b5a3171114c20651 : SUCCESS in 7m 05s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/cbb46148c0c6476892deba63aa0100c7 : SUCCESS in 38m 42s","accounts_in_message":[],"_revision_number":2},{"id":"89c47620ed6900bc8c76d48b9385365d81f96c08","author":{"_account_id":12156,"name":"Jeremy Moffitt","email":"jmoffitt@suse.com","username":"jmoffitt"},"date":"2019-09-03 15:14:15.000000000","message":"Patch Set 2: Code-Review+1\n\nlgtm, thanks for fixing the commit message","accounts_in_message":[],"_revision_number":2},{"id":"a624fdd560ecc6cf313e734c6721b7439bd27d31","author":{"_account_id":11880,"name":"Gloria Gu","email":"gloriafanggu@gmail.com","username":"gugl"},"date":"2019-09-03 19:29:14.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"03eee4ab578f60de5138190e7bde7ed17305715c","author":{"_account_id":11880,"name":"Gloria Gu","email":"gloriafanggu@gmail.com","username":"gugl"},"date":"2019-09-03 19:30:38.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"81bb3af2d3b4f6ab5a35a41847e50ec667a7085e","author":{"_account_id":12156,"name":"Jeremy Moffitt","email":"jmoffitt@suse.com","username":"jmoffitt"},"date":"2019-09-03 20:26:19.000000000","message":"Patch Set 4: Code-Review+1\n\nstill looks good, I didnt notice or mind the \"check-in\", though I agree \"commit\" is better","accounts_in_message":[],"_revision_number":4},{"id":"fa78fefef85ef286d38421c39ff126c58888e4c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 20:48:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/1a560cbb9167467d97bcf5888df7bff5 : SUCCESS in 5m 26s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/8d8b14d641b84d918fa42fe85442462e : SUCCESS in 7m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d70e9bf4ec4d42ebbbeb06529d10ef3b : SUCCESS in 9m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae8562a6fe4748259196f130360a85c0 : SUCCESS in 6m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ce051f4651274ad7926bf713caca5c00 : SUCCESS in 11m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0701444aca714db9b9317c9c45f1e541 : SUCCESS in 9m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8b280d8a66ce4099a7f1c19a101013d7 : SUCCESS in 10m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0c37b5a85044ac2a05b969167809290 : SUCCESS in 6m 13s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/e915f0403f4f42eb8d449d539dce5df7 : SUCCESS in 9m 47s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/12e4268bd8da4ceb8fffb3c2833a9170 : FAILURE in 8m 49s (non-voting)\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/b9a35a8245164a138c8e6962b632ba3a : SUCCESS in 6m 57s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/76efd4a6528f47d2b5c81b5f72debd75 : SUCCESS in 1h 10m 44s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/d9d9301cfa024b29a4e4866be68b8d9a : SUCCESS in 38m 18s\n- horizon-dsvm-tempest-plugin-py27 https://zuul.opendev.org/t/openstack/build/cedb5d8dee1148e8a04c6c42f2f7fc10 : SUCCESS in 41m 17s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bc5ae5aed279401b9df28e6c1ac66e51 : SUCCESS in 5m 30s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/574a948f4f8a4b9ab8b962a1ddb201a2 : SUCCESS in 41m 14s","accounts_in_message":[],"_revision_number":4},{"id":"1d52b4438a9582cbbdd6efe203e31d6bc7dcb712","author":{"_account_id":6825,"name":"Gary W. Smith","email":"garysmith123@gmail.com","username":"gary-w-smith"},"date":"2019-09-05 15:06:57.000000000","message":"Patch Set 4: Code-Review+1\n\nThis is a welcome change, and it\u0027s nice to kill two bugs with one change","accounts_in_message":[],"_revision_number":4},{"id":"d0b2252320df65a166ea19cf23da3f002108078f","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2019-09-10 03:39:21.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"716662aceb48e9de59b948d6e4104b529b134c74","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-09-23 07:10:23.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"218e05bfdb9ef657b4d148d8c3b4e21cec18ab1e","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2019-09-26 14:07:48.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"61002b3e61a2395e101827c3542d4c03dfa07180","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-09-30 03:33:33.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"4247cee9049b99c0bf263f1f940909e90f8552b3","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-09-30 04:45:42.000000000","message":"Patch Set 4:\n\nfriendly ping ...","accounts_in_message":[],"_revision_number":4},{"id":"4e288c4b5a9500a46cb75af99296bc3b867de1ba","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-10-01 05:48:14.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nSorry for late. I was checking the ideal solution, but it turns out we need some refactoring in horizon workflow code. It looks good as a workaround. \n\nI see the following problems in the current quota workflow with manila-ui integration.\n- Cannot update the quota of manila\n- Cannot retrieve the current quota of manila\n\nThis workaround fixes the first one, but the second one still remains.\n\nI will investigate how to change the project quota workflow and horizon workflows codes with minimum impact (so that we can backport a fix hopefully)","accounts_in_message":[],"_revision_number":4},{"id":"695be8c85066647f4a430a4e3e7ab6345c25be75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 05:48:31.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"986fc29be10b790b00b121852460d8b3bd42322e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 06:57:27.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/d9f43974ad004326a81cc7540eccbeb9 : SUCCESS in 5m 39s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/7bb380f196d94b7f8bc7f65cbd34c5dc : SUCCESS in 6m 17s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/7c1433bd98144854bd618e066edf442f : SUCCESS in 9m 01s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/81f407034ca446ff8c03dae2778c43df : SUCCESS in 9m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f91cedfdd14943d6860ba5fe83f50821 : SUCCESS in 10m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e95c51c579d46d79686b2fce799723c : SUCCESS in 6m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/43788e19fa7340b9bc286001b48fa321 : SUCCESS in 9m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a33e2f82bb84f769d75a3aac45da75a : SUCCESS in 7m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/99ffecb369b6404ea9a3b0ff50d6d6c1 : SUCCESS in 7m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34371ef041154368b0d22cc541f1b9e6 : SUCCESS in 6m 38s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/65614193b14248e085f84cde898a4389 : SUCCESS in 40m 13s\n- horizon-dsvm-tempest-plugin-py27 https://zuul.opendev.org/t/openstack/build/85626f3512de407b81baaf6c2c2c3911 : SUCCESS in 34m 00s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/8e4be77a39e1430f83ea7a709a510728 : SUCCESS in 5m 45s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/58cf84e9f8e648d0808b88c233817fc9 : SUCCESS in 1h 06m 45s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/bdecd81d8ac648dda92fbec614ba5665 : SUCCESS in 40m 25s","accounts_in_message":[],"_revision_number":4},{"id":"718b7a068a3ecd4da95331818b57632bf78b32a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 06:57:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"48a07c42c032f597657ef4c2a9ac88a60cc7a469","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 06:58:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a29aa87040d647cb97f6e35cfc044891 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":4},{"id":"7a45c11712e680dd60801044d35161107c7ef738","author":{"_account_id":11880,"name":"Gloria Gu","email":"gloriafanggu@gmail.com","username":"gugl"},"date":"2019-10-01 16:35:26.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit be4d2301bb9654a772990d5d6f7bb2b126c7c127","accounts_in_message":[],"_revision_number":4},{"id":"3e37680a78542dee689f4fcdd9a95c57deb019cc","author":{"_account_id":11880,"name":"Gloria Gu","email":"gloriafanggu@gmail.com","username":"gugl"},"date":"2019-10-01 17:06:01.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 5c284f228af2aafc2861f2035930e420991200dc","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"44987f02be671fd81b3dcd8d22f9fa1b219488c2","revisions":{"4aa5d2c6d00778669dbb5717b6694189e99f19ac":{"kind":"REWORK","_number":1,"created":"2019-08-30 21:00:04.000000000","uploader":{"_account_id":11880,"name":"Gloria Gu","email":"gloriafanggu@gmail.com","username":"gugl"},"ref":"refs/changes/13/679513/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/13/679513/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/679513/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/679513/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/679513/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/13/679513/1"}}},"commit":{"parents":[{"commit":"30b7cfb9b8de020122669421d4fce60f1cf09495","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/30b7cfb9b8de020122669421d4fce60f1cf09495"}]}],"author":{"name":"Gloria Gu","email":"gfgu@suse.com","date":"2019-08-30 20:47:39.000000000","tz":-420},"committer":{"name":"Gloria Gu","email":"gfgu@suse.com","date":"2019-08-30 20:47:39.000000000","tz":-420},"subject":"Remove the check which causes plugin\u0027s quotas update failure","message":"Remove the check which causes plugin\u0027s quotas update failure\n\nThe check causes external plugin\u0027s quotas fields unknown\nexception because the quatos fields are not part of\nQUOTA_FIELDS which only includes the fields for\ncinder, neutron and nova.\n\nMeanwhile the exception message throws TypeError.\n\nThis checkin removes the uncessary check. It also helps\nfix the manila-ui\u0027s bug #1842119\n\nChange-Id: If093f4ecf1e344792b347c5c338b441b1ab42a04\nClose-bug: #1798048\nClose-bug: #1842119\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/4aa5d2c6d00778669dbb5717b6694189e99f19ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/4aa5d2c6d00778669dbb5717b6694189e99f19ac"}]},"branch":"refs/heads/master"},"e26a8b3971849bfc68243e7f337a3c2e647bda83":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-08-30 22:01:53.000000000","uploader":{"_account_id":11880,"name":"Gloria Gu","email":"gloriafanggu@gmail.com","username":"gugl"},"ref":"refs/changes/13/679513/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/13/679513/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/679513/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/679513/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/679513/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/13/679513/2"}}},"commit":{"parents":[{"commit":"30b7cfb9b8de020122669421d4fce60f1cf09495","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/30b7cfb9b8de020122669421d4fce60f1cf09495"}]}],"author":{"name":"Gloria Gu","email":"gfgu@suse.com","date":"2019-08-30 20:47:39.000000000","tz":-420},"committer":{"name":"Gloria Gu","email":"gfgu@suse.com","date":"2019-08-30 22:00:54.000000000","tz":-420},"subject":"Remove the check which causes plugin\u0027s quotas update failure","message":"Remove the check which causes plugin\u0027s quotas update failure\n\nThe check causes external plugin\u0027s quotas fields unknown\nexception because the quotas fields are not part of\nQUOTA_FIELDS which only includes the fields for\ncinder, neutron and nova.\n\nMeanwhile the exception message throws TypeError.\n\nThis checkin removes the uncessary check. It also helps\nfix the manila-ui\u0027s bug #1842119\n\nChange-Id: If093f4ecf1e344792b347c5c338b441b1ab42a04\nClose-bug: #1798048\nClose-bug: #1842119\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/e26a8b3971849bfc68243e7f337a3c2e647bda83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/e26a8b3971849bfc68243e7f337a3c2e647bda83"}]},"branch":"refs/heads/master"},"afff624473c5d86ac940ec70cd5279bb29011674":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-09-03 19:29:14.000000000","uploader":{"_account_id":11880,"name":"Gloria Gu","email":"gloriafanggu@gmail.com","username":"gugl"},"ref":"refs/changes/13/679513/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/13/679513/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/679513/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/679513/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/679513/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/13/679513/3"}}},"commit":{"parents":[{"commit":"30b7cfb9b8de020122669421d4fce60f1cf09495","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/30b7cfb9b8de020122669421d4fce60f1cf09495"}]}],"author":{"name":"Gloria Gu","email":"gfgu@suse.com","date":"2019-08-30 20:47:39.000000000","tz":-420},"committer":{"name":"Gloria Gu","email":"gfgu@suse.com","date":"2019-09-03 19:28:35.000000000","tz":-420},"subject":"Remove the check which causes plugin\u0027s quotas update failure","message":"Remove the check which causes plugin\u0027s quotas update failure\n\nThe check causes external plugin\u0027s quotas fields unknown\nexception because the quotas fields are not part of\nQUOTA_FIELDS which only includes the fields for\ncinder, neutron and nova.\n\nMeanwhile the exception message throws TypeError.\n\nThis checkin removes the uncessary check. It also helps\nfix the manila-ui\u0027s bug #1842119\n\nChange-Id: If093f4ecf1e344792b347c5c338b441b1ab42a04\nCloses-bug: #1798048\nCloses-bug: #1842119\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/afff624473c5d86ac940ec70cd5279bb29011674"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/afff624473c5d86ac940ec70cd5279bb29011674"}]},"branch":"refs/heads/master"},"44987f02be671fd81b3dcd8d22f9fa1b219488c2":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-09-03 19:30:38.000000000","uploader":{"_account_id":11880,"name":"Gloria Gu","email":"gloriafanggu@gmail.com","username":"gugl"},"ref":"refs/changes/13/679513/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/13/679513/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/679513/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/679513/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/679513/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/13/679513/4"}}},"commit":{"parents":[{"commit":"30b7cfb9b8de020122669421d4fce60f1cf09495","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/30b7cfb9b8de020122669421d4fce60f1cf09495"}]}],"author":{"name":"Gloria Gu","email":"gfgu@suse.com","date":"2019-08-30 20:47:39.000000000","tz":-420},"committer":{"name":"Gloria Gu","email":"gfgu@suse.com","date":"2019-09-03 19:29:53.000000000","tz":-420},"subject":"Remove the check which causes plugin\u0027s quotas update failure","message":"Remove the check which causes plugin\u0027s quotas update failure\n\nThe check causes external plugin\u0027s quotas fields unknown\nexception because the quotas fields are not part of\nQUOTA_FIELDS which only includes the fields for\ncinder, neutron and nova.\n\nMeanwhile the exception message throws TypeError.\n\nThis commit removes the uncessary check. It also helps\nfix the manila-ui\u0027s bug #1842119\n\nChange-Id: If093f4ecf1e344792b347c5c338b441b1ab42a04\nCloses-bug: #1798048\nCloses-bug: #1842119\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/44987f02be671fd81b3dcd8d22f9fa1b219488c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/44987f02be671fd81b3dcd8d22f9fa1b219488c2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
