)]}'
{"id":"openstack%2Fhorizon~685656","triplet_id":"openstack%2Fhorizon~master~Icbf26a0f34f1a562c809c641768357238d019a04","project":"openstack/horizon","branch":"master","topic":"bug/1845908","hashtags":[],"change_id":"Icbf26a0f34f1a562c809c641768357238d019a04","subject":"Fix localized string (ja)","status":"ABANDONED","created":"2019-09-30 07:02:26.000000000","updated":"2019-10-01 01:35:28.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"297885a16e2be0aa5830ef3b33e6a560a9dfaf9e","_number":685656,"virtual_id_number":685656,"owner":{"_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"date":"2019-09-30 09:39:40.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-09-30 08:25:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-09-30 09:44:04.000000000","_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":-2,"date":"2019-09-30 08:47:20.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-30 08:25:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-30 08:47:20.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2019-09-30 09:39:40.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"}],"messages":[{"id":"9b1390aeff07575c94aba69ab80fb50445bec062","author":{"_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"},"date":"2019-09-30 07:02:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"119e2caefa6c54818d25b57126245c3065827bb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 08:25:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/2b466e601cf94bd7adf06be17aac7c7e : SUCCESS in 6m 56s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/5032428c8612432c842f55f18562e0a7 : SUCCESS in 6m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d754d147d9b24daf91789d7c4e951db4 : SUCCESS in 9m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7634259be3e64d24934a3715364e27a7 : SUCCESS in 17m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fb4cefa1a1254481870426d3cf693c46 : SUCCESS in 14m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac6c2c7d73664dc0b6c55c15e14f6149 : SUCCESS in 12m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1f999ed600d04f2db486112b55ac3ae7 : SUCCESS in 10m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cee683e839c452dbb4e45c5ce115054 : SUCCESS in 14m 56s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/b075178466f84d4b9360b7c4628fe3c2 : SUCCESS in 1h 16m 21s","accounts_in_message":[],"_revision_number":1},{"id":"978bb30f16fca513e053d774f00039cf06419120","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-09-30 08:47:20.000000000","message":"Patch Set 1: Code-Review-2\n\nPlease, propose localization fixes via http://translate.openstack.org","accounts_in_message":[],"_revision_number":1},{"id":"371569bc2ee3c5db79d69a699ee1a503876902b6","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-09-30 09:39:40.000000000","message":"Patch Set 1:\n\nLuckily I am a member of the Japanese translation team and fixed it.","accounts_in_message":[],"_revision_number":1},{"id":"3bb8d26e8b40455d2a79fec42e761d2006066322","author":{"_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"},"date":"2019-09-30 09:44:04.000000000","message":"Patch Set 1:\n\n@Ivan\nThank you very much for pointing out to the correct workflow.\n\n@Motoki-san thanks for the swift fix. I\u0027ve applied to join the Japanese translation since I was on the site already.","accounts_in_message":[],"_revision_number":1},{"id":"d5e92ae8ef3c923c0f4a5f92e88d2b11ff357fe2","author":{"_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"},"date":"2019-10-01 01:35:28.000000000","message":"Abandoned\n\nNobody should change translation files (the .po files in the locale directory) besides the bot. [1]\n\n\n[1] https://docs.openstack.org/i18n/latest/reviewing-translation-import.html","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1eff5491e37ad23f146605bea5033f25931edfc4","revisions":{"1eff5491e37ad23f146605bea5033f25931edfc4":{"kind":"REWORK","_number":1,"created":"2019-09-30 07:02:26.000000000","uploader":{"_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"},"ref":"refs/changes/56/685656/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/56/685656/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/56/685656/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/56/685656/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/56/685656/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/56/685656/1"}}},"commit":{"parents":[{"commit":"4783feea79bb4dd53d48de68983dcbe99e234084","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/4783feea79bb4dd53d48de68983dcbe99e234084"}]}],"author":{"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","date":"2019-09-30 06:59:29.000000000","tz":540},"committer":{"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","date":"2019-09-30 06:59:29.000000000","tz":540},"subject":"Fix localized string (ja)","message":"Fix localized string (ja)\n\nChange-Id: Icbf26a0f34f1a562c809c641768357238d019a04\nCloses-Bug: #1845908\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/1eff5491e37ad23f146605bea5033f25931edfc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/1eff5491e37ad23f146605bea5033f25931edfc4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
