)]}'
{"id":"openstack%2Fhorizon~686341","triplet_id":"openstack%2Fhorizon~master~I5f3c4c4c72847f52b5e23d5ffa52262fd8a998a6","project":"openstack/horizon","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"I5f3c4c4c72847f52b5e23d5ffa52262fd8a998a6","subject":"Imported Translations from Zanata","status":"MERGED","created":"2019-10-03 07:28:52.000000000","updated":"2019-10-03 09:54:19.000000000","submitted":"2019-10-03 09:52:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686341-1570096358678-4cdd5882","meta_rev_id":"cda8859877253337d1cee408929db65811786692","_number":686341,"virtual_id_number":686341,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-03 09:52:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":2,"date":"2019-10-03 08:42:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":1,"date":"2019-10-03 08:42:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-03 08:42:36.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-10-03 09:52:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"910be40ab510ae0406d856812b9e685bbbae44e2","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-03 07:28:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"21b07f7be250562c1e28d3915416274572a35099","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 08:39:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/fa53594cc4b041eb8a106cde6ce6dddf : SUCCESS in 4m 24s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/6ab8fe04162b40239e3add62b03979db : SUCCESS in 6m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9943e54469cb494eadb4b705b09c65cc : SUCCESS in 10m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/006d025832fe4176b890a935be81a446 : SUCCESS in 5m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6ad6c909269d439abb6ad5b635f4a8fa : SUCCESS in 10m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f65eda2eb1848c8bee166c34b43914c : SUCCESS in 8m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1d5ae514f4184994b09069a6fb5d47ed : SUCCESS in 8m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12ea7a25eff54a059b667c1257e5ded0 : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/766fa118699e46778f65291d0cf6d644 : SUCCESS in 25m 10s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/430c5e9081584154a488da7c7f4be9e8 : SUCCESS in 1h 02m 13s","accounts_in_message":[],"_revision_number":1},{"id":"afb2acfe4b6f593efe54fe4675d6f5c5860c01ba","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-10-03 08:42:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d20fb1597fe302bbf58ac123f5fb17be510fb51b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 08:42:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0dd08469bd90c2f1a43d39589aab9a9c379a81c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 09:52:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/5c841f12f4a94189af71178dfdd20433 : SUCCESS in 4m 12s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/acc6be3e58d84ea8bb6124c0086cb384 : SUCCESS in 7m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b1e70981c02451bad67cc5461ff089f : SUCCESS in 8m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ae1e07f6c004cb18384ffd8a386fbff : SUCCESS in 5m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/422f44dff02f4819862c9b3d0e42ca80 : SUCCESS in 9m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4549db04def4d4985cf7a51d0d6d8e6 : SUCCESS in 7m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bc669ee6134e4880a62a4a096ccde6ba : SUCCESS in 8m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cc05aa0c34b48e3902d2a08b4cab566 : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9693c741f4bf4508baffa3525b9efbf6 : SUCCESS in 55m 17s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/758c6ab3f3f24404b76ce31591a7e3d0 : SUCCESS in 1h 08m 12s","accounts_in_message":[],"_revision_number":1},{"id":"a5e3c18323cfd9790fd1520e121b0227ee0f041f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 09:52:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"df0aeda6d2404cf9bffeccfd4c9e0a135d24aa3e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 09:54:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2e6b4d2d5d84d53bdf1d109b21fbfc7 : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b728c9b86084facab243b511a32762c : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a035ea6aff464d0eb9f35f453e3cc3e5b312b240","revisions":{"a035ea6aff464d0eb9f35f453e3cc3e5b312b240":{"kind":"REWORK","_number":1,"created":"2019-10-03 07:28:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/41/686341/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/41/686341/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/41/686341/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/41/686341/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/41/686341/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/41/686341/1"}}},"commit":{"parents":[{"commit":"cf904d119418404bdcf1ff7b6a6a50157fc8ba03","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/cf904d119418404bdcf1ff7b6a6a50157fc8ba03"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-03 07:28:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-03 07:28:50.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I5f3c4c4c72847f52b5e23d5ffa52262fd8a998a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/a035ea6aff464d0eb9f35f453e3cc3e5b312b240"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/a035ea6aff464d0eb9f35f453e3cc3e5b312b240"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
