)]}'
{"id":"openstack%2Fhorizon~687439","triplet_id":"openstack%2Fhorizon~master~I2d71392eb8569ffb8cb15af29eea76e120a158cc","project":"openstack/horizon","branch":"master","topic":"app-cred-access-rules","hashtags":[],"change_id":"I2d71392eb8569ffb8cb15af29eea76e120a158cc","subject":"Add support for keystone access rules","status":"MERGED","created":"2019-10-08 23:39:56.000000000","updated":"2019-11-25 09:28:08.000000000","submitted":"2019-11-25 09:25:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687439-1574673956313-29596582","meta_rev_id":"c019ff992acd51ef3613b3cb20d2523b34180717","_number":687439,"virtual_id_number":687439,"owner":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"date":"2019-11-22 19:46:04.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-25 09:25:56.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":2,"date":"2019-11-25 07:59:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":2,"date":"2019-10-30 05:49:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":1,"date":"2019-11-25 07:59:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-30 05:49:48.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-11-25 07:59:01.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2019-11-25 09:25:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5099d89c2956d6742d8246584b879545d25a546c","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-10-08 23:39:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d5076ea46011f659bc1df6aa3cabaf4e2b7391b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 00:47:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/40d9ed2682784bc0b9fcc00d5cdc5694 : SUCCESS in 4m 53s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/645d5866a95f49ef9fba88222c2946f7 : SUCCESS in 6m 09s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/9e9433fc0c6946659ed59a5bec9adb3f : FAILURE in 8m 56s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/37bc3e62cea042db84056cddf3ce46c0 : FAILURE in 10m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70eeab5ca6164b18bf049413b941bc36 : FAILURE in 10m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/078605aa426442f8abbfe057d06028c6 : SUCCESS in 4m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f0dd46a1e02941c9a60db4447108da9c : FAILURE in 10m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/969cb9a8b91245f68d1c2e8ecff00a81 : FAILURE in 9m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e0881cde4b394936bfd325226e98560f : FAILURE in 8m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d04834cffbf54653a49eb9129954e8ab : SUCCESS in 6m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b2eba23d03c4e1889e10ee1b0ca67b4 : SUCCESS in 24m 55s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/c4643b6574954da694167655e0e7975b : SUCCESS in 5m 24s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/711ed61899934fb28a43228a81a864ba : SUCCESS in 1h 06m 30s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/ae754aaf22b14819bfe67c2c5fcdf0d7 : SUCCESS in 37m 28s\n- horizon-dsvm-tempest-plugin-py27 https://zuul.opendev.org/t/openstack/build/14e802b51ab04ead9afd048510fe75e6 : SUCCESS in 26m 15s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2086d7e5d4c64e018144e1afb2ebafc6 : SUCCESS in 3m 53s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/3e5a76a2314741b7a22afda7be628120 : SUCCESS in 33m 50s","accounts_in_message":[],"_revision_number":1},{"id":"905070bc59baad99c328d7b59af0c54e6530bc1f","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-10-09 01:34:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2abfa8c3dac15f00798e1feece395fdf6b0c03b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 02:17:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/561a3e6908594907983d6ea80f6a30a7 : SUCCESS in 2m 09s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/cacacd3fe80a45239869dc9f79046426 : RETRY_LIMIT in 3m 57s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/e81061a02f4a4611ba57e18137f81365 : RETRY_LIMIT in 4m 27s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/91534cb51c7241eb822e4bdc0681d48e : FAILURE in 3m 38s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/cbe8127b45d84c6fb8da222b0695db94 : FAILURE in 3m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61d36c2c8c0e464ca7bbbca1c1db711f : FAILURE in 2m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ea9d390083b4b708d698981990dc054 : FAILURE in 3m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/97ede0e0e6db49c69e331718742ec51e : FAILURE in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d43f5f53d5f42b4bb4fb00ffeb77aae : FAILURE in 5m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3797d5fdd91541f2b52969d1f816561d : FAILURE in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2029ab145bb4480bd875abaa8481578 : FAILURE in 3m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd73e6a572834ff68e0b3763b624d5bf : SUCCESS in 21m 46s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/a0c51f1f9c8942eaa0364ce293fb0516 : FAILURE in 4m 28s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/f94f4222bb804bc7a82a04302d2a59a8 : FAILURE in 39m 38s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/fa566cc280cb4e97a89073c82efbe5d2 : SUCCESS in 39m 54s\n- horizon-dsvm-tempest-plugin-py27 https://zuul.opendev.org/t/openstack/build/af8b2fbb0b244a3bb2db0e3c5e5fb5de : SUCCESS in 33m 43s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4a152977dd084e03b2d23b288c5e1ea6 : FAILURE in 4m 13s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/cef5a5d710c04fd2ae6730c78a23f54c : SUCCESS in 40m 05s","accounts_in_message":[],"_revision_number":2},{"id":"9559d8f693d5171f81749d9b9f777a3cd67165ce","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-10-22 19:14:27.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ed5c2a21d4d00c8c074ea7256f73476a4a4397ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 20:00:12.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3d8f93270daf46a090638f26f68945be : SUCCESS in 3m 17s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/67205a618da64aad80d5a3cdadf7abf2 : RETRY_LIMIT in 4m 54s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/621703cdbadc46a2826f18df622e436f : RETRY_LIMIT in 4m 42s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/30fa0834ffca459c85f8c2f376bab225 : FAILURE in 5m 02s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/165e820435384e679cf2c86787d9f399 : FAILURE in 6m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bdfbc4a160564fcc8bfc73f9c4c4ee57 : FAILURE in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef23977dcf2a4dabb1ea73612d5d0eff : FAILURE in 4m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/85f1af8b1a694a0fb8f213e04553a3fa : FAILURE in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8bdc31212f6d4e1c9cb68b6c1d9a44eb : FAILURE in 6m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/55734ab34081490a93af76defec63131 : FAILURE in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f20c316450ed4c4db7aa042337edf3d5 : FAILURE in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5420662e67c64e17874ea134716bec1e : SUCCESS in 25m 50s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/c153d79e10984d8c9fc41eec4fc1f128 : FAILURE in 5m 25s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/7e0bde3413064386abb777f021186a10 : FAILURE in 40m 46s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/d70c9d60703b4643af60b7d2786f97a0 : SUCCESS in 35m 08s\n- horizon-dsvm-tempest-plugin-py27 https://zuul.opendev.org/t/openstack/build/1e9e150fad234c3aa55363037aaf1c4b : SUCCESS in 34m 42s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8043ee20085d42699e918828cb322f1b : FAILURE in 4m 11s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/545026d183ef4c7cb699fe4013771148 : SUCCESS in 34m 17s","accounts_in_message":[],"_revision_number":2},{"id":"a607fc4601bc82945ccdf186f02b59351a6e6a2f","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-10-23 17:33:29.000000000","message":"Patch Set 2:\n\nrecheck\n\nu-c bumped this morning","accounts_in_message":[],"_revision_number":2},{"id":"f99e01ac5c99359f762bf46d7309b87c0fa4d0e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 18:54:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6b131ebbee1144ae8ecd8dc6e583ac30 : SUCCESS in 4m 09s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/40c70f0d29634095b942847a6b90cb3c : SUCCESS in 9m 29s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/da7855ba9f014cc7bf6878eddfdd1138 : SUCCESS in 6m 53s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/544606cd39574344bf53ac6cef24ebe0 : SUCCESS in 8m 53s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/d6a05eb81bd04b20b3323a2b416a2670 : SUCCESS in 15m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/90427eebc40e43b1b4fa2471b24a6fdd : SUCCESS in 9m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8458f485532247049e12122f47ed7435 : SUCCESS in 5m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e6e5c1701e7c4bde83ca52d8f2c3cd08 : SUCCESS in 9m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd5337c9b0a64674b781782874e42351 : SUCCESS in 10m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a415f7dac1074d609645b1c3af533348 : SUCCESS in 9m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/376dc9f7075d45ba8fd5e741d4ba5fe8 : SUCCESS in 6m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79bcc3cbd17b462d9e95b3bcc90e6529 : SUCCESS in 21m 48s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/d14ff3732e0c4cf2bfed37827ea6f4aa : SUCCESS in 7m 02s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/c8eb724201a54f27ad1e6c0c51d82e7c : SUCCESS in 1h 08m 34s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/7c3ade44ff04480d99706a02ef88c487 : SUCCESS in 49m 39s\n- horizon-dsvm-tempest-plugin-py27 https://zuul.opendev.org/t/openstack/build/a7b54c18daf54b59ac38d7327b987655 : SUCCESS in 51m 07s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3b660b99303a4a7fb1c7cf3724579da4 : SUCCESS in 9m 30s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/11ece3fc71204b2e927e78703323d2d5 : SUCCESS in 41m 18s","accounts_in_message":[],"_revision_number":2},{"id":"a2373764fa05d90a3a1bbd4c9f842658479d75ab","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-10-25 05:36:55.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"13b07cc02f00f80f46d2f63136fe48fe55413a65","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-10-29 20:35:10.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"66d9e2086824d8dd4c68a0ee4222c70f6f193299","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-10-29 20:35:30.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"e14d02236f279234ea1dffa3f72441320669a4c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 22:21:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/1d44ccdabbad4e5ba663ea29af6833d2 : SUCCESS in 3m 25s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/446dfb5937474121979d95f4848c2867 : SUCCESS in 5m 03s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/195a465603c54ed6bf12e1a238f6db06 : SUCCESS in 6m 12s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/353c1d43874a4cd19538a83c528f9bd1 : SUCCESS in 8m 01s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/5d9e86d40b8440268d66661934deacb4 : SUCCESS in 9m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce8ffb01df34454586572654e7bf74b3 : SUCCESS in 10m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56c5fda0e6ac4f2c92750aa763be68a3 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c01fbc11868e4e21aa0cc727bc7ab2ad : SUCCESS in 8m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2ed740bcdf3547b489ff3cc44f73cf06 : SUCCESS in 8m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/671ef55c0a344738bc8de33614349781 : SUCCESS in 7m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f97bf828fa24621a02a6b1da84c4c2d : SUCCESS in 20m 11s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/7f70cc84bd404161be6e4aa8d1a48fda : SUCCESS in 5m 51s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/46054221bd0542e28b0ef6d4ca116d76 : SUCCESS in 1h 08m 57s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/656ebfdfd5c84480827cf0f09721448c : SUCCESS in 38m 21s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/da5c6f7976174b2babdce9a5e56a9e45 : FAILURE in 6m 29s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/6d0e45662dfc46e7a76a2356d1af551f : SUCCESS in 1h 09m 33s","accounts_in_message":[],"_revision_number":3},{"id":"6442183cef105728b544210e1e0cfe913f04b239","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-10-29 22:34:09.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b699c801016fa304b76bd89474e127d1a14a7874","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 00:25:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/e79d60ba276c4bd8ab00662c20149ddf : SUCCESS in 2m 40s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/27a2d6231f1c471b9f22c744e6c0f51b : SUCCESS in 7m 10s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/96411ca2cf05435a9ad1458cc39edbee : SUCCESS in 6m 54s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/959d11d5c5d54cb79ebdf0b5e8f53df8 : SUCCESS in 10m 41s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/db57be50b3c148d9abc43b0036590fe4 : SUCCESS in 9m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3ffa6b60a4e410aa24b722f57c63f8e : SUCCESS in 8m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/072310be0bc7469ea31c8e5d1d924a47 : SUCCESS in 8m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4409bddcc6cc4df2ba6eb1337124527b : SUCCESS in 7m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/01433e268f5d45b4af59228d4c57d29f : SUCCESS in 8m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1539ab766d6a480cb31fb8d086b046da : SUCCESS in 7m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/907a07f569cd49498399f0d6e69e7ee8 : SUCCESS in 20m 04s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/580d2a3d027646ed87e49096a56c2fe7 : SUCCESS in 9m 40s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/79c9562edbc84045a0a82916d2d8bb2c : SUCCESS in 1h 15m 51s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/33cb66c2ce3249d29e26df3e0a8f298d : SUCCESS in 37m 38s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/02b7c5349e884a66839caa5af111a5d8 : SUCCESS in 5m 06s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/9b525cab82634d7a9a0307edaa086247 : SUCCESS in 33m 08s","accounts_in_message":[],"_revision_number":4},{"id":"b33a778f83271799a41ad2728ef765eb5eb5e004","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-10-30 05:47:29.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f5c5370bd1c58b7a4da1af5c3273a6727df656f5","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-10-30 05:49:48.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"43512265b176243fef01118a65fbd8ff7998705d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 07:28:21.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/04c587a5edc94361b487805087b6877d : SUCCESS in 2m 33s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/ae91bb771649442488d473b46590b7d2 : SUCCESS in 5m 29s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/876657af68594d1a8448c8553b965afe : SUCCESS in 6m 55s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/29b84ea535cf413f8da18a35572e90f7 : SUCCESS in 7m 37s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/42538d488cc94f4bbe52f4ce2fe2c7d9 : SUCCESS in 9m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b5be1b4bf7444508c8b9ada1af40ab7 : SUCCESS in 8m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b097800c728c4bd6b711553a875e94e4 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cdb79b604e214ead8685d6d8f4946207 : SUCCESS in 10m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/32ba83b0039d45fb9752cdbd308b1cb6 : SUCCESS in 8m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c035bc595a0f4610bbc2d1cbcb0aa58b : SUCCESS in 6m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b57b486165714e958bf16abd227bb2cf : SUCCESS in 23m 50s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/4a7b78a789064eb7b35a10e9a7127260 : SUCCESS in 6m 33s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/71d25825521e41528521d009a2d1d54c : SUCCESS in 1h 08m 16s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/10e496eb4c844876976282c0ccd5a316 : SUCCESS in 33m 13s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3f945c9442f8486d942d34b1417061fb : SUCCESS in 4m 37s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/a492a4ea20c64c199714844886561a8b : SUCCESS in 32m 44s","accounts_in_message":[],"_revision_number":5},{"id":"4224a35afa93c5cf3618132014a2e227e5c943fe","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-11-21 07:20:08.000000000","message":"Patch Set 5: Code-Review-1\n\nHi Collen, thanks for the patch but when I tested this patch in my devstack environment I am not able to create \"Application Credentials\". Could you please try the same?\nSo -1 for that.","accounts_in_message":[],"_revision_number":5},{"id":"83da919f6f2bc8b6894d30d7f72ad5d80f163821","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-11-22 19:46:04.000000000","message":"Patch Set 5:\n\n\u003e Hi Collen, thanks for the patch but when I tested this patch in my\n \u003e devstack environment I am not able to create \"Application\n \u003e Credentials\". Could you please try the same?\n \u003e So -1 for that.\n\nVishal - I of course tested this before submitting it, it works well for all cases I tested. Can you describe how you were testing it and what errors were emitted by the horizon UI and/or the horizon and keystone logs?","accounts_in_message":[],"_revision_number":5},{"id":"fb2d8ca8697ae194d7563f282088e3f46ba2b52d","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-11-25 07:56:33.000000000","message":"Patch Set 5: -Code-Review\n\n\u003e \u003e Hi Collen, thanks for the patch but when I tested this patch in\n \u003e my\n \u003e \u003e devstack environment I am not able to create \"Application\n \u003e \u003e Credentials\". Could you please try the same?\n \u003e \u003e So -1 for that.\n \u003e \n \u003e Vishal - I of course tested this before submitting it, it works\n \u003e well for all cases I tested. Can you describe how you were testing\n \u003e it and what errors were emitted by the horizon UI and/or the\n \u003e horizon and keystone logs?\n\nOhh sorry mybad .\nI think there was some issue with my environment, since I tested in fresh environment it worked successfully.","accounts_in_message":[],"_revision_number":5},{"id":"cae32c18b7d964f496045d4fe0d26af4ea5fd1c8","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-11-25 07:59:01.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nThanks, Colleen for working on this.\nLGTM.","accounts_in_message":[],"_revision_number":5},{"id":"2e94aebdd54f5b772c040e55f66bc5f18d319777","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 07:59:14.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"0e003384a95cd3edd188a025db6a080f9d3fb8e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 09:25:56.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f24c5c4b69f644089271e36026044b06 : SUCCESS in 1m 48s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/e843baf8a4f24bb5bce403a5ce48004c : SUCCESS in 4m 19s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/0bcf8cc32dfa466199489bd98a83ba12 : SUCCESS in 5m 30s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/84c34302eb4d4f9d91154331da8fadd0 : SUCCESS in 8m 26s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/285c36af466549c8819e29599088008a : SUCCESS in 9m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8effab36b9224b7aba4c1f87934289f6 : SUCCESS in 8m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a7e4aaadd0248efb3217688042531a2 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59091b1ea22e486daf5dbc78b3797013 : SUCCESS in 8m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a94fe4fa63b8401ca97cd381edbf3661 : SUCCESS in 8m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a86b1e039c0c4ee3a011beece4f651a8 : SUCCESS in 8m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b45e90d604294f7da9804fcae53ac590 : SUCCESS in 22m 07s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/b29f7daab5c14218af861781ed50a577 : SUCCESS in 46m 14s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/e28dcb3f6667488882d903f238bc3aac : SUCCESS in 8m 07s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/cc505106ad0a490e8f75c464cb2299c9 : SUCCESS in 1h 24m 01s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/f5653863e9bc487c89d492f9f8eb2d2e : SUCCESS in 38m 28s","accounts_in_message":[],"_revision_number":5},{"id":"75b0353a0bc32c6d4da01d6f68f10e87b062eac1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 09:25:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"eb15e961137ce999557bf6e39d0ff0612e948dc3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 09:28:08.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24f2239d54604268ade7a9d030180ef1 : SUCCESS in 1m 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc6f89d8654040c49a1154b2d71afb1d : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"4d1786c68724c91c2686cda89540fa91b9c21bcf","revisions":{"e8b0e7016a5898461ff7e71c0f500b47aa12e907":{"kind":"REWORK","_number":1,"created":"2019-10-08 23:39:56.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/39/687439/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/39/687439/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/39/687439/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/39/687439/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/39/687439/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/39/687439/1"}}},"commit":{"parents":[{"commit":"307e884eaa799691f081d73362a287d20a1ad9a8","subject":"Fix grammar error in a recent release notes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/307e884eaa799691f081d73362a287d20a1ad9a8"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-10-08 23:37:55.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-10-08 23:37:55.000000000","tz":-420},"subject":"Add support for keystone access rules","message":"Add support for keystone access rules\n\nKeystone implemented the ability to apply fine-grained access control\nrestrictions to application credentials[1]. This patch adds new fields\nto the application credential creation form and detail view so that\nhorizon users can use this feature.\n\n[1] http://specs.openstack.org/openstack/keystone-specs/specs/keystone/train/capabilities-app-creds.html\n\nChange-Id: I2d71392eb8569ffb8cb15af29eea76e120a158cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/e8b0e7016a5898461ff7e71c0f500b47aa12e907"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/e8b0e7016a5898461ff7e71c0f500b47aa12e907"}]},"branch":"refs/heads/master"},"b5079527c0ada9f1dc1835c832f5486c8b789651":{"kind":"REWORK","_number":2,"created":"2019-10-09 01:34:05.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/39/687439/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/39/687439/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/39/687439/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/39/687439/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/39/687439/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/39/687439/2"}}},"commit":{"parents":[{"commit":"307e884eaa799691f081d73362a287d20a1ad9a8","subject":"Fix grammar error in a recent release notes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/307e884eaa799691f081d73362a287d20a1ad9a8"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-10-08 23:37:55.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-10-09 01:33:48.000000000","tz":-420},"subject":"Add support for keystone access rules","message":"Add support for keystone access rules\n\nKeystone implemented the ability to apply fine-grained access control\nrestrictions to application credentials[1]. This patch adds new fields\nto the application credential creation form and detail view so that\nhorizon users can use this feature.\n\n[1] http://specs.openstack.org/openstack/keystone-specs/specs/keystone/train/capabilities-app-creds.html\n\nDepends-on: https://review.opendev.org/677585\n\nChange-Id: I2d71392eb8569ffb8cb15af29eea76e120a158cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/b5079527c0ada9f1dc1835c832f5486c8b789651"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/b5079527c0ada9f1dc1835c832f5486c8b789651"}]},"branch":"refs/heads/master"},"63615c29644a47afced6d6081bc6d99567dc3ad2":{"kind":"REWORK","_number":3,"created":"2019-10-29 20:35:10.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/39/687439/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/39/687439/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/39/687439/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/39/687439/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/39/687439/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/39/687439/3"}}},"commit":{"parents":[{"commit":"307e884eaa799691f081d73362a287d20a1ad9a8","subject":"Fix grammar error in a recent release notes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/307e884eaa799691f081d73362a287d20a1ad9a8"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-10-08 23:37:55.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-10-29 20:35:00.000000000","tz":-420},"subject":"Add support for keystone access rules","message":"Add support for keystone access rules\n\nKeystone implemented the ability to apply fine-grained access control\nrestrictions to application credentials[1]. This patch adds new fields\nto the application credential creation form and detail view so that\nhorizon users can use this feature.\n\n[1] http://specs.openstack.org/openstack/keystone-specs/specs/keystone/train/capabilities-app-creds.html\n\nDepends-on: https://review.opendev.org/677585\n\nChange-Id: I2d71392eb8569ffb8cb15af29eea76e120a158cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/63615c29644a47afced6d6081bc6d99567dc3ad2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/63615c29644a47afced6d6081bc6d99567dc3ad2"}]},"branch":"refs/heads/master"},"3b578bc89cfc7491ec4193c2426df55951f814ee":{"kind":"REWORK","_number":4,"created":"2019-10-29 22:34:09.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/39/687439/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/39/687439/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/39/687439/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/39/687439/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/39/687439/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/39/687439/4"}}},"commit":{"parents":[{"commit":"307e884eaa799691f081d73362a287d20a1ad9a8","subject":"Fix grammar error in a recent release notes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/307e884eaa799691f081d73362a287d20a1ad9a8"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-10-08 23:37:55.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-10-29 22:33:26.000000000","tz":-420},"subject":"Add support for keystone access rules","message":"Add support for keystone access rules\n\nKeystone implemented the ability to apply fine-grained access control\nrestrictions to application credentials[1]. This patch adds new fields\nto the application credential creation form and detail view so that\nhorizon users can use this feature.\n\n[1] http://specs.openstack.org/openstack/keystone-specs/specs/keystone/train/capabilities-app-creds.html\n\nDepends-on: https://review.opendev.org/677585\n\nChange-Id: I2d71392eb8569ffb8cb15af29eea76e120a158cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/3b578bc89cfc7491ec4193c2426df55951f814ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/3b578bc89cfc7491ec4193c2426df55951f814ee"}]},"branch":"refs/heads/master"},"4d1786c68724c91c2686cda89540fa91b9c21bcf":{"kind":"REWORK","_number":5,"created":"2019-10-30 05:47:29.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/39/687439/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/39/687439/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/39/687439/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/39/687439/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/39/687439/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/39/687439/5"}}},"commit":{"parents":[{"commit":"307e884eaa799691f081d73362a287d20a1ad9a8","subject":"Fix grammar error in a recent release notes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/307e884eaa799691f081d73362a287d20a1ad9a8"}]}],"author":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-10-08 23:37:55.000000000","tz":-420},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-10-30 05:47:05.000000000","tz":540},"subject":"Add support for keystone access rules","message":"Add support for keystone access rules\n\nKeystone implemented the ability to apply fine-grained access control\nrestrictions to application credentials[1]. This patch adds new fields\nto the application credential creation form and detail view so that\nhorizon users can use this feature.\n\n[1] http://specs.openstack.org/openstack/keystone-specs/specs/keystone/train/capabilities-app-creds.html\n\nDepends-on: https://review.opendev.org/677585\n\nChange-Id: I2d71392eb8569ffb8cb15af29eea76e120a158cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/4d1786c68724c91c2686cda89540fa91b9c21bcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/4d1786c68724c91c2686cda89540fa91b9c21bcf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
