)]}'
{"id":"openstack%2Fhorizon~687813","triplet_id":"openstack%2Fhorizon~master~Id16c9fd4114316ade63c74097e486623d0c5be9f","project":"openstack/horizon","branch":"master","topic":"oslo-config-integration-tests","hashtags":[],"change_id":"Id16c9fd4114316ade63c74097e486623d0c5be9f","subject":"Avoid duplicated maintenance of integration test config","status":"MERGED","created":"2019-10-10 07:53:23.000000000","updated":"2020-01-27 01:46:05.000000000","submitted":"2020-01-27 01:44:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687813-1580089473083-04bc3352","meta_rev_id":"ffb0e0eefe52bde48970fb62e34524cf6d75fcce","_number":687813,"virtual_id_number":687813,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":23467,"name":"Pallav Gupta","email":"pallavgupta84@gmail.com","username":"pallav"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-27 01:44:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":2,"date":"2020-01-27 00:21:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2020-01-14 08:12:13.000000000","_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":1,"date":"2020-01-24 19:49:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23467,"name":"Pallav Gupta","email":"pallavgupta84@gmail.com","username":"pallav"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":1,"date":"2020-01-27 00:21:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":23467,"name":"Pallav Gupta","email":"pallavgupta84@gmail.com","username":"pallav"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":23467,"name":"Pallav Gupta","email":"pallavgupta84@gmail.com","username":"pallav"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23467,"name":"Pallav Gupta","email":"pallavgupta84@gmail.com","username":"pallav"},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-14 08:12:13.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-01-24 19:49:37.000000000","updated_by":{"_account_id":23467,"name":"Pallav Gupta","email":"pallavgupta84@gmail.com","username":"pallav"},"reviewer":{"_account_id":23467,"name":"Pallav Gupta","email":"pallavgupta84@gmail.com","username":"pallav"},"state":"REVIEWER"},{"updated":"2020-01-27 00:21:24.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2020-01-27 01:44:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bb19665e8a165d50dafbdc45151577bf2e8413f3","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-10-10 07:53:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"555178fb620534d133dce543ba43b6bc278afdf8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 09:09:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1b13bc6dccf644239d94575894c801eb : SUCCESS in 2m 12s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/4725732c1dc24c8eac08d9125de70d72 : SUCCESS in 4m 24s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/f4290928272845afa6f66af5528e0420 : SUCCESS in 5m 37s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/4093ce75053a458c80596195b259955a : SUCCESS in 9m 27s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/5419ff35afb9461ba0f8d5c723c061f3 : SUCCESS in 10m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/299df86a099e4d6890e12176976d9f9c : SUCCESS in 10m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9ce9bf7cde84ce1bea586cf31907de5 : SUCCESS in 5m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5bc57538d7f84e4dbfed1f06150b46f5 : SUCCESS in 8m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8501697c5c364318a1d68f25a51ad6f7 : SUCCESS in 9m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e7f6f006d36344b6afadeac06851a96d : SUCCESS in 9m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/878131ac38e747d58bab5298c4b00317 : SUCCESS in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/598f60bb4c78454781baabd807ef44cd : SUCCESS in 27m 17s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/7e1f8cd68a9645c6af70b686d16a0c76 : SUCCESS in 5m 18s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/b6ce35d823254bdcbd80b43be10eb44a : SUCCESS in 1h 09m 54s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/bdd5e04c4b4d41e8b8ab47f7494738c3 : SUCCESS in 34m 14s\n- horizon-dsvm-tempest-plugin-py27 https://zuul.opendev.org/t/openstack/build/707b09ab65c2473c82d7715503a8d0e0 : SUCCESS in 36m 32s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/63914da2cc554660a2d9a56565e802f5 : SUCCESS in 4m 25s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/15574c016da143d7b1792465f35b2975 : SUCCESS in 36m 27s","accounts_in_message":[],"_revision_number":1},{"id":"f22d1b12a8692708f6fd49fb329dd29c40516918","author":{"_account_id":23467,"name":"Pallav Gupta","email":"pallavgupta84@gmail.com","username":"pallav"},"date":"2019-10-11 18:40:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"83a1105511502470203a5aaddd47df6f13a465f3","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-10-25 10:08:03.000000000","message":"Patch Set 1: Code-Review+2\n\nGood cleanup!","accounts_in_message":[],"_revision_number":1},{"id":"073a71de3da3d02054a6baa133d480a962c29442","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-10-29 14:03:55.000000000","message":"Patch Set 1: Workflow-1\n\nMerge conflicts easily happen when we merge integration test patches from Paliav, so this should be -W until the current two +2\u0027ed fixes land.","accounts_in_message":[],"_revision_number":1},{"id":"fb93b91b38d8df54e8950ea54f9b1aa231df0af2","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-12-04 14:12:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b6ea915d038b7aa6228e857bfce9af647a1eea00","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-12-04 14:14:11.000000000","message":"Patch Set 2:\n\n\u003e Uploaded patch set 2.\n\nrebased on master and merge conflict resolved.","accounts_in_message":[],"_revision_number":2},{"id":"48a997b7a59c5daafbe64928822e888194a72f50","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-12-04 15:31:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"398706c692acb538afffd14b9ae01a12ca1188d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 15:38:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8392cea0d7a34cda8a8e2f37c0e56dc1 : SUCCESS in 2m 47s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/eef9ae004d1749498c26b8e4f75146de : SUCCESS in 5m 40s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/9622897731524d22b009ba5c40049aa9 : SUCCESS in 7m 52s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/a5cab291deba403aa188e62ff02715c6 : SUCCESS in 11m 27s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/93a4919e7e5b4a75b5e9c60c790bfb25 : SUCCESS in 9m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/514fded7c55f4fb89eb4c57266886a43 : SUCCESS in 9m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48506767c4154aa7b093b3cfab641856 : SUCCESS in 7m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb6f1ff76c934e06be32d7a48480ec5b : SUCCESS in 10m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bfb58392a74b42c08b0d90580f99a9fd : SUCCESS in 10m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a8f7eedc49046f9a381c26cb020bac2 : SUCCESS in 6m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15ab8efefe7a4f9da09d3445f8a4f9f6 : SUCCESS in 27m 47s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/4ed4a334bf2e4130bc59c563b2fd0cd8 : SUCCESS in 6m 00s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/b515ba0320f84ac79f185e8b472c7644 : SUCCESS in 1h 22m 53s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/c522c63b893d4e0888009e977e72a683 : SUCCESS in 33m 40s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7efad1d223484aeba839110421243e69 : SUCCESS in 5m 04s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/3054205728b049ad9fbe2b6b003aec02 : SUCCESS in 40m 49s","accounts_in_message":[],"_revision_number":2},{"id":"1dbbf2c03ced3613b67b12107a952cf5a931b2c8","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-09 15:55:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"573d0152192a81be6bb7eaaa8b5e7baa5640a771","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-09 15:56:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e4a604bbb465189cad2da7861f9068309b583f99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 17:37:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b707f65464f54631a4cff265b49fcd71 : SUCCESS in 4m 24s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/ff0921d4ec7b45cca083be1ae95a6a5d : SUCCESS in 6m 03s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/2a26c4bcaf2d497a957e76eb2179edae : SUCCESS in 11m 04s\n- horizon-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/267279d746af4f2289bfb6f09638b97f : SUCCESS in 12m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6df2487c0bbe4798ac3bfb86b53acbe5 : SUCCESS in 11m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a81c145359bf4793b9f65dc66f31fa70 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88ac94b925864e28920971873d9271f2 : SUCCESS in 11m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/33a50c75b22642f0a244f2f32611103f : SUCCESS in 10m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c3e3b58950a45a1b4d604181872727b : SUCCESS in 8m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5be442fc1a334356a713fda879e9c7c5 : SUCCESS in 26m 58s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/4ff7e95e1b38406a968cf0fc7a4f031a : SUCCESS in 7m 02s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/604095b9e5454e2b84d9ff88b1bd2943 : SUCCESS in 1h 35m 15s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/510ddacbda1c41feac28ff26aac0f0f1 : SUCCESS in 37m 46s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/21abade081a34f24a1f8cf29c0c98422 : SUCCESS in 5m 46s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/2876e47062ac40d4bd75bb66310a882d : SUCCESS in 36m 23s","accounts_in_message":[],"_revision_number":3},{"id":"f8d16a37a44ea1583fbcdb7190fb702853ea28a8","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-01-14 08:12:13.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"431ad8ca720fbcd0a29a0610b89ae18a225e23da","author":{"_account_id":23467,"name":"Pallav Gupta","email":"pallavgupta84@gmail.com","username":"pallav"},"date":"2020-01-24 19:49:37.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"2166be48a8a352266b59ee275d12f0bd84a96bac","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-01-27 00:21:24.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"df58ee9c8e4432e1ed0b97ffae85c6ce1687fffc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 00:21:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"185bf7ca530e2ae60723e75c37a27528a58aa00b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 01:44:32.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/15046aa8dc424682bed101f994167e35 : SUCCESS in 2m 24s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/0b4c34c8abd24447b7dacea6dd7e1364 : SUCCESS in 4m 57s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/418dc618f65e4a5f82603141f6ed9a6f : SUCCESS in 6m 22s\n- horizon-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/c8fac57d1a9b4392aa1d587968d5e229 : SUCCESS in 9m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e760dd783c044bb5863d82df47a6c717 : SUCCESS in 9m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fc5d81d96b14b218a1a84039f3a13f1 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4cb87ee70214a93af3ad2d6ce49fb3b : SUCCESS in 9m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d40b42e9becf403698e3462ae5958c9a : SUCCESS in 9m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9360fc0778c94ade8922e653c4051c74 : SUCCESS in 5m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ef4b74d56ac4a28b008a0e4dabd3df6 : SUCCESS in 24m 17s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/ae821ad50e7440099fbb04fb19ffb399 : SUCCESS in 32m 11s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/e9e8a4e5b9b540b1bf51fffa83af55f3 : SUCCESS in 6m 25s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/9790b9a158ec420ba9ac4cb724816fc6 : SUCCESS in 1h 20m 58s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/03a0a2fbbf304f32ad6df9a3d354e0b9 : SUCCESS in 31m 02s","accounts_in_message":[],"_revision_number":3},{"id":"8f4246ef9694c48f1aeca38c5975ca85e2801a4d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 01:44:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"21d71b747a9f2effe4a2f63f6a7b2ab46f4dc381","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 01:46:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4185c37a3d794a80a92c061fe901af6b : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5036b3723b2f4223bdea20817fdb93d3 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4ef9ba94c076f132a7d54cb5a3c8b2628cd1b378","revisions":{"20296351b3d7c02a9037d363c7386456e1c8b78d":{"kind":"REWORK","_number":1,"created":"2019-10-10 07:53:23.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/13/687813/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/13/687813/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/687813/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/687813/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/687813/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/13/687813/1"}}},"commit":{"parents":[{"commit":"307e884eaa799691f081d73362a287d20a1ad9a8","subject":"Fix grammar error in a recent release notes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/307e884eaa799691f081d73362a287d20a1ad9a8"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-10-04 06:52:00.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-10-10 07:53:04.000000000","tz":540},"subject":"Avoid duplicated maintenance of integration test config","message":"Avoid duplicated maintenance of integration test config\n\nConfigurations of the integration tests are maintained both\nin horizon.conf and in-code configuration. It is redundant.\n\nConfigurations of the integration tests are defined by oslo.config.\nWe can use default values from them and generate a sample config file\nusing oslo-config-generator.\n\nThis commit drops horizon.conf and generates the sample config.\nWe no longer have horizon.conf, so the logic to load the config files\nis updated accordingly.\n\nAlso makes the usage of single and double quotes in config.py consistent.\n\nChange-Id: Id16c9fd4114316ade63c74097e486623d0c5be9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/20296351b3d7c02a9037d363c7386456e1c8b78d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/20296351b3d7c02a9037d363c7386456e1c8b78d"}]},"branch":"refs/heads/master"},"1396e161cea80c16197ea7f9700f5966028a1015":{"kind":"REWORK","_number":2,"created":"2019-12-04 14:12:42.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/13/687813/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/13/687813/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/687813/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/687813/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/687813/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/13/687813/2"}}},"commit":{"parents":[{"commit":"621a954046cc6ab6726b1335729cedebe9180e89","subject":"Merge \"Change py27integration to integration.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/621a954046cc6ab6726b1335729cedebe9180e89"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-10-04 06:52:00.000000000","tz":540},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2019-12-04 14:11:40.000000000","tz":0},"subject":"Avoid duplicated maintenance of integration test config","message":"Avoid duplicated maintenance of integration test config\n\nConfigurations of the integration tests are maintained both\nin horizon.conf and in-code configuration. It is redundant.\n\nConfigurations of the integration tests are defined by oslo.config.\nWe can use default values from them and generate a sample config file\nusing oslo-config-generator.\n\nThis commit drops horizon.conf and generates the sample config.\nWe no longer have horizon.conf, so the logic to load the config files\nis updated accordingly.\n\nAlso makes the usage of single and double quotes in config.py consistent.\n\nChange-Id: Id16c9fd4114316ade63c74097e486623d0c5be9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/1396e161cea80c16197ea7f9700f5966028a1015"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/1396e161cea80c16197ea7f9700f5966028a1015"}]},"branch":"refs/heads/master"},"4ef9ba94c076f132a7d54cb5a3c8b2628cd1b378":{"kind":"REWORK","_number":3,"created":"2020-01-09 15:56:07.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/13/687813/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/13/687813/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/687813/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/687813/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/687813/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/13/687813/3"}}},"commit":{"parents":[{"commit":"0a4785c4894d2117343382e038494da8811dec2b","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/0a4785c4894d2117343382e038494da8811dec2b"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-10-04 06:52:00.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-01-09 15:55:43.000000000","tz":540},"subject":"Avoid duplicated maintenance of integration test config","message":"Avoid duplicated maintenance of integration test config\n\nConfigurations of the integration tests are maintained both\nin horizon.conf and in-code configuration. It is redundant.\n\nConfigurations of the integration tests are defined by oslo.config.\nWe can use default values from them and generate a sample config file\nusing oslo-config-generator.\n\nThis commit drops horizon.conf and generates the sample config.\nWe no longer have horizon.conf, so the logic to load the config files\nis updated accordingly.\n\nAlso makes the usage of single and double quotes in config.py consistent.\n\nChange-Id: Id16c9fd4114316ade63c74097e486623d0c5be9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/4ef9ba94c076f132a7d54cb5a3c8b2628cd1b378"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/4ef9ba94c076f132a7d54cb5a3c8b2628cd1b378"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
