)]}'
{"id":"openstack%2Fhorizon~690134","triplet_id":"openstack%2Fhorizon~master~Iebb1e78c718b931d632445e4de6d7a29ccb92be2","project":"openstack/horizon","branch":"master","topic":"bug/1849351","hashtags":[],"change_id":"Iebb1e78c718b931d632445e4de6d7a29ccb92be2","subject":"Remove dependency on novaclient list_extensions API","status":"MERGED","created":"2019-10-22 17:17:59.000000000","updated":"2019-11-05 22:15:23.000000000","submitted":"2019-10-25 14:08:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690134-1572012488747-fa70c53b","meta_rev_id":"5e4f23327ad86e3b47bfc40731adb8811e9e00ed","_number":690134,"virtual_id_number":690134,"owner":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-25 14:08:08.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2019-10-25 12:49:56.000000000","_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":2,"date":"2019-10-25 11:22:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":1,"date":"2019-10-25 12:49:56.000000000","_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-22 17:48:05.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2019-10-23 01:49:48.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2019-10-25 11:22:23.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-10-25 12:49:56.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2019-10-25 14:08:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"366a51a79b45c87ed359b4c1869443f82438231a","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-22 17:17:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a9f7d4dd14301c4f253ed205c30545bc2e707faf","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-10-22 17:48:05.000000000","message":"Patch Set 1:\n\nThanks for the patch but the same changes are already purposed in another patch [1].\n\n[1] https://review.opendev.org/#/c/689320/","accounts_in_message":[],"_revision_number":1},{"id":"0a88055c45dd1d7b601edc352468de3375311596","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-22 18:23:32.000000000","message":"Patch Set 1:\n\n\u003e Thanks for the patch but the same changes are already purposed in another patch [1].\n\nThanks, I wasn\u0027t aware of that. I left some comments on that since I think it\u0027s a much bigger change than this and also drops support for the blacklist configuration without removing that from the docs/default settings nor does it have a release note. In other words, my fix is much more targeted and probably quicker to land and unblock using novaclient 16.0.0 in upper-constraints.","accounts_in_message":[],"_revision_number":1},{"id":"cf058fb08e6587a3c874b2ab5cf39080d98c1be7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 18:28:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/4a1db007eeeb4dd488c51f2a2a1bf962 : SUCCESS in 4m 03s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/8cbe29a91c0a43e480f6990d671ca29a : SUCCESS in 5m 34s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/6045af9851de43b4b4c3a9212202a476 : FAILURE in 8m 23s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/cd1ac6c293b04f5db590477e014e2fae : FAILURE in 10m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0349db71ec8b434abcf476be4618eece : FAILURE in 10m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aae5dbdeed724536a841b508a7a507db : SUCCESS in 5m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/04a3a85555ad493bae5190a104d9bf9a : FAILURE in 10m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/836dba3c96534736b40ddbb7519f80fa : FAILURE in 8m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7919c5f2ea684fc2875da23ec8dbcd36 : FAILURE in 8m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5386b4c261134c9d9a069a20a1b78834 : SUCCESS in 5m 10s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/d4edc05e93974c36b9908a20226e7865 : SUCCESS in 6m 05s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/8a38df5cc068496dbd0f906618f0a5a1 : SUCCESS in 1h 05m 21s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/0f482888ff8b4cf09b8ba84ba3b39281 : FAILURE in 10m 08s\n- horizon-dsvm-tempest-plugin-py27 https://zuul.opendev.org/t/openstack/build/6360c2752ed54e7ba2569bf32e5c6130 : SUCCESS in 34m 20s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/af99b3e5bc744c1e9efdd50214f55ff5 : SUCCESS in 4m 20s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/33ee294c78b64534960a2222c15af1b0 : FAILURE in 12m 17s","accounts_in_message":[],"_revision_number":1},{"id":"44418edc2dbf294f7076b001fa982b6c63516923","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-22 20:26:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"18ac64a3b0e9c6fa97941e8e9ea7e15d7cb1dc5d","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-22 20:26:36.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"3736520ea3b050a2be35b020c3457f1b9fcf3b8a","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-22 21:32:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2c7328faa1ef41b9aa44e6f9ec166505962299ae","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-22 21:50:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bbdb2692aa2c3c4dea3a97b90fbf32af46945b9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 23:00:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/0c1c33ac72cf48188f19886978e6b84f : SUCCESS in 4m 36s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/19f82a828793439f840bf9b70467f137 : SUCCESS in 5m 57s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/9668ffc6595448889c4deb1d75bebaf6 : SUCCESS in 8m 33s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/cad1bf67d81c4c5abc7f8d5105fac810 : SUCCESS in 9m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5cdc6c18d7034862a2b63f00dcf55584 : SUCCESS in 10m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/728c9fccf86546e0aaa9722ae5dbd989 : SUCCESS in 5m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/591cb724839440b8a5b7ffaaee30eb69 : SUCCESS in 10m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fbce3e7d39d1481b86596056234441ef : SUCCESS in 8m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8496dfb1f0734f069af15fee2147500a : SUCCESS in 8m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2d763bba0834e69b41bb090ca777048 : SUCCESS in 5m 13s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/466b936db88c408e84ad01d5ff7df37b : SUCCESS in 5m 40s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/00b85786bb80454ea82adcd9e3fb9bcb : SUCCESS in 1h 05m 14s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/42ef4c4eacf6466ab90bacada6e1795a : SUCCESS in 36m 26s\n- horizon-dsvm-tempest-plugin-py27 https://zuul.opendev.org/t/openstack/build/179ea75b82174a3289580cf668b5dc13 : SUCCESS in 40m 50s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7224342ba29d4515ae9732d73fbaf403 : SUCCESS in 4m 02s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/a1bd20e25b88460ca9198f503b3f19fc : SUCCESS in 40m 02s","accounts_in_message":[],"_revision_number":2},{"id":"1be3abf24d505af049046403c7a91951aa671e97","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2019-10-23 01:49:48.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"245f75c6e2fc79c9be14230df73d54d57c2f113c","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-10-23 06:13:42.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"8296ca5f1153c7565628c5261b027a14e988db78","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-23 16:02:20.000000000","message":"Patch Set 2: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6a5ddb5a3518fff57e4a09f722ab302b2a93ceb4","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-23 16:04:04.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"0f79ffe2e63eb0d215fe6d9136a337559ee9ddfb","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-10-23 16:14:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6dad31bbfefbcc8bb480bb4dd36e99612988ec93","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-23 16:33:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"29ee5b8c361677eba59683436e643bf5d0512284","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-23 16:37:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0f31cbdb2ac53800cef0532908d9fc23fad01b48","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-23 16:37:47.000000000","message":"Patch Set 3: Workflow-1\n\nWIP until we sort out the javascript code issue.","accounts_in_message":[],"_revision_number":3},{"id":"0cd35060839125a7e023ba37dbdabea7c8a79775","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 17:55:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/ceb2671d14ff4817b71700c04d312fc3 : SUCCESS in 4m 37s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/017c19cbd8bd4dfd92bf418f68087c01 : SUCCESS in 5m 36s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/bf48051454004daab7848a1c5ab4d3a3 : SUCCESS in 9m 27s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/f176c7013859457180b8f6b1626cfa49 : SUCCESS in 9m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/747aad2f85034de593778c7d4fa0bf38 : SUCCESS in 9m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/897e7a3cf4ef44a5bcde7b49c747ced9 : SUCCESS in 6m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/49b9c46dc5bd4c9b9df6c304379a8079 : SUCCESS in 11m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a1f0fa7f0fd41099087e5c29310c854 : SUCCESS in 8m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2aa670b2c28f4d9f8dd530083bfc4b11 : SUCCESS in 10m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59b5a2e84fab4f91bdc0811395bb42cb : SUCCESS in 7m 46s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/5c0165df79714ab88975870fa57b6bb8 : SUCCESS in 5m 53s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/825d2d0e5e1342b893428d4b9b273ffd : SUCCESS in 1h 14m 27s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/c35ece1d7a6947018dfebcdf3de90e24 : SUCCESS in 38m 44s\n- horizon-dsvm-tempest-plugin-py27 https://zuul.opendev.org/t/openstack/build/4ef28e2326de4e6a9f33b26df4935814 : SUCCESS in 38m 22s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/360cbcff8e764144ba5544887e60796f : SUCCESS in 4m 28s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/ea692b27b6c441d5ac2805c5c51e4062 : SUCCESS in 37m 57s","accounts_in_message":[],"_revision_number":3},{"id":"933690dd936fc93a05ae65682828a09311d8474d","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-10-24 13:31:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5e51af20fbdef5a74dc750208603ab92bd81928d","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-24 15:33:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"806b971596adb98c51101f6b645498367d6633a9","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-10-24 16:28:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8f49bf4591410395e587c0ea7fa1ba8d9a0d7d77","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-24 17:17:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"065e0fe1326cb05a07a9e2771b53c940e600ca70","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-24 19:18:46.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"af1e505a3871a92c228a1c4f73f051256afda36b","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-24 19:19:55.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"a48b7df9488289620279c7b7e4c82725f8258b54","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-24 19:20:45.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"55c5b0887de4ec249efb9c9a9e7f5d097800dea5","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-24 19:24:48.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c080526481310e785a2e79689a779bd343f9399c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 20:52:43.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/50b8df6ab9364130b81d288a30effa44 : SUCCESS in 4m 06s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/9326a2a97768473fad8c299ef1faba77 : SUCCESS in 7m 49s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/76da66e5d3794b29ac399dcd06b2c31e : SUCCESS in 9m 27s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/b3d36261c5ab4558b152ff260c09fc74 : SUCCESS in 8m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc84bca558dd465c9742726f4e41107c : SUCCESS in 11m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34e3d4d79ddf4457a765a17b707bfb11 : SUCCESS in 6m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1fad1e44846249b1862e3fc795fbfded : SUCCESS in 10m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bef042a1e4514804a4441bcd36c2a828 : SUCCESS in 12m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b19872558d134ba6aa99fb0833e3bfd5 : SUCCESS in 10m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0efcb67d1fcd4aae881cac9a375d7757 : SUCCESS in 7m 55s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/658626ca64204646893ec95aab0746f5 : SUCCESS in 6m 45s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/95c20c4601e442c6a3358184480e7810 : SUCCESS in 1h 21m 03s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/399e9f8d39c8418cbd9e8931ce5b8411 : SUCCESS in 39m 22s\n- horizon-dsvm-tempest-plugin-py27 https://zuul.opendev.org/t/openstack/build/cbd6e1c9ddbf4bcb95ec62ec6243b1f2 : SUCCESS in 42m 06s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5702094500c846b4848f451f3c5a4537 : SUCCESS in 5m 33s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/c232325fb06d46168683d93b7d7401f1 : SUCCESS in 34m 26s","accounts_in_message":[],"_revision_number":5},{"id":"58ad498e2183e73406c991756fc9444c00efd9c6","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-10-25 11:22:23.000000000","message":"Patch Set 5: Code-Review+2\n\nThanks! I confirmed this works well.","accounts_in_message":[],"_revision_number":5},{"id":"65a9783f5f8f4b9bfd2b6c6059882f467cd92b5b","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-10-25 12:49:56.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nThank you, Matt!","accounts_in_message":[],"_revision_number":5},{"id":"6ad6212ebbb0887d112f1e911230aada39323dee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 12:50:09.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"d2873bf12a41d6c04e562ca147796968b9279584","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 14:08:08.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/ad45a392f3654e3d80100e42b520ebf5 : SUCCESS in 5m 11s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/9eea23f4446442ac96109203cb406280 : SUCCESS in 5m 53s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/e2a8a3124ef34e658da77121a32f120a : SUCCESS in 9m 10s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/f3029e427fe64d93b3e481da3dd21698 : SUCCESS in 10m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72af323e5ba94e35830886716959861f : SUCCESS in 10m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cca25da7a6c54535a708b6d4d86f60fd : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5bf4a5cb85854de282eca07ac107b29e : SUCCESS in 9m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/88a1d5d31a584655a2b2710bf3473032 : SUCCESS in 9m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae4bb2ac5e82402cb94303df3aad8c4e : SUCCESS in 5m 34s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/58753a6936bd4f998e85d86e33e69c28 : SUCCESS in 33m 45s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/28067711cfa44e4f821892666bf0af67 : SUCCESS in 5m 51s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/8e31c368e7be4a2c83b9547d5bf5574d : SUCCESS in 1h 12m 38s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/ae3f77e2db764af9b453d263609c993c : SUCCESS in 39m 54s","accounts_in_message":[],"_revision_number":5},{"id":"0652791bd2ae44125a079621199b854e5cbc2609","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 14:08:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"baf159c491337c49f30bd3aa02ec186705adb429","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 14:09:41.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/060517ad18574eb4b33ce2ec8223a782 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":5},{"id":"3e41f935b6fed613d5f97eb18b33a7a6e7ad87f9","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-11-05 22:15:23.000000000","message":"Patch Set 5:\n\namotoki, e0ne: Are you planning a release for horizon with this fix? manila-ui test jobs are failing at the moment but don\u0027t get this fix because they\u0027re pulling in horizon 16.0.0... \n\n[1] https://review.opendev.org/#/c/692081/","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"b148c9207580731863e4dc49771e132a8c31edbc","revisions":{"82e17675d984df5e4b001ce572ff5ef5417747dd":{"kind":"REWORK","_number":1,"created":"2019-10-22 17:17:59.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/34/690134/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/34/690134/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/34/690134/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/34/690134/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/34/690134/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/34/690134/1"}}},"commit":{"parents":[{"commit":"85a1dddf126691921924edcecaee5c054c7df6c2","subject":"Fix aes-xts key length in Horizon Admin Guide / Manage Volumes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/85a1dddf126691921924edcecaee5c054c7df6c2"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-10-22 17:14:13.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-10-22 17:14:13.000000000","tz":-240},"subject":"Remove dependency on novaclient list_extensions API","message":"Remove dependency on novaclient list_extensions API\n\nThe novaclient list_extensions API binding was removed in the\n16.0.0 release [1]. The ability to enable/disable extensions\nin nova has been deprecated since Liberty [2] and was removed\nin Newton [3].\n\nFor horizon this only matters for the OPENSTACK_NOVA_EXTENSIONS_BLACKLIST\nconfig setting. In order to work with novaclient 16.0.0+, this\nchange removes the list_extensions usage since all extensions\nare enabled in nova and thus for horizon a nova extension is only\nnot supported if it\u0027s in the configured blacklist.\n\n[1] https://review.opendev.org/686516/\n[2] https://review.opendev.org/214592/\n[3] https://review.opendev.org/351362/\n\nChange-Id: Iebb1e78c718b931d632445e4de6d7a29ccb92be2\nCloses-Bug: #1849351\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/82e17675d984df5e4b001ce572ff5ef5417747dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/82e17675d984df5e4b001ce572ff5ef5417747dd"}]},"branch":"refs/heads/master"},"b4f934bfbb4bdf581e0c2a3ea69edd9e692ca129":{"kind":"REWORK","_number":2,"created":"2019-10-22 21:50:15.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/34/690134/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/34/690134/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/34/690134/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/34/690134/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/34/690134/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/34/690134/2"}}},"commit":{"parents":[{"commit":"85a1dddf126691921924edcecaee5c054c7df6c2","subject":"Fix aes-xts key length in Horizon Admin Guide / Manage Volumes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/85a1dddf126691921924edcecaee5c054c7df6c2"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-10-22 17:14:13.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-10-22 21:50:06.000000000","tz":-240},"subject":"Remove dependency on novaclient list_extensions API","message":"Remove dependency on novaclient list_extensions API\n\nThe novaclient list_extensions API binding was removed in the\n16.0.0 release [1]. The ability to enable/disable extensions\nin nova has been deprecated since Liberty [2] and was removed\nin Newton [3].\n\nFor horizon this only matters for the OPENSTACK_NOVA_EXTENSIONS_BLACKLIST\nconfig setting. In order to work with novaclient 16.0.0+, this\nchange removes the list_extensions usage since all extensions\nare enabled in nova and thus for horizon a nova extension is only\nnot supported if it\u0027s in the configured blacklist.\n\n[1] https://review.opendev.org/686516/\n[2] https://review.opendev.org/214592/\n[3] https://review.opendev.org/351362/\n\nChange-Id: Iebb1e78c718b931d632445e4de6d7a29ccb92be2\nCloses-Bug: #1849351\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/b4f934bfbb4bdf581e0c2a3ea69edd9e692ca129"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/b4f934bfbb4bdf581e0c2a3ea69edd9e692ca129"}]},"branch":"refs/heads/master"},"54b11af9508799c2f990702c06274f6fbe2854b1":{"kind":"REWORK","_number":3,"created":"2019-10-23 16:33:32.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/34/690134/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/34/690134/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/34/690134/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/34/690134/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/34/690134/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/34/690134/3"}}},"commit":{"parents":[{"commit":"85a1dddf126691921924edcecaee5c054c7df6c2","subject":"Fix aes-xts key length in Horizon Admin Guide / Manage Volumes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/85a1dddf126691921924edcecaee5c054c7df6c2"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-10-22 17:14:13.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-10-23 16:33:11.000000000","tz":-240},"subject":"Remove dependency on novaclient list_extensions API","message":"Remove dependency on novaclient list_extensions API\n\nThe novaclient list_extensions API binding was removed in the\n16.0.0 release [1]. The ability to enable/disable extensions\nin nova has been deprecated since Liberty [2] and was removed\nin Newton [3].\n\nFor horizon this only matters for the OPENSTACK_NOVA_EXTENSIONS_BLACKLIST\nconfig setting. In order to work with novaclient 16.0.0+, this\nchange removes the list_extensions usage since all extensions\nare enabled in nova and thus for horizon a nova extension is only\nnot supported if it\u0027s in the configured blacklist.\n\n[1] https://review.opendev.org/686516/\n[2] https://review.opendev.org/214592/\n[3] https://review.opendev.org/351362/\n\nChange-Id: Iebb1e78c718b931d632445e4de6d7a29ccb92be2\nCloses-Bug: #1847959\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/54b11af9508799c2f990702c06274f6fbe2854b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/54b11af9508799c2f990702c06274f6fbe2854b1"}]},"branch":"refs/heads/master"},"6273231c009e987ecc0dd3cf300d895ae2d3fbb6":{"kind":"REWORK","_number":4,"created":"2019-10-24 19:18:46.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/34/690134/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/34/690134/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/34/690134/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/34/690134/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/34/690134/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/34/690134/4"}}},"commit":{"parents":[{"commit":"85a1dddf126691921924edcecaee5c054c7df6c2","subject":"Fix aes-xts key length in Horizon Admin Guide / Manage Volumes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/85a1dddf126691921924edcecaee5c054c7df6c2"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-10-22 17:14:13.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-10-24 19:16:37.000000000","tz":-240},"subject":"Remove dependency on novaclient list_extensions API","message":"Remove dependency on novaclient list_extensions API\n\nThe novaclient list_extensions API binding was removed in the\n16.0.0 release [1]. The ability to enable/disable extensions\nin nova has been deprecated since Liberty [2] and was removed\nin Newton [3].\n\nFor horizon this only matters for the OPENSTACK_NOVA_EXTENSIONS_BLACKLIST\nconfig setting and some javascript code used to compile panels based on\nenabled extensions. In order to work with novaclient 16.0.0+, this\nchange removes the list_extensions usage since all extensions\nare enabled in nova and thus for horizon a nova extension is only\nnot supported if it\u0027s in the configured blacklist. To continue supporting\nthe javascript code which uses the getExtensions function, the extension\nnames are hard-coded. Note that the method meant to test that code,\n_test_extension_list, was wrong but never ran because of the underscore\nprefix on the method name. That is fixed here.\n\n[1] https://review.opendev.org/686516/\n[2] https://review.opendev.org/214592/\n[3] https://review.opendev.org/351362/\n\nChange-Id: Iebb1e78c718b931d632445e4de6d7a29ccb92be2\nCloses-Bug: #1847959\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/6273231c009e987ecc0dd3cf300d895ae2d3fbb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/6273231c009e987ecc0dd3cf300d895ae2d3fbb6"}]},"branch":"refs/heads/master"},"b148c9207580731863e4dc49771e132a8c31edbc":{"kind":"REWORK","_number":5,"created":"2019-10-24 19:24:48.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/34/690134/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/34/690134/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/34/690134/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/34/690134/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/34/690134/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/34/690134/5"}}},"commit":{"parents":[{"commit":"85a1dddf126691921924edcecaee5c054c7df6c2","subject":"Fix aes-xts key length in Horizon Admin Guide / Manage Volumes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/85a1dddf126691921924edcecaee5c054c7df6c2"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-10-22 17:14:13.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-10-24 19:24:39.000000000","tz":-240},"subject":"Remove dependency on novaclient list_extensions API","message":"Remove dependency on novaclient list_extensions API\n\nThe novaclient list_extensions API binding was removed in the\n16.0.0 release [1]. The ability to enable/disable extensions\nin nova has been deprecated since Liberty [2] and was removed\nin Newton [3].\n\nFor horizon this only matters for the OPENSTACK_NOVA_EXTENSIONS_BLACKLIST\nconfig setting and some javascript code used to compile panels based on\nenabled extensions. In order to work with novaclient 16.0.0+, this\nchange removes the list_extensions usage since all extensions\nare enabled in nova and thus for horizon a nova extension is only\nnot supported if it\u0027s in the configured blacklist. To continue supporting\nthe javascript code which uses the getExtensions function, the extension\nnames are hard-coded. Note that the method meant to test that code,\n_test_extension_list, was wrong but never ran because of the underscore\nprefix on the method name. That is fixed here.\n\n[1] https://review.opendev.org/686516/\n[2] https://review.opendev.org/214592/\n[3] https://review.opendev.org/351362/\n\nChange-Id: Iebb1e78c718b931d632445e4de6d7a29ccb92be2\nCloses-Bug: #1847959\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/b148c9207580731863e4dc49771e132a8c31edbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/b148c9207580731863e4dc49771e132a8c31edbc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
