)]}'
{"id":"openstack%2Fhorizon~698191","triplet_id":"openstack%2Fhorizon~stable%2Ftrain~Ic563fa8f86e155ef05222dd152ddcc5c0b712081","project":"openstack/horizon","branch":"stable/train","topic":"zanata/translations","hashtags":[],"change_id":"Ic563fa8f86e155ef05222dd152ddcc5c0b712081","subject":"Imported Translations from Zanata","status":"MERGED","created":"2019-12-10 07:17:39.000000000","updated":"2019-12-19 17:51:46.000000000","submitted":"2019-12-19 17:49:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698191-1576777786261-a981c286","meta_rev_id":"e3e323460b026f396f4b9afd4b9c0a07787d89e3","_number":698191,"virtual_id_number":698191,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-19 17:49:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":2,"date":"2019-12-19 16:28:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":1,"date":"2019-12-19 16:28:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-14 18:15:18.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2019-12-19 16:28:27.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-12-19 17:49:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a84842da577f278f3af98025290f9ebb055762a1","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-10 07:17:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f98951ceb1b6e6cbdf0d9d4429dafe13608a98a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 08:25:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/dd4777a178dc49a39b6f634eab4e3e51 : SUCCESS in 4m 29s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/afe249d3b22043f9968a43bc93e3ab61 : SUCCESS in 12m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0dd9ad2c9fc14ff2805e56facbdcaa15 : SUCCESS in 11m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75de7114409c4e6895168a45c17edbf8 : SUCCESS in 5m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/24aee245076d4900ae9919c809d22841 : SUCCESS in 9m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f871510a3720439ca17063ec839eec28 : SUCCESS in 9m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/920395b1e34c4c0283870475b64c5a38 : SUCCESS in 8m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dd50bb54d0049439c2116fce75bee2e : SUCCESS in 6m 32s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/edb7cf2effca4d79911a2925be124d81 : SUCCESS in 1h 05m 47s","accounts_in_message":[],"_revision_number":1},{"id":"4cef1730c6d5c231dd96ee09b7e51a218aa9b43d","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-12 08:39:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"df7c4e586489d9fa87f40aaf6c539941f9d91fca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 10:00:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/a7258267fd784d4e99ade7a9534c43da : SUCCESS in 5m 49s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/ce50a348636c46a386efc72496853746 : SUCCESS in 7m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1343cf86a32a4e76bb0bb26e268dfe00 : SUCCESS in 10m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51032b8b2eed42fc992e749aee7652a3 : SUCCESS in 6m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7ec094a759534c70b217bad0273e6440 : SUCCESS in 10m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/698c75625fe7488eb6834360ab5c089f : SUCCESS in 9m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/09626eca3db948fa8c91420870496a67 : SUCCESS in 9m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70563d5c698c4fe0b101499a2be4a709 : SUCCESS in 6m 15s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/8e77704a92334c3dba7597821a510582 : SUCCESS in 1h 15m 30s","accounts_in_message":[],"_revision_number":2},{"id":"597487f05d7560dc683a47abcfee41d25d4cf1db","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-14 07:55:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7c742aa07c081a3b47eab33807d9e8057801f373","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-14 09:11:57.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/43ae47a897764ad9816adcfec6cf0515 : SUCCESS in 5m 12s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/44bcec1cd5034261ad61325da88df2a4 : SUCCESS in 6m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e6cfe0121b1e42e5bdc46c58a8bae35e : SUCCESS in 8m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1e350d7811e4349b77da0e6a6d33b87 : FAILURE in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a3de0afa2f804ccc93a142f00341d2fa : SUCCESS in 11m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0dc719873aff4fbf9fc18915fc7e1147 : SUCCESS in 9m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e0372b02c6c8421dbabed45df8ff3f01 : SUCCESS in 9m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a41af3375cb4fd5b7ae4f8a92ec5530 : SUCCESS in 7m 22s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/1f92229b718145da95de6d327ad50f16 : SUCCESS in 1h 12m 08s","accounts_in_message":[],"_revision_number":3},{"id":"9bec69730786beb1accb95d0720fc7ac6c7dc56c","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-12-14 18:15:18.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"5a54200c6eba8ce3fd586207aa7e82cbe85274f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-14 19:27:09.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/a8ce499930dd4f559db61578949bb054 : SUCCESS in 4m 56s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/7fbe82d1f990404cbf129e80b1a48884 : SUCCESS in 6m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2d2ee853678467b8c7f89e3d30ebd92 : SUCCESS in 11m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46efc7908e704d0d903ef5c74c74939a : FAILURE in 4m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cbac6b746a474632b92ae62aafee6467 : SUCCESS in 11m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/247f7f9e6e644684a77680678ce8aaa2 : SUCCESS in 9m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ccf051b5474472c869e175007256f1e : SUCCESS in 7m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d26aef2b2a34a5fb9eeeca456947abd : SUCCESS in 6m 23s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/0680882237d4444b9aa8cb53589962ad : SUCCESS in 1h 06m 38s","accounts_in_message":[],"_revision_number":3},{"id":"c6c668f992cc761c122f5fe4391311f054631ad2","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-17 08:04:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"de0acaf300e21d2ec928168e0ccfca9afca8ceed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 09:22:13.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/a07955032a8c4a52819139f18357c488 : SUCCESS in 6m 21s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/327b72c10f004b67a0045b95dface351 : SUCCESS in 7m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c63507f5d164c648b64d5a2a75de38c : SUCCESS in 12m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4382b68dce7b444594f93214392512c7 : FAILURE in 4m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bcf393fc7ec74afe9aef7080cebaabaf : SUCCESS in 12m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67b41d5a89454e66a207ad86c95e0123 : SUCCESS in 11m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6d40e7009b824ed79aa2503ae3e49e78 : SUCCESS in 11m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80d5d3f9a9354acaa581a3de35467a87 : SUCCESS in 6m 49s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/27e822a70cc941239c3ccb1ac52934b8 : SUCCESS in 1h 15m 52s","accounts_in_message":[],"_revision_number":4},{"id":"8b0feb121818caa49c610b37d69d24afa4c0b31b","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-18 08:41:31.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b59daeed15d977e7f767eb82f6798ba8c879efb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 09:47:34.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/1eef2c59284143efafdda0754a7c4b8c : SUCCESS in 5m 35s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/dd5f0d242ad44c1483ed9d439cf92656 : SUCCESS in 9m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a620b82eff524cfeaaeebf645c58d663 : SUCCESS in 9m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e756792129d4ae0856a0a334b4338cb : FAILURE in 6m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/95e2b68812814365aa629a9159366f4e : SUCCESS in 10m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61b2293992314d4496c3423323f27267 : SUCCESS in 9m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/134c02e6396f4fbb9c1fa326afb4eadc : SUCCESS in 10m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2fd36f03f284a0ea72ca092238f21b0 : SUCCESS in 7m 46s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/c9b3c55687c44e0282983018229c0a6f : SUCCESS in 1h 03m 15s","accounts_in_message":[],"_revision_number":5},{"id":"ef7a61e66d6149bd06f4c9ef03a4f12e1126b4f5","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-12-19 14:34:44.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"b023e19efa4976a6a52c1053229a445bedea1937","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 15:49:44.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/998cf1a41ce34f1fb8e07d2c9c791ee3 : SUCCESS in 5m 56s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/d9b2ec36eaf24ad5ac9c5b4ff143ee05 : SUCCESS in 10m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d59fc71ffa0c48babb4d7275c44298bf : SUCCESS in 11m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c7b96346b904a14a340d4bd2fd661fd : SUCCESS in 6m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3a536386bbba427c9f4e58861c6808fe : SUCCESS in 11m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9af0ec3b4904189b8ff1cb5f2ade02d : SUCCESS in 10m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cee3d22cbd6142fcbfe0c4e7473332fa : SUCCESS in 9m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbab6159df674021995c54009aa55033 : SUCCESS in 7m 16s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/a72ce0727ab2402e85d3f9631a185633 : SUCCESS in 1h 08m 11s","accounts_in_message":[],"_revision_number":5},{"id":"c67c174e18e36e9c09cba183b8662356cee93edc","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-12-19 16:28:27.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"172d75b6d56be9e4b6b363b55c06e557010a9835","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 16:28:40.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"7066290d882e17d86583f53d81a8573c3ce4acda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 17:49:46.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/dadf7b62f4454e63b677d5a61dd37d8c : SUCCESS in 5m 52s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/f529cf2dbb6e48d9b4f42f4df46c99ec : SUCCESS in 6m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77ebb46d41274491ab7da2f37b49d05b : SUCCESS in 14m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed3164d338744f398a78c9f79bd76fc0 : SUCCESS in 7m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/31849f8ae4064aa8b3ea62e3e8bb3b52 : SUCCESS in 14m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1bc9382781624f93ae51b96c632161a1 : SUCCESS in 10m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/67c1b8a67d45458a80bf4e1af89240b4 : SUCCESS in 10m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d4201e37cd7435392d25301558445cd : SUCCESS in 7m 11s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/61c9336f976f4875ac42acc5c0567e29 : SUCCESS in 1h 14m 11s","accounts_in_message":[],"_revision_number":5},{"id":"b17b3de05d68581c53a44c229ec6bc5c958e9c90","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 17:49:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"3221b9c6bafffe8ed8ae4031246a95ff1673160f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 17:51:46.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66632bfbd43342b6a21fe8b3f8eb5077 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"a122315c73dd488e47006436902a6d3205988e84","revisions":{"d19d72d6c1cca5c6a45c7f210a916fd1d8807b9d":{"kind":"REWORK","_number":1,"created":"2019-12-10 07:17:39.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/91/698191/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/91/698191/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/698191/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/698191/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/698191/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/91/698191/1"}}},"commit":{"parents":[{"commit":"4b693661b565550207c6556eedbbc7414f9c7848","subject":"Merge \"Fix change expired password feature\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/4b693661b565550207c6556eedbbc7414f9c7848"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-10 07:17:34.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-10 07:17:34.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Ic563fa8f86e155ef05222dd152ddcc5c0b712081\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/d19d72d6c1cca5c6a45c7f210a916fd1d8807b9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/d19d72d6c1cca5c6a45c7f210a916fd1d8807b9d"}]},"branch":"refs/heads/stable/train"},"664b30546e731f782c5042b73b7f419bc339c289":{"kind":"REWORK","_number":2,"created":"2019-12-12 08:39:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/91/698191/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/91/698191/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/698191/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/698191/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/698191/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/91/698191/2"}}},"commit":{"parents":[{"commit":"4b693661b565550207c6556eedbbc7414f9c7848","subject":"Merge \"Fix change expired password feature\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/4b693661b565550207c6556eedbbc7414f9c7848"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-12 08:39:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-12 08:39:31.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Ic563fa8f86e155ef05222dd152ddcc5c0b712081\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/664b30546e731f782c5042b73b7f419bc339c289"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/664b30546e731f782c5042b73b7f419bc339c289"}]},"branch":"refs/heads/stable/train"},"14d57011f03aa13a0460760f8972a0ee9001c711":{"kind":"REWORK","_number":3,"created":"2019-12-14 07:55:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/91/698191/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/91/698191/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/698191/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/698191/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/698191/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/91/698191/3"}}},"commit":{"parents":[{"commit":"4b693661b565550207c6556eedbbc7414f9c7848","subject":"Merge \"Fix change expired password feature\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/4b693661b565550207c6556eedbbc7414f9c7848"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-14 07:55:34.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-14 07:55:34.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Ic563fa8f86e155ef05222dd152ddcc5c0b712081\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/14d57011f03aa13a0460760f8972a0ee9001c711"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/14d57011f03aa13a0460760f8972a0ee9001c711"}]},"branch":"refs/heads/stable/train"},"99a46fea6bddb119552742b2efd65858507ca3c1":{"kind":"REWORK","_number":4,"created":"2019-12-17 08:04:18.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/91/698191/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/91/698191/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/698191/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/698191/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/698191/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/91/698191/4"}}},"commit":{"parents":[{"commit":"4b693661b565550207c6556eedbbc7414f9c7848","subject":"Merge \"Fix change expired password feature\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/4b693661b565550207c6556eedbbc7414f9c7848"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-17 08:04:16.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-17 08:04:16.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Ic563fa8f86e155ef05222dd152ddcc5c0b712081\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/99a46fea6bddb119552742b2efd65858507ca3c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/99a46fea6bddb119552742b2efd65858507ca3c1"}]},"branch":"refs/heads/stable/train"},"a122315c73dd488e47006436902a6d3205988e84":{"kind":"REWORK","_number":5,"created":"2019-12-18 08:41:31.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/91/698191/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/91/698191/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/698191/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/698191/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/698191/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/91/698191/5"}}},"commit":{"parents":[{"commit":"4b693661b565550207c6556eedbbc7414f9c7848","subject":"Merge \"Fix change expired password feature\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/4b693661b565550207c6556eedbbc7414f9c7848"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-18 08:41:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-18 08:41:28.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Ic563fa8f86e155ef05222dd152ddcc5c0b712081\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/a122315c73dd488e47006436902a6d3205988e84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/a122315c73dd488e47006436902a6d3205988e84"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
