)]}'
{"id":"openstack%2Fhorizon~698652","triplet_id":"openstack%2Fhorizon~stable%2Fstein~I0d31c7406027d7396ec341818eb154f6d1101938","project":"openstack/horizon","branch":"stable/stein","topic":"zanata/translations","hashtags":[],"change_id":"I0d31c7406027d7396ec341818eb154f6d1101938","subject":"Imported Translations from Zanata","status":"MERGED","created":"2019-12-12 08:09:51.000000000","updated":"2019-12-21 06:31:14.000000000","submitted":"2019-12-21 06:28:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698652-1576909684886-cc347a4a","meta_rev_id":"fafc896996d23337cb5e7f34fe184b208bc01eaa","_number":698652,"virtual_id_number":698652,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-21 06:28:04.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":2,"date":"2019-12-21 05:20:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":1,"date":"2019-12-21 05:20:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-14 18:15:02.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2019-12-21 05:20:08.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-12-21 06:28:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a4db1ed6c7438a43689d0de00512cd5b537efb8f","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-12 08:09:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3f108c6367798f129d5104590d9d175b01c8e26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 09:03:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs4-npm-run-lint https://zuul.opendev.org/t/openstack/build/27967fb26f2b40c7a61fd3f260401f25 : SUCCESS in 5m 09s\n- nodejs4-npm-run-test https://zuul.opendev.org/t/openstack/build/866ecbc6d9784e189182b515f9c95e8b : SUCCESS in 8m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57e09452c7fc46f0813674b8c4fd6e55 : SUCCESS in 9m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed13f1aad08344b385f98c2edfaae5ea : SUCCESS in 6m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a14e92ab42744f5eb285117813405d6b : SUCCESS in 10m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/726cdf3cdedd462ba6aae71d2a16d9c9 : SUCCESS in 9m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b322169ff0734753b0dfe73cee6e11d0 : SUCCESS in 7m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ed22c41057974d6ca4921d191f73fb3e : SUCCESS in 10m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93c859e30e824ad0bbf79ab00603336f : SUCCESS in 5m 01s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/92d63c0cdd2f47e6963bb4dafa23a0c8 : SUCCESS in 52m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fcf1e29a4b4b2a1b4ea1408cce73f39711f955e2","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2019-12-14 18:15:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3f9e16af7a7303bd0b8662a97f211e1a22824576","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-12-16 06:22:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cb0dde3bc656dfcf528e0e69da66aa33ca8e6837","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 08:07:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- nodejs4-npm-run-lint https://zuul.opendev.org/t/openstack/build/47c4abfb129a464fbbda6136f49a33ab : SUCCESS in 6m 44s\n- nodejs4-npm-run-test https://zuul.opendev.org/t/openstack/build/f3b10a9658944ecab45111dd3c525f85 : SUCCESS in 8m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e13784f40864062b6074cc0c1179dfd : FAILURE in 10m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d24e4781ad714bf484bcb69c686329c9 : FAILURE in 8m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/07eaf55efa504f2fb8b6c9f47d374252 : SUCCESS in 13m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3181ee6fd35d40f3a1e6a6e7c2374738 : SUCCESS in 10m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2292c934399049aa9fdb66832014ebbc : SUCCESS in 13m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0e533f8d2bbe47d39ec71c4d8ead6101 : SUCCESS in 10m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0be930f45c3b4e16a31809b2bb043593 : SUCCESS in 6m 22s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/4aab9cd310ca478b84b64819cedd413d : SUCCESS in 1h 01m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e37537d7f11e49be976e8ead7fa414a9c5e45057","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-12-20 08:43:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a96a5aa143a3bf42eb517cf762206604ff16f6d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 09:52:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs4-npm-run-lint https://zuul.opendev.org/t/openstack/build/0ee5b2c455214f9a9e1156bda7fc8921 : SUCCESS in 6m 10s\n- nodejs4-npm-run-test https://zuul.opendev.org/t/openstack/build/41f31d0ffadc4dfeb80d1c882a78d9dd : SUCCESS in 16m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1e05fae8a34b41478082c6e71ec2b3fe : SUCCESS in 9m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecafc8146c544b37bead23dfaa67690a : SUCCESS in 6m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2b840f4f045f42cb80a6e87d24f12704 : SUCCESS in 12m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2c835b9b4dcb4f35a089d4ee7a62ed1d : SUCCESS in 8m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7fdda6ef9ff448049aeb9f3685f8d152 : SUCCESS in 9m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b7929f6b312b4cb5b037d0094806c34e : SUCCESS in 13m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51fbf08ee4ea40378d6dfb533b3923b5 : SUCCESS in 6m 31s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/559bc533fac9417faa727ab63728471e : FAILURE in 59m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6754c523c09b64c757e98e52e6d4564e01146d4c","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-12-21 05:20:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f35ae30e8bf75c485c0889818eb1d5357d380158","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-21 05:20:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6356aeee075cf6f44d48e6833dd400bbcfeafa02","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-21 06:28:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"4f8fdb3ceb9cdfd83470225a4367d8039f887b82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-21 06:28:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- nodejs4-npm-run-lint https://zuul.opendev.org/t/openstack/build/6606343b13fd4e1c993f4e828ebccd83 : SUCCESS in 4m 42s\n- nodejs4-npm-run-test https://zuul.opendev.org/t/openstack/build/41d20f0f976a48e78124195ecd5aa8ef : SUCCESS in 7m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92bfc82c64424e9faeb61ed2ffec471d : SUCCESS in 8m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9f8b808dd9a4a508cfc6dc1f1f415ee : SUCCESS in 5m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/977336d040054a3b9dd221c5ecd3033b : SUCCESS in 10m 15s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ac40ef52f18c4233b7e9098345dd72d4 : SUCCESS in 10m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e69ec39d37f2489fb66abf0c19a5f7ad : SUCCESS in 8m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/46eceb2c45c547d1914b1727af59bc69 : SUCCESS in 9m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a04754db7ebf446ebed5aed7dbfafbed : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":2},{"id":"6f8bd2e3257cacae4505a5e6fae4cf0b0dca7360","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-21 06:31:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be94303c88f7455c8db44316b1b83e03 : SUCCESS in 2m 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d6056bf466ebdc65e75c3d2df6d434b491969d3c","revisions":{"de0457f21c885e3ffe8c16e20d5c050c8f772d8b":{"kind":"REWORK","_number":1,"created":"2019-12-12 08:09:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/52/698652/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/52/698652/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/52/698652/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/52/698652/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/52/698652/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/52/698652/1"}}},"commit":{"parents":[{"commit":"4b319012235f941aa0b6a4cac46f5e733473c55b","subject":"Merge \"Fix deleting port from port details page\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/4b319012235f941aa0b6a4cac46f5e733473c55b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-12 08:09:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-12 08:09:48.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I0d31c7406027d7396ec341818eb154f6d1101938\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/de0457f21c885e3ffe8c16e20d5c050c8f772d8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/de0457f21c885e3ffe8c16e20d5c050c8f772d8b"}]},"branch":"refs/heads/stable/stein"},"d6056bf466ebdc65e75c3d2df6d434b491969d3c":{"kind":"REWORK","_number":2,"created":"2019-12-20 08:43:39.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/52/698652/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/52/698652/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/52/698652/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/52/698652/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/52/698652/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/52/698652/2"}}},"commit":{"parents":[{"commit":"1f1ac980ef9ee288a2e8ff27fba2e80cdb12d611","subject":"Cap hacking version to \u003c2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/1f1ac980ef9ee288a2e8ff27fba2e80cdb12d611"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-20 08:43:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-12-20 08:43:36.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I0d31c7406027d7396ec341818eb154f6d1101938\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/d6056bf466ebdc65e75c3d2df6d434b491969d3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/d6056bf466ebdc65e75c3d2df6d434b491969d3c"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
