)]}'
{"id":"openstack%2Fhorizon~700402","triplet_id":"openstack%2Fhorizon~master~Id56be13fb93c6dc123e7cfb1c906357cdf59732b","project":"openstack/horizon","branch":"master","topic":"autogen-plugin-registry","hashtags":[],"change_id":"Id56be13fb93c6dc123e7cfb1c906357cdf59732b","subject":"Generate plugin registry based on releases repo","status":"MERGED","created":"2019-12-23 09:43:50.000000000","updated":"2020-01-14 09:41:30.000000000","submitted":"2020-01-14 09:39:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700402-1578994773477-17819b5d","meta_rev_id":"0a8f1623bbb9e700fd72aebca20092e2ff7aa59b","_number":700402,"virtual_id_number":700402,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"date":"2020-01-09 06:27:53.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-14 09:39:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":2,"date":"2019-12-23 18:03:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2020-01-08 20:37:53.000000000","_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":1,"date":"2020-01-14 08:07:30.000000000","_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-23 18:03:02.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2020-01-14 08:07:30.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-01-14 09:39:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"722cd3061ebabb186dd21c683aa7d09a68870faf","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-12-23 09:43:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d37a8fd0c6671a3c2446ac2f02bb838f81f7e82b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-23 11:52:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/d48e47c0e81c41c8a927b8f2ae91f97f : SUCCESS in 5m 05s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/3f6f4f449d964bb6b3eb3c21c6214212 : SUCCESS in 6m 41s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/5c9776f1bebb4d2a8b62056f5b9efbfd : SUCCESS in 7m 01s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/b639c66147364e678a2eb2ee836bf7ce : SUCCESS in 11m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5136a264256e442c9d1bce40fa4877db : SUCCESS in 9m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62eee1714a8347438bed001400dc10b6 : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4f8a1e3620a4d3893c17921ba0fdd92 : SUCCESS in 9m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/151cdfb08a934653b3fc313875c27f83 : SUCCESS in 7m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba1ac507f0124ec9ad7e11f5ac01ea4c : SUCCESS in 6m 53s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/02a3c75d355d48d091d0a0c8385606dd : SUCCESS in 7m 13s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/3bb2e7f1b71a46b1b765b40dbf4730a2 : TIMED_OUT in 2h 06m 16s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/e912163f87984ef695acb53e47626e12 : SUCCESS in 35m 48s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ac6acfb3051c49e380ac5b3caf4b4e61 : SUCCESS in 4m 47s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/dc4493335e434e99b14ba33dfa897b87 : SUCCESS in 38m 09s","accounts_in_message":[],"_revision_number":1},{"id":"341cfbe159d750539bea33706c812896d91a92ab","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-12-23 13:01:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"565cccb57af261dae25ebe47498bb144e54d3661","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-23 14:32:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/c42c451060784ee4b0017e3605e9badf : SUCCESS in 5m 33s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/4dddba75a78f4d9999233ac1eb3f66ee : SUCCESS in 8m 03s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/ca3f2f246f064f2b965b867fd01dd433 : SUCCESS in 8m 45s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/eae25bbbab3e450686a11e08d8e424c5 : SUCCESS in 11m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3c309f4fb4e4e559ae853db30138a1c : SUCCESS in 8m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56f90c3be0754a63b2d2df7761f3ce85 : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/269e2031987a4b019086b7d68b3e633b : SUCCESS in 10m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/54a73cdb81b349af85fd073c6926edff : SUCCESS in 12m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ea0365324134eef982cecd7c000f17c : SUCCESS in 9m 19s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/fcb1f88c6a2e4a279fe5b6ffb721f90b : SUCCESS in 6m 39s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/27e3c8dd78724b28a2bcda0bce8f592b : SUCCESS in 1h 27m 29s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/7cf05dcd4ef04711bac87d059addde88 : SUCCESS in 43m 42s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/dc8aa8f4ef5c434b82471bac6839124d : SUCCESS in 6m 13s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/13dc2eb9cbf542b99ecc4bd497e9f2e0 : SUCCESS in 40m 56s","accounts_in_message":[],"_revision_number":1},{"id":"8399a4675bbd62595d3cc827a7eeaa1d2dafe6c4","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-12-23 18:03:02.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM.","accounts_in_message":[],"_revision_number":1},{"id":"26d640b0cc9a439ea7fc4abd16fd21007654d30c","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-01-08 20:37:53.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2b97e0e5437ed16888c7a00e3e1b9544be9e92ce","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-09 06:27:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b0b2746388971b55f2d41ba5c5dd1730c1898367","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-01-14 08:07:30.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3adf3cf80cf0e726f8129576818835cda6bce3eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 08:07:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0fe12ef13d0b47ad94be7b66be3e14995544e6f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 09:39:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e32b859cbb1625836baefc8108a7cd3d2593069c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 09:39:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/cce4116183214d78a4cb04ec5b41b522 : SUCCESS in 4m 58s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/ecd38496ec6d4548822d2f2f8658e00f : SUCCESS in 6m 33s\n- horizon-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/4f2ba62c0d654f10ac9570d9520679a9 : SUCCESS in 10m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7de30f40b9ad47549dbbc3c323cc385a : SUCCESS in 11m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4da9c1a20885407d8b505932c2d3281d : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35846a8e3c9c448483119a1561bb5774 : SUCCESS in 10m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3267772dabaa47d8bf21122c21513afa : SUCCESS in 11m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82c424e97b034845bd30470cd4ef05a8 : SUCCESS in 6m 19s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/9660e42ac4bf4a0d839c31e2a0fb5425 : SUCCESS in 37m 27s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/834547e8916149689edf2bc33bd0fdb9 : SUCCESS in 9m 39s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/9d0af63e013f4067bca4814e18e575d9 : SUCCESS in 1h 26m 02s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/a719d475a71e46699116837655c8f6a1 : SUCCESS in 39m 08s","accounts_in_message":[],"_revision_number":1},{"id":"700fcd62fc28a23238c255e51047784b988731f9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 09:41:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b99d641c864f4df1b066209790076237 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cd80207d0952bea9e03b4565b4f8fca8df566283","revisions":{"cd80207d0952bea9e03b4565b4f8fca8df566283":{"kind":"REWORK","_number":1,"created":"2019-12-23 09:43:50.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/02/700402/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/02/700402/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/02/700402/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/02/700402/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/02/700402/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/02/700402/1"}}},"commit":{"parents":[{"commit":"3a829be260a6c1ee8c8a73a0ff59cc68bb8d245a","subject":"Merge \"Add support to get disk_formats from glance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/3a829be260a6c1ee8c8a73a0ff59cc68bb8d245a"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-12-20 09:43:01.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2019-12-23 09:43:24.000000000","tz":540},"subject":"Generate plugin registry based on releases repo","message":"Generate plugin registry based on releases repo\n\nThe plugin registry is maintained manually, so it tends to be out-of-date.\nThe deliverble files in openstack/releases repository provide information\non repositories and bug trackers. It sounds reasonable to me to generate\nthe plugin registry based on infomration in the openstack/releases repo.\n\nThe script added picks up deliverables marked as horizon-plugin.\nmistral-dashboard and networking-bgpvpn dashboard are released in\ndifferent ways, so they are handled specially in the script.\n\nHow to run:\n  ./tools/list-horizon-plugins.py \\\n      --csv-file plugin-registry.csv \\\n      ../releases ussuri\n\nNote that this commit does not autogenerate the plugin registry.\nThe script still needs to be run manually.\nFurther automation would be the next step.\n\nNote that bug trackers of senlin-dashboard and trove-dashboard in\nthe releases repo looks out-of-date. I will submit a fix for them soon.\n\nThe downside of the automation is that we cannot list non-official\nhorizon plugins but I believe it is a good compromise.\nThe only non-offical plugin listed now is the group-based-policy UI,\nbut it is questionable whether it is still maintained well.\n\nChange-Id: Id56be13fb93c6dc123e7cfb1c906357cdf59732b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/cd80207d0952bea9e03b4565b4f8fca8df566283"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/cd80207d0952bea9e03b4565b4f8fca8df566283"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
