)]}'
{"id":"openstack%2Fhorizon~701902","triplet_id":"openstack%2Fhorizon~stable%2Focata~I8fbcb834d9817e7eea1b2c39fc4caab6004b981b","project":"openstack/horizon","branch":"stable/ocata","topic":"tox_failure_fix-stable/ocata","hashtags":[],"change_id":"I8fbcb834d9817e7eea1b2c39fc4caab6004b981b","subject":"Ensure tox registers failures properly","status":"MERGED","created":"2020-01-10 08:36:42.000000000","updated":"2020-02-05 10:35:17.000000000","submitted":"2020-02-05 10:35:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701902-1580898917335-e685ee51","meta_rev_id":"791e99aeb2eaada492136f1af0e852ddd4d5bc6f","_number":701902,"virtual_id_number":701902,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-05 10:35:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"recommended":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":1,"date":"2020-02-03 00:43:31.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2020-01-14 08:12:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":2,"date":"2020-02-05 09:39:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":1,"date":"2020-02-05 09:39:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-14 08:12:39.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-02-03 00:43:31.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2020-02-05 09:39:56.000000000","updated_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"REVIEWER"},{"updated":"2020-02-05 10:35:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cfd02f878d236e27cd319f2a664c2f6f98d6a606","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-10 08:36:42.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"da45e01a2c302166f727af10125216ed77f127e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 08:58:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/868e0208cd8a4ba6900d95f024cd708c : FAILURE in 4m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/67bec3b1ece246bfb141edb1128a93c9 : SUCCESS in 9m 52s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3dfb54fe46c34d4e9249bc6ca669cdc3 : FAILURE in 13m 01s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/e641dd43b0ee4c8193d0f6799833b774 : SUCCESS in 7m 36s\n- nodejs-npm-run-lint https://zuul.opendev.org/t/openstack/build/6de4bf40f0f54cc49cde81aaed8b311e : SUCCESS in 7m 20s\n- nodejs-npm-run-test https://zuul.opendev.org/t/openstack/build/3bc24b850d7c466a9d26fae1eb0bb50f : FAILURE in 7m 00s\n- horizon-openstack-tox-py27dj110 https://zuul.opendev.org/t/openstack/build/5bab2cfa8b5d4d798f7dfca59f339399 : FAILURE in 7m 46s\n- horizon-openstack-tox-py27dj19 https://zuul.opendev.org/t/openstack/build/bdb9a734669542f4afc9d1240267adff : FAILURE in 7m 38s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/e5202594e6794f1aa6e7eec2b24e5185 : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":1},{"id":"1e96818583736f7290bed069e537547c5c9a7a26","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-10 09:37:12.000000000","message":"Patch Set 1:\n\nopenstack-tox-py35 failed as expected. Previously failed tests are ignored.\n\nI confirmed the change works expectedly. I will rebase it on top of th patch of \"operation_log\".","accounts_in_message":[],"_revision_number":1},{"id":"f91fab9b70c10112ee053d8d4288fde3e78395df","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-10 09:38:09.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"33234fea43e6182b8019be3df59605d40891b438","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 09:48:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7d088bbeba94a7383735c1a1da2f330 : SUCCESS in 4m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ed37c7374d704d939980fdc1df78ad71 : SUCCESS in 8m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/078301fca383476fbdc83c4184feb433 : SUCCESS in 9m 04s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/36fdbc53f1c64a33970f7efe92588784 : SUCCESS in 4m 37s\n- nodejs-npm-run-lint https://zuul.opendev.org/t/openstack/build/dd0f6993704548aea055cd8af64c7045 : SUCCESS in 5m 13s\n- nodejs-npm-run-test https://zuul.opendev.org/t/openstack/build/b08d80751ab8493ca0620df158361254 : SUCCESS in 6m 52s\n- horizon-openstack-tox-py27dj110 https://zuul.opendev.org/t/openstack/build/b213f23a0ada4f69ab03aa18d8d543d8 : SUCCESS in 8m 19s\n- horizon-openstack-tox-py27dj19 https://zuul.opendev.org/t/openstack/build/13e4d7c2ddd94fe6ba281d507a5272e5 : SUCCESS in 8m 42s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/6c73120099fb44d199dae8fbd3254fcf : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":2},{"id":"15cf1d53c273c41852019356f3d9dfcde39388eb","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-01-14 08:12:39.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"923e17b7aef47c5409236cccf4d1e200b439de74","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-02-03 00:43:31.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e6ee1908e82c6b8152e7fd45c36d80f1e26ae023","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2020-02-05 09:39:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"42d7e56d4f5159f6152721e3045a21a98150afa8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 09:41:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"264f863d16b4ebe2b9a2d40c28aa5a1fc099c303","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 10:35:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"994981435a34c98bbd32996d0925924351afc078","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 10:35:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c522314af5b14a2a9bd2a633fa1d233f : SUCCESS in 4m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/25f9c33906814fa2a589de4b95b4266a : SUCCESS in 9m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ad96f59366d44e0994f1ea81ea7ded8d : SUCCESS in 12m 14s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ec0adb19ff9e40d699ec083dfee0865e : SUCCESS in 5m 39s\n- nodejs-npm-run-lint https://zuul.opendev.org/t/openstack/build/995be669680848c99c1055e8131c1001 : SUCCESS in 6m 03s\n- nodejs-npm-run-test https://zuul.opendev.org/t/openstack/build/0cc8deb11cec45148d2aeca35164d578 : SUCCESS in 10m 11s\n- horizon-openstack-tox-py27dj110 https://zuul.opendev.org/t/openstack/build/3a83e9608a464b0ea85337432198bb79 : SUCCESS in 10m 55s\n- horizon-openstack-tox-py27dj19 https://zuul.opendev.org/t/openstack/build/23c15a0ecc2c4572832ada6311db8afb : SUCCESS in 11m 06s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/a582baf0de9d475e9f5e20ed47e40f8e : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"420d34414aeafe835cde2960c8c6d00d1b1cef94","revisions":{"1e83dacd3095f43ef34529aa3144f45fb5db261b":{"kind":"REWORK","_number":1,"created":"2020-01-10 08:36:42.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/02/701902/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/02/701902/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/02/701902/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/02/701902/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/02/701902/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/02/701902/1"}}},"commit":{"parents":[{"commit":"bef5eaef67a1feaef28c53636031f1c818332cbe","subject":"Merge \"Fix Angular errors in openstack_auth\" into stable/ocata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/bef5eaef67a1feaef28c53636031f1c818332cbe"}]}],"author":{"name":"adrian-turjak","email":"adriant@catalyst.net.nz","date":"2017-07-19 13:42:20.000000000","tz":720},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-01-10 08:36:42.000000000","tz":0},"subject":"Ensure tox registers failures properly","message":"Ensure tox registers failures properly\n\nTox runs tests via unit_tests.sh which when not\nrunning a test subset will mean tox only checks for\nthe final exit code. This means horizon test failues\nwill not be seen by tox unless openstack_dashboard\ntests also fail.\n\nThis change forces an exit code which !\u003d0 if either\ntest run fails.\n\nChange-Id: I8fbcb834d9817e7eea1b2c39fc4caab6004b981b\n(cherry picked from commit 280848c55ffeb8cfe948acd5f2ca234d2c1264f5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/1e83dacd3095f43ef34529aa3144f45fb5db261b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/1e83dacd3095f43ef34529aa3144f45fb5db261b"}]},"branch":"refs/heads/stable/ocata"},"420d34414aeafe835cde2960c8c6d00d1b1cef94":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-01-10 09:38:09.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/02/701902/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/02/701902/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/02/701902/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/02/701902/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/02/701902/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/02/701902/2"}}},"commit":{"parents":[{"commit":"ceb93c2fdd5192dd6480285b8d8171c640b7d893","subject":"operation_log: Log mask_fields only when exists in request","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/ceb93c2fdd5192dd6480285b8d8171c640b7d893"}]}],"author":{"name":"adrian-turjak","email":"adriant@catalyst.net.nz","date":"2017-07-19 13:42:20.000000000","tz":720},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-01-10 09:38:09.000000000","tz":0},"subject":"Ensure tox registers failures properly","message":"Ensure tox registers failures properly\n\nTox runs tests via unit_tests.sh which when not\nrunning a test subset will mean tox only checks for\nthe final exit code. This means horizon test failues\nwill not be seen by tox unless openstack_dashboard\ntests also fail.\n\nThis change forces an exit code which !\u003d0 if either\ntest run fails.\n\nChange-Id: I8fbcb834d9817e7eea1b2c39fc4caab6004b981b\n(cherry picked from commit 280848c55ffeb8cfe948acd5f2ca234d2c1264f5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/420d34414aeafe835cde2960c8c6d00d1b1cef94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/420d34414aeafe835cde2960c8c6d00d1b1cef94"}]},"branch":"refs/heads/stable/ocata"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
