)]}'
{"id":"openstack%2Fhorizon~702539","triplet_id":"openstack%2Fhorizon~master~I17add5dc20846be052577db1ed8ca8bf98819846","project":"openstack/horizon","branch":"master","topic":"remove-pint","hashtags":[],"change_id":"I17add5dc20846be052577db1ed8ca8bf98819846","subject":"Remove Pint from requirements.txt","status":"MERGED","created":"2020-01-14 21:34:13.000000000","updated":"2020-01-22 17:44:46.000000000","submitted":"2020-01-22 17:42:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702539-1579714924606-5aeb02ef","meta_rev_id":"ea9999dca2625104a93917828126aa12bca5e011","_number":702539,"virtual_id_number":702539,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-22 17:42:04.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"recommended":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-01-15 13:34:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-01-15 07:30:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":1,"date":"2020-01-15 07:37:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2020-01-22 16:03:12.000000000","_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-01-22 16:03:12.000000000","_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-15 07:30:08.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2020-01-15 07:37:21.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-01-15 13:34:40.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-01-22 16:03:12.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-01-22 17:42:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1287051183ba98a015a0d2dde8a22dfe49ed9c2b","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-14 21:34:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1ab3a7c8365541d3b797a435c56dfa702de53b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 01:01:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/cfcfec9634fc4e109526583872ce5850 : SUCCESS in 4m 45s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/a8f74e0bd1bb4d169e9b339119bcfce6 : SUCCESS in 7m 08s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/1c6ad1e835c34603a3a386777698dd8f : SUCCESS in 11m 23s\n- horizon-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/848f806967fc417d9a87c72b22aa5da0 : SUCCESS in 10m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ec38855fbdc4192a19da5400ddff4d0 : SUCCESS in 12m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d22023b8dd264a8190bc522d1ca9da8a : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d96557162e7d4295a44697581c41731e : SUCCESS in 11m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/51204ae1b0a242adacdc4d054a5d2ba3 : SUCCESS in 12m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47a91a9571e14541a0bfd46d376098e1 : SUCCESS in 9m 05s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/5788f63c30b148e0b3d0153bbe9ef653 : SUCCESS in 9m 36s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/a9fb631db7534688b93bdfc32f033ddd : SUCCESS in 1h 26m 07s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/70eb41d8bae14c7284f7396c25b5c0ce : SUCCESS in 56m 02s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/dba1ee13ca3549e1883521a98cac8710 : SUCCESS in 7m 11s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/e3c9c198dfa04884b76bbcb4d4910920 : SUCCESS in 54m 11s","accounts_in_message":[],"_revision_number":1},{"id":"dd12bf4bbb796f7da2f9a856211db1e68015aeb6","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-01-15 02:26:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ff09c1bf21ab1f0a6ac6e0935b4054c47c049f68","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-15 04:13:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"25153d4bffcd8676227471994fd20f1707143790","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 05:34:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c63f4f78d4dd44e2ba3d6db247276010 : SUCCESS in 2m 12s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/0239e39dca5e47ad8b7cb035b79d0ef3 : SUCCESS in 4m 15s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/bd958a01018046ee8fffd644be5aac73 : SUCCESS in 6m 52s\n- horizon-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/38a9f8383bc941289f7d9233a07fc7be : SUCCESS in 9m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0f3b3f5800349699f286b215a89bf35 : SUCCESS in 8m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/417e2f6b00df4fe691851a752f922b18 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ef9aa34669b426e8dc3c96477e2a0a6 : SUCCESS in 7m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6bceea1a6e7a4040b70ee9221a72e521 : SUCCESS in 8m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d7529acba5149038e5efcb5abc6827a : SUCCESS in 6m 26s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/9bb2f30d32904a78b886b51815d3a4ee : SUCCESS in 6m 06s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/0c216baeccb94660aa51011f746b456e : SUCCESS in 1h 20m 14s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/bc30409b11484287a5e5ffd56e0aaff1 : SUCCESS in 32m 02s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a86cda68496648519533b9c01d3a0bbd : SUCCESS in 4m 34s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/d3ec398585e4400388bb7f3a90874fb7 : SUCCESS in 39m 31s","accounts_in_message":[],"_revision_number":2},{"id":"0c8073ea73822a82f09b7f51fb89f28d0a2dc6cb","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-01-15 07:30:08.000000000","message":"Patch Set 2: Code-Review+2\n\nlGTM","accounts_in_message":[],"_revision_number":2},{"id":"367095f98c7a150318ffb1eaf202e60fb3e90830","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-15 07:37:21.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"670d911a2cbb0612ae7bb610fbd8a04dd0df98ee","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-15 13:34:40.000000000","message":"Patch Set 2: Code-Review+1\n\nLooks happy without it. Great to clean up things that aren\u0027t actually needed.","accounts_in_message":[],"_revision_number":2},{"id":"7d0f35b389672aa03fc1c38a60e0541145ed4436","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-01-22 16:03:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b7eceba07f917dc138cc9fd79f212f3bcf60cd9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 16:03:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cdc9c0be0d54ca505462fc5309f2ab696e22c360","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 17:42:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/9ae2c3c4ba124821a7530fa9e3d26819 : SUCCESS in 3m 32s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/3447a6efb2c14c169625b6f983227865 : SUCCESS in 5m 33s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/0c75cf18170c4a76bac709695cceb997 : SUCCESS in 6m 51s\n- horizon-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/ae96401d3602459f9a77038d000a693e : SUCCESS in 11m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/987e8c3970674c1991f9b8c18a6f5866 : SUCCESS in 10m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d424764a22f4f698836f8e335e01df8 : SUCCESS in 7m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9ee9f4508fb4cf385cfaacde4eeb2d8 : SUCCESS in 9m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8aebf08bd254401babf81da9fcb50fa1 : SUCCESS in 11m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2093f99865e649dc944bdc1d7b4987b4 : SUCCESS in 7m 43s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/f3522ab69bf14b6daed808fabbf77697 : SUCCESS in 32m 04s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/ef0571882f3d4ac2b2a3549513ce1226 : SUCCESS in 6m 56s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/ea4f97f3d9384c0f8b7a0c360ac8ff48 : SUCCESS in 1h 30m 30s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/bd0d509180834fd9868edae3d0c81bbd : SUCCESS in 37m 47s","accounts_in_message":[],"_revision_number":2},{"id":"aec72d8c7064f291e3186c29c5a22238c9856a6d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 17:42:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"4daede0a9623da14279a2680847e65e1c0259da9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 17:44:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb60e5a24cf84fcba805b1017280c51a : SUCCESS in 1m 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1570e5c3edc624b9a566e7115149f574c0973167","revisions":{"23ccea4d5bd2338932c2c65de458ef446349acbc":{"kind":"REWORK","_number":1,"created":"2020-01-14 21:34:13.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/39/702539/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/39/702539/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/39/702539/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/39/702539/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/39/702539/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/39/702539/1"}}},"commit":{"parents":[{"commit":"91ed3a0695211532f18110555cfa0ef66c77c6cc","subject":"Merge \"Fix dupliated words issue like \" by by which\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/91ed3a0695211532f18110555cfa0ef66c77c6cc"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-01-14 21:32:29.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-01-14 21:32:29.000000000","tz":540},"subject":"Remove Pint from requirements.txt","message":"Remove Pint from requirements.txt\n\nPint is no longer used after horizon/utils/units.py was cleanup\nby commit 26eebd4abbbabc1475bae61c51d1cfcf14dba770.\n\nChange-Id: I17add5dc20846be052577db1ed8ca8bf98819846\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/23ccea4d5bd2338932c2c65de458ef446349acbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/23ccea4d5bd2338932c2c65de458ef446349acbc"}]},"branch":"refs/heads/master"},"1570e5c3edc624b9a566e7115149f574c0973167":{"kind":"REWORK","_number":2,"created":"2020-01-15 04:13:16.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/39/702539/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/39/702539/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/39/702539/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/39/702539/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/39/702539/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/39/702539/2"}}},"commit":{"parents":[{"commit":"91ed3a0695211532f18110555cfa0ef66c77c6cc","subject":"Merge \"Fix dupliated words issue like \" by by which\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/91ed3a0695211532f18110555cfa0ef66c77c6cc"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-01-14 21:32:29.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-01-15 04:12:50.000000000","tz":540},"subject":"Remove Pint from requirements.txt","message":"Remove Pint from requirements.txt\n\nPint is no longer used after horizon/utils/units.py was cleanup\nby commit 26eebd4abbbabc1475bae61c51d1cfcf14dba770.\n\nChange-Id: I17add5dc20846be052577db1ed8ca8bf98819846\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/1570e5c3edc624b9a566e7115149f574c0973167"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/1570e5c3edc624b9a566e7115149f574c0973167"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
