)]}'
{"id":"openstack%2Fhorizon~702813","triplet_id":"openstack%2Fhorizon~stable%2Fstein~I47f73ff94664d315a2400feb8ce8a25f4e6beced","project":"openstack/horizon","branch":"stable/stein","topic":"bug/1838522-stable/stein","hashtags":[],"change_id":"I47f73ff94664d315a2400feb8ce8a25f4e6beced","subject":"Fixes a series of bugs related to Floating IPs.","status":"MERGED","created":"2020-01-16 08:00:51.000000000","updated":"2020-02-18 17:05:43.000000000","submitted":"2020-02-18 17:02:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702813-1582045377684-86a07e3f","meta_rev_id":"dd3b8d27bdfa4559f0d7c405da5718e7ca572aca","_number":702813,"virtual_id_number":702813,"owner":{"_account_id":26546,"name":"Marek Lyčka","email":"marek.lycka@ultimum.io","username":"mareklycka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-18 17:02:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"recommended":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":1,"date":"2020-01-27 04:08:57.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2020-02-18 16:22:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":2,"date":"2020-01-24 07:50:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":1,"date":"2020-02-18 16:22:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-24 07:50:07.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-01-27 04:08:57.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2020-02-18 16:22:55.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-02-18 17:02:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b7acf556dafc9162906101ccd6129ada5aef2b9f","author":{"_account_id":26546,"name":"Marek Lyčka","email":"marek.lycka@ultimum.io","username":"mareklycka"},"date":"2020-01-16 08:00:51.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"272cdad498fb84d81573ae772d25144827fb305f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 09:00:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs4-npm-run-lint https://zuul.opendev.org/t/openstack/build/e0167721ce154fb2963c493d3053cf4b : SUCCESS in 5m 28s\n- nodejs4-npm-run-test https://zuul.opendev.org/t/openstack/build/70d76abb6a3e4388a93582950c028ba8 : SUCCESS in 9m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd7b7904dd784e1884912fed44488753 : SUCCESS in 10m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcb5902f107c4fc4831f2d9a0a065054 : SUCCESS in 6m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8628d57da86b4fe8b8a7f4297018e865 : SUCCESS in 10m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/397004ee045845268a7efe2d143ff3d6 : SUCCESS in 9m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d689fd8d5c024900971b7e8869dcaa67 : SUCCESS in 10m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5905e58a589244bf965507982daa587c : SUCCESS in 9m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7ce40f67abf42249e086e6606e77c14 : SUCCESS in 4m 50s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/207fd96a0a674c85bea05c97df28e26c : SUCCESS in 9m 38s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/3f388fe2a9ae44b194b0a7f34482c7c7 : SUCCESS in 6m 52s (non-voting)\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/ec9cf4d7e9d04ad3af2091260a29cbe9 : SUCCESS in 58m 24s (non-voting)\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/dfbaec263b6e43e5a7cf31e79da72f46 : SUCCESS in 36m 20s\n- horizon-dsvm-tempest-plugin-py27 https://zuul.opendev.org/t/openstack/build/f1956dc5352b46b4921c060924c45bd0 : SUCCESS in 36m 21s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/974214bf02fb481586760b6123ac122e : SUCCESS in 4m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9b15777e31c2af77343f9c9603dabf8b279ca0a4","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-24 07:50:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c61019684b33e0877c4c86006a50c5302371f1db","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-01-27 04:08:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1155afd76cc6769c4e1b57064c53dee73a9407c3","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-02-18 16:22:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2186966ad27495c849e15bdd40666680809fa024","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 16:23:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"353aa1294f696178577db10d9e6522af1a4f3d78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 17:02:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- nodejs4-npm-run-lint https://zuul.opendev.org/t/openstack/build/60935a2d2639473aae3def878aa7fdf9 : SUCCESS in 10m 30s\n- nodejs4-npm-run-test https://zuul.opendev.org/t/openstack/build/e58dc00800f14a1abcac0faecd7b9ad4 : SUCCESS in 11m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2cb8f72ec6854cc3b8fe74ba8fe054c3 : SUCCESS in 14m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4887363a576f48f298c6bc19bd867e69 : SUCCESS in 9m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a246469d593548778402a5c67fb2bd87 : SUCCESS in 13m 07s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/578ddea059204cba9bc14e62dd03fe15 : SUCCESS in 12m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18f2e4c0a3aa479b960757e373f58414 : SUCCESS in 11m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/073b17c36e534c8aabecab3f75d64ec1 : SUCCESS in 12m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0f9de8aecc741e99f669c179231d1e5 : SUCCESS in 7m 48s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/cdbb6165364c4ea3ad4c711269472560 : SUCCESS in 13m 04s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/6b2ec4fce3744738ad905e1089c79f04 : SUCCESS in 36m 48s\n- horizon-dsvm-tempest-plugin-py27 https://zuul.opendev.org/t/openstack/build/c7a27fd28f99402fbc23de5e46d0cad4 : SUCCESS in 35m 12s","accounts_in_message":[],"_revision_number":1},{"id":"5e73832a86658997e2157202adbf54f777e59821","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 17:02:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8d48a1da0347bba65d51ca057ea8a3e58a441ae2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 17:05:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64d0ebf1167e40548473f86b0c3cbd29 : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9fdb6291377e3b5bcdb3a89d998c6f5a1ab0043c","revisions":{"9fdb6291377e3b5bcdb3a89d998c6f5a1ab0043c":{"kind":"REWORK","_number":1,"created":"2020-01-16 08:00:51.000000000","uploader":{"_account_id":26546,"name":"Marek Lyčka","email":"marek.lycka@ultimum.io","username":"mareklycka"},"ref":"refs/changes/13/702813/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/13/702813/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/702813/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/702813/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/702813/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/13/702813/1"}}},"commit":{"parents":[{"commit":"29b3f06ad4ff808c59066cdc9bbd4cf7840b9bb4","subject":"Merge \"GUI hangs on creation of project in User creation form\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/29b3f06ad4ff808c59066cdc9bbd4cf7840b9bb4"}]}],"author":{"name":"Marek","email":"marek.lycka@ultimum.io","date":"2019-10-30 08:38:29.000000000","tz":60},"committer":{"name":"Marek Lyčka","email":"marek.lycka@ultimum.io","date":"2020-01-16 08:00:51.000000000","tz":0},"subject":"Fixes a series of bugs related to Floating IPs.","message":"Fixes a series of bugs related to Floating IPs.\n\n- Fixes KeyErrors when accessing \u0027floatingip\u0027 values in usages, which\n  broken Floating IP allocation.\n- The quota display in the bottom right of the Allocation dialog are\n  only displayed if \u0027enabled_quotas\u0027 is True\n- Adds security group rule tallying for the usages overview page, which\n  fixes a KeyError crash for installations where Horizon \u0027enable_quotas\u0027\n  is set to true, but the \u0027quota_details\u0027 extension is not installed on\n  in Neutron\n- Adds a policy check to show and hide The plus/add button in\n  Instances-\u003eAssociate Floating IP to match the Allocate IP To Project\n  button in Floating IPs\n- Fixed the page title not being set for the non-modal version of the\n  modal allocation dialog/form\n- Added an \u0027allowed\u0027 functionality for network usage overview charts to\n  allow for them to be dynamically disabled\n- Added tests and mocks for the above\n- Added tests for non-legacy quota tallying for networks\n- Added test for disabled network quotas in overview\n\nChange-Id: I47f73ff94664d315a2400feb8ce8a25f4e6beced\ncloses-bug: #1838522\n(cherry picked from commit 161b4ae5d4a7a29f9e5c91d9b391dee49e075993)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/9fdb6291377e3b5bcdb3a89d998c6f5a1ab0043c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/9fdb6291377e3b5bcdb3a89d998c6f5a1ab0043c"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
