)]}'
{"id":"openstack%2Fhorizon~710101","triplet_id":"openstack%2Fhorizon~master~I73a741a9b7d80eb0475a07cca13138659a9a31b0","project":"openstack/horizon","branch":"master","topic":"remove-mox","hashtags":[],"change_id":"I73a741a9b7d80eb0475a07cca13138659a9a31b0","subject":"trivial: Use literals inline","status":"MERGED","created":"2020-02-26 17:52:52.000000000","updated":"2020-03-03 02:21:21.000000000","submitted":"2020-03-03 02:19:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710101-1583201981663-b95df229","meta_rev_id":"e4ad7db745b85870e34abe999fdecde1e1c07763","_number":710101,"virtual_id_number":710101,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-03 02:19:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":2,"date":"2020-03-03 00:45:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2020-03-02 23:01:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":1,"date":"2020-03-03 00:45:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-02 23:01:13.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-03-03 00:45:57.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2020-03-03 02:19:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"47db6bfc095e03583bedde76229dab75f2d6e882","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-02-26 17:52:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"278cf914ffb37fbb1d68365a5dc1f08ec2dae2fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 19:48:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/640efef312f2454f80545b12254dde41 : SUCCESS in 7m 07s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/42b52ff9e80f4fb580e368199a7712c6 : SUCCESS in 11m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e624b200dbb24afaafe481c2ccb5e2ae : SUCCESS in 12m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0fb2a78216b040e1bee5555de6f3dddb : SUCCESS in 13m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b914a7f9256449889f53477aeafff68 : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72b710d689f64490ba2bdbd19cd11394 : SUCCESS in 13m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c5866e788474f5a867d73da660f8eb8 : SUCCESS in 11m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a56d7c4b4ef8427e93f7c81bf99a7767 : SUCCESS in 8m 21s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/9399560fc8dd4722ab61b432667f6911 : SUCCESS in 8m 59s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/254a2e73f26743c399a9380d12794538 : SUCCESS in 1h 40m 19s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c43b460aa43c4ab391f8d524dbb78fec : SUCCESS in 7m 07s","accounts_in_message":[],"_revision_number":1},{"id":"d499b85840d41d60c62c805c5885ad82912b8f51","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-03-02 23:01:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a96c514c4d61184c517fbd98b091b168eed6dfd1","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-03-03 00:45:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1a73ef2204e2809b71a0f62e32657e89814b8d12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 00:46:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e3d89a4882163a0db4656b2e827d1636f95a17ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 02:19:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/e2777ab2b1e74ce98c9a03a17a3e7908 : SUCCESS in 6m 40s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/f7c712e2435b432ca4425f6800c63b74 : SUCCESS in 8m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cb6feb109fb74faa893f75127ff53029 : SUCCESS in 13m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d7cd71ef76ed42fcb769dca77385a83a : SUCCESS in 13m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b115b7ca03b7493e908d628060eceedc : SUCCESS in 8m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6dba04eada684f67a3a26f3fafc9b9e3 : SUCCESS in 11m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4751437739ff45bd9caf63d9d4f2a7a2 : SUCCESS in 9m 26s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/2442bbcaa8b74f9790a6b4dfb6aabe0d : SUCCESS in 9m 18s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/ab2613844bb24d839715deea6e0deca8 : SUCCESS in 1h 28m 04s","accounts_in_message":[],"_revision_number":1},{"id":"fb57378f940f1cfdbeb4fb4619dca237b9e08408","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 02:19:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fe2c81347bd87fe0f1dc40ac4798a68352c79e59","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 02:21:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31a76934fcfa4b1aaf5cec031c02927e : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2c30f892751a580c32f15992ea5fb414b40a543d","revisions":{"2c30f892751a580c32f15992ea5fb414b40a543d":{"kind":"REWORK","_number":1,"created":"2020-02-26 17:52:52.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/01/710101/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/01/710101/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/01/710101/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/01/710101/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/01/710101/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/01/710101/1"}}},"commit":{"parents":[{"commit":"e23fe2bc42f7ccf55ce52af5aeb103ab9bc9ac98","subject":"tests: Convert remaining mox tests to mock (pt. 2.8)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/e23fe2bc42f7ccf55ce52af5aeb103ab9bc9ac98"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-02-26 16:29:36.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-02-26 17:49:20.000000000","tz":0},"subject":"trivial: Use literals inline","message":"trivial: Use literals inline\n\nAnother step towards the demise of Mox. This makes the move from\nmox-style calls to mock ones much easier to grok. We also some \u0027if\u0027\nblocks that were not being used.\n\nChange-Id: I73a741a9b7d80eb0475a07cca13138659a9a31b0\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/2c30f892751a580c32f15992ea5fb414b40a543d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/2c30f892751a580c32f15992ea5fb414b40a543d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
