)]}'
{"id":"openstack%2Fhorizon~710174","triplet_id":"openstack%2Fhorizon~stable%2Ftrain~I776d2c8c6864067e7b44ce4b2f107c80b8b6d7fb","project":"openstack/horizon","branch":"stable/train","topic":"bug/1744670","hashtags":[],"change_id":"I776d2c8c6864067e7b44ce4b2f107c80b8b6d7fb","subject":"Add custom SSL CA Cert support for api.cinder.get_microversion","status":"MERGED","created":"2020-02-27 02:55:18.000000000","updated":"2020-10-15 12:08:37.000000000","submitted":"2020-10-15 12:06:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710174-1602763609361-dfe5b5ea","meta_rev_id":"f5121d13ffed0f5aaa78daf0bd9d9e945ff52be3","_number":710174,"virtual_id_number":710174,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-15 12:06:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"recommended":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":1,"date":"2020-10-05 11:04:55.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2020-10-02 20:35:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":2,"date":"2020-10-15 11:01:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":1,"date":"2020-10-15 11:01:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-19 08:36:58.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-06-27 09:03:25.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-10-02 20:35:31.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-10-05 11:04:55.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2020-10-15 11:01:32.000000000","updated_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"REVIEWER"},{"updated":"2020-10-15 12:06:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"51189eeb39a7a7ada9b6011abf82eb06e5eb5c1e","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-02-27 02:55:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"123944195501bc446fa3552615bec4dbf8eb2d4f","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-02-27 02:56:25.000000000","message":"Patch Set 1: Workflow-1\n\nI first would like to get feedback on the approach before adding the unit test coverage, so -W.","accounts_in_message":[],"_revision_number":1},{"id":"e67290a3c44a22a4a62b86c185a6f58524cb3528","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-02-27 04:28:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"87e8cad26fc6713361fc2b069fcddac52d96fecb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 04:38:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/33d2232210944760bd2ccf986ad3de88 : SUCCESS in 8m 17s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/e37e8a44ce1b43b299a0f526ca8eaab5 : SUCCESS in 11m 19s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/f82e4d4a5e2c4e90ba01c0e890b09237 : SUCCESS in 12m 29s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/52263a63a0cd456c8725005083a749de : SUCCESS in 12m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bbdbab89112846c6bb50126ea211e8b4 : SUCCESS in 12m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7408b7a656da49f59cc4b86ae8072361 : SUCCESS in 10m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eada763dba674d4c9ed5afdbf4595f38 : SUCCESS in 13m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7750c867246e46d2824dfef006ddc21f : SUCCESS in 11m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e23713bc2973450a98cb1af94deb1f60 : SUCCESS in 11m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/451148bfbf1d4df3af729748afbb9c6c : SUCCESS in 9m 37s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/41b0d249426a4f5ca7db8015ca117bcd : SUCCESS in 10m 22s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/e0f8710c4da340aeadd09af4a78eab02 : SUCCESS in 1h 02m 41s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/b07c438d32b34d8f8b51533e5d5bb596 : SUCCESS in 36m 01s\n- horizon-dsvm-tempest-plugin-py27 https://zuul.opendev.org/t/openstack/build/41bc0f0974ea4a36851cdd20b6fde651 : FAILURE in 23m 19s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6abff478dbc04399ac776b71a2dc586f : SUCCESS in 6m 10s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/c345aebd13bb438c83b238319eed4bde : FAILURE in 22m 27s","accounts_in_message":[],"_revision_number":1},{"id":"eb3601f3d5ea52e25f3584efebe681210a02d28b","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-02-28 06:26:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0dd07d77534cfe734ff57bc68ae8f793e9a8950b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 07:34:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/34cafa7016454c2f9c069fd186cf019f : SUCCESS in 10m 38s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/b885c053eaa84c2fba2b3b1acb5bd161 : SUCCESS in 8m 34s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/dca38cc146a2461ea4e07c10f276b763 : SUCCESS in 12m 13s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/f0a5b7e7be074932b874ef6fea576d5c : SUCCESS in 11m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a04ffcd1fced4dc1924b12830552c045 : SUCCESS in 11m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a864a15a9e448fe86424bcda5f108ee : SUCCESS in 8m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2a54db5c2698424e9d7167cdbbfc0932 : SUCCESS in 13m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5138b1bc42e648bdaa4dc071ac46cb1c : SUCCESS in 10m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ee41bb242c3e46b8a12b0d136fd9edf0 : SUCCESS in 12m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfb124ff532748f9a3486e93b1eea18c : SUCCESS in 7m 55s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/b79389beddc34e9d932e5c83024ce62c : SUCCESS in 10m 27s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/d12b8a8cd0524e35b8cda077465320ec : SUCCESS in 1h 04m 42s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/9d2f5e9daca64f0aa4aed9845832378d : SUCCESS in 39m 27s\n- horizon-dsvm-tempest-plugin-py27 https://zuul.opendev.org/t/openstack/build/91f40ba4d5f5474a8949673479ec00a9 : SUCCESS in 42m 28s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/20b27cba4c354ba2aeccd4a6a2eb8877 : SUCCESS in 6m 29s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/0b53625b8a444b2d8ba47bf684e94e66 : SUCCESS in 51m 50s","accounts_in_message":[],"_revision_number":1},{"id":"ab270eedf048f5fccd1a1fc5b96932230e4a683b","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-04-02 14:56:18.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM as a workaround for old releases.","accounts_in_message":[],"_revision_number":1},{"id":"2c6690b20d2985ab9c281188efe9473059c846f3","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-04-22 06:22:49.000000000","message":"Patch Set 1:\n\nThanks Ivan. I forgot to check your reply.\nI will work on adding unit tests.","accounts_in_message":[],"_revision_number":1},{"id":"b89739df623db04243ebb7ea8dbcd1d8c20c3513","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-19 08:36:58.000000000","message":"Patch Set 1: Code-Review+1\n\nTried this with horizon 16.2.0 and python-cinderclient 5.0.1, it worked. Would love to see it in a Train stable release!","accounts_in_message":[],"_revision_number":1},{"id":"f78190fffe4fbebefd85d87ee67f06ba74625056","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-27 09:03:25.000000000","message":"Patch Set 1: Code-Review+1\n\n+1 for any fixing approach really; centos packages 5.0.1, upper constraints has 5.0.1 so it should be fine either way","accounts_in_message":[],"_revision_number":1},{"id":"ab7cf7a514735f7b1a3ad581d93157f637ddacce","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-09-28 06:56:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"31a0f5d9e29ab33c1fc3ef24a02533442d261112","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-09-28 08:06:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"281964c85eefd655ddec7f0c5454acbed65544fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 09:06:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/b611e402949c47edbfe2800dd8cdb2dc : SUCCESS in 6m 29s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/eea8156befa247b5ad16cb9bee17288d : SUCCESS in 6m 31s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/4e6ba7d136be455ea0f46652f67a5c2e : SUCCESS in 7m 56s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/db16ad3cca58494f88131db411ae8758 : SUCCESS in 9m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8792f5333d746849858d98ec8a20d5a : SUCCESS in 9m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e8bfe5865b0442fbc29d53281745f54 : SUCCESS in 5m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e3f1f056c0224bf1b257f206f2447da4 : SUCCESS in 9m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aaf7e719df594ee9b930ce77bb6b74e4 : SUCCESS in 7m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/16f202bf5de94f28b532ca322c563481 : SUCCESS in 8m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcb8e0a10b86458998c16044f83e4663 : SUCCESS in 6m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac812fb17a214cc4bb86233bf823b047 : SUCCESS in 9m 49s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/e1ac449377f3424d9c659131055f2063 : SUCCESS in 6m 36s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/32385d281b154f379861e81d2e145a3c : SUCCESS in 56m 49s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/558f9639bebd4f5b8affa934cadb8f9f : SUCCESS in 28m 42s\n- horizon-dsvm-tempest-plugin-py27 https://zuul.opendev.org/t/openstack/build/550aeea218624a2ba6c3537b4cf7675a : SUCCESS in 28m 11s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a570ba553e7945579d1bc8ca795bf5c5 : SUCCESS in 4m 39s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/ba1b064f289840109023e7afa0eb1bf9 : SUCCESS in 29m 01s","accounts_in_message":[],"_revision_number":3},{"id":"df46cb12de2a1012fbd97586ec4253bd5c36954b","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-10-02 20:35:31.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"7bb9d8c506ba5cae240412d650758aaa154a15d9","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-10-05 11:04:55.000000000","message":"Patch Set 3: Code-Review+1\n\nLGTM code is borrowed from cinderclient https://opendev.org/openstack/python-cinderclient/src/branch/master/cinderclient/client.py#L74","accounts_in_message":[],"_revision_number":3},{"id":"8142bffb8c67be5f66a1dab26f99ea33b0663b24","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2020-10-15 11:01:32.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e8a6d0c01c338716a5bd6811e749aeb0051f6a10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 11:01:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e9e8bfbe095d04946ee86b5b08b97bdd5be86ce0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 12:06:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/497826715fba44d68e17bd369399fa9e : SUCCESS in 5m 18s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/6ce077cf9e184e578db91dd991adeba3 : SUCCESS in 7m 04s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/72ddc059e0064ac2b853eba6f8a4f483 : SUCCESS in 10m 46s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/f9ecdf991bbb4ad594eb2398c7c0ff65 : SUCCESS in 10m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/789e9d9401c343458f84976608eff280 : SUCCESS in 10m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8dac4a1f142644568bcca9854d73e1b3 : SUCCESS in 7m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/18bda57683d34559a9b090edeb224135 : SUCCESS in 11m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb5ec71ad02342fc9ddcccff5e333ee7 : SUCCESS in 6m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/db8611a7787f4211be225cc92daf9ba3 : SUCCESS in 12m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/175a7c251c6c44debde5844f8cf63ef8 : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/12dad65422aa44c5aae3a94574319434 : SUCCESS in 11m 08s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/9a1cbc39b7214ca89ae567801eaec319 : SUCCESS in 38m 02s\n- horizon-dsvm-tempest-plugin-py27 https://zuul.opendev.org/t/openstack/build/5ef4aea63b884ee68593a1385abe3b28 : SUCCESS in 36m 32s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/c5d8803d6609440eb5ca31dcea5946b3 : SUCCESS in 7m 06s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/261424b7e4944051b736099e3e2aeab0 : SUCCESS in 1h 01m 14s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/bdf248263b8e4859b069944957dff849 : SUCCESS in 38m 16s","accounts_in_message":[],"_revision_number":3},{"id":"121228c3872f8c0b9a6f48cf17d9db8e12c4e366","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 12:06:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"6052591b8e4451c71fe4daa2d1164354a1260604","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 12:08:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37359623bae54472a78a4df07ce3e809 : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6ddd6f290c8416eb02d7e2c815570af : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"68d4c5e78db61ea401ee89c84a69676e26a636a1","revisions":{"5d9b7da4799241707f0f221bfe1d12a782d12188":{"kind":"REWORK","_number":1,"created":"2020-02-27 02:55:18.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/74/710174/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/74/710174/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/74/710174/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/74/710174/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/74/710174/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/74/710174/1"}}},"commit":{"parents":[{"commit":"e058e381607e93439a59f47c36e6a9b7cf61cc71","subject":"Allow to evacuate without specifying a target host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/e058e381607e93439a59f47c36e6a9b7cf61cc71"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-02-27 02:29:23.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-02-27 02:29:23.000000000","tz":540},"subject":"Add custom SSL CA Cert support for api.cinder.get_microversion","message":"Add custom SSL CA Cert support for api.cinder.get_microversion\n\nThis is a stable branch version of\ncommit 285c51f7e1b27a786fa9684abdc489998285b4e2 in the master branch.\ncustom SSL CA Cert support in python-cinderclient was introduced\nin 5.0.0, but horizon train supports python-cinderclient \u003e\u003d4.0.1,\nso we cannot backport it as-is.\n\nThis commit borrowed the logic of get_server_version() from\npython-cinderclient 5.0.0 so that this can be backported to\nolder releases than train.\n\nFrom the above reason, this is directly proposed to stable/train.\n\nChange-Id: I776d2c8c6864067e7b44ce4b2f107c80b8b6d7fb\nCloses-Bug: #1744670\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/5d9b7da4799241707f0f221bfe1d12a782d12188"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/5d9b7da4799241707f0f221bfe1d12a782d12188"}]},"branch":"refs/heads/stable/train"},"035498db5ef1479f83eacaeb1b2ec11f086656e8":{"kind":"REWORK","_number":2,"created":"2020-09-28 06:56:58.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/74/710174/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/74/710174/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/74/710174/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/74/710174/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/74/710174/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/74/710174/2"}}},"commit":{"parents":[{"commit":"e058e381607e93439a59f47c36e6a9b7cf61cc71","subject":"Allow to evacuate without specifying a target host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/e058e381607e93439a59f47c36e6a9b7cf61cc71"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-02-27 02:29:23.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-09-28 06:54:40.000000000","tz":540},"subject":"Add custom SSL CA Cert support for api.cinder.get_microversion","message":"Add custom SSL CA Cert support for api.cinder.get_microversion\n\nThis is a stable branch version of\ncommit 285c51f7e1b27a786fa9684abdc489998285b4e2 in the master branch.\ncustom SSL CA Cert support in python-cinderclient was introduced\nin 5.0.0, but horizon train supports python-cinderclient \u003e\u003d4.0.1,\nso we cannot backport it as-is.\n\nThis commit borrowed the logic of get_server_version() from\npython-cinderclient 5.0.0 so that this can be backported to\nolder releases than train.\n\nFrom the above reason, this is directly proposed to stable/train.\n\nChange-Id: I776d2c8c6864067e7b44ce4b2f107c80b8b6d7fb\nCloses-Bug: #1744670\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/035498db5ef1479f83eacaeb1b2ec11f086656e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/035498db5ef1479f83eacaeb1b2ec11f086656e8"}]},"branch":"refs/heads/stable/train"},"68d4c5e78db61ea401ee89c84a69676e26a636a1":{"kind":"REWORK","_number":3,"created":"2020-09-28 08:06:58.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/74/710174/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/74/710174/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/74/710174/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/74/710174/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/74/710174/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/74/710174/3"}}},"commit":{"parents":[{"commit":"e058e381607e93439a59f47c36e6a9b7cf61cc71","subject":"Allow to evacuate without specifying a target host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/e058e381607e93439a59f47c36e6a9b7cf61cc71"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-02-27 02:29:23.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-09-28 08:06:32.000000000","tz":540},"subject":"Add custom SSL CA Cert support for api.cinder.get_microversion","message":"Add custom SSL CA Cert support for api.cinder.get_microversion\n\nThis is a stable branch version of\ncommit 285c51f7e1b27a786fa9684abdc489998285b4e2 in the master branch.\ncustom SSL CA Cert support in python-cinderclient was introduced\nin 5.0.0, but horizon train supports python-cinderclient \u003e\u003d4.0.1,\nso we cannot backport it as-is.\n\nThis commit borrowed the logic of get_server_version() from\npython-cinderclient 5.0.0 so that this can be backported to\nolder releases than train.\n\nFrom the above reason, this is directly proposed to stable/train.\n\nChange-Id: I776d2c8c6864067e7b44ce4b2f107c80b8b6d7fb\nCloses-Bug: #1744670\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/68d4c5e78db61ea401ee89c84a69676e26a636a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/68d4c5e78db61ea401ee89c84a69676e26a636a1"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
