)]}'
{"id":"openstack%2Fhorizon~712902","triplet_id":"openstack%2Fhorizon~stable%2Ftrain~I45e7cb5711b422cd9584b671f4a5549d5eca39ed","project":"openstack/horizon","branch":"stable/train","hashtags":[],"change_id":"I45e7cb5711b422cd9584b671f4a5549d5eca39ed","subject":"DNR. Test CI","status":"ABANDONED","created":"2020-03-13 10:29:20.000000000","updated":"2020-03-26 15:23:12.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0ad33c504ead8a433bdae0a1f64dcce293d9d94b","_number":712902,"virtual_id_number":712902,"owner":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-03-21 20:13:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":1,"date":"2020-03-24 10:26:12.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-21 20:13:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-24 10:26:12.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"}],"messages":[{"id":"a80b7281847bc1f08f02cb9a8bb648d9c9aec83f","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-03-13 10:29:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac31acfa90fd68878e9abedd37dfe5d97d902dad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 11:39:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/b3352da7632f41c5b044325edf327b94 : SUCCESS in 5m 45s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/972338f430474c8585d160e6fceeab0e : SUCCESS in 7m 44s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/7b375ca75e4c43bdb3741ef94d3e876c : SUCCESS in 8m 28s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/5bc8a2ec98034f778563c3b387c461b3 : SUCCESS in 19m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3e9ad709f984749bbd9c54ecf53c468 : SUCCESS in 9m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b790fa9404842368ac7179ef70551df : SUCCESS in 6m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/17b369c43e8f47f296e317c83726d906 : SUCCESS in 9m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5d7310f54154839b35944fd364ebf65 : SUCCESS in 9m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5a1350b3785b4625a5922ed40e14c17e : FAILURE in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fffdc8035e040f8b2d48a9c5310b723 : SUCCESS in 6m 32s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/02da1065ce0f4a11a0f3d3770bd19906 : SUCCESS in 7m 40s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/73ee02ebfda74b6aa5bba84a3d052fce : SUCCESS in 1h 07m 53s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/418e925f55ec4f7aa2d7e79cf207fa14 : SUCCESS in 50m 25s\n- horizon-dsvm-tempest-plugin-py27 https://zuul.opendev.org/t/openstack/build/d0930d44561d4d1b8411252573d7402d : SUCCESS in 36m 33s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/208b28ffaf724bea8743469940d8c5eb : SUCCESS in 4m 34s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/6ffec03b4a8342358a08a8b0208dcb82 : SUCCESS in 51m 26s","accounts_in_message":[],"_revision_number":1},{"id":"43bfe2334ffad2ed9ccba6d031d647552c0c9dc9","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-03-18 15:22:17.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit d655bbc5ba21a6937cab2b28863c76808e2470d4","accounts_in_message":[],"_revision_number":1},{"id":"9e3a5d82112d861b043ad5e23d4b4dcd8de7e2cd","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-03-21 18:45:25.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"257c7236725b053f577f23fe4f3c7c5f743e46a0","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-03-21 19:05:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f3656f8e22a803f521cf6b8293dbdf15c931b808","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 20:13:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/22f8e571cbc74c36b500b085a8a11053 : SUCCESS in 4m 28s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/11de8e3dd0064220a8cc04510a5eefc5 : SUCCESS in 6m 49s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/8048ed5e837c489489730ffd4b606be0 : SUCCESS in 9m 43s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/b90c71e0030c402faeada06faea806cc : SUCCESS in 10m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd8682db1aa04d74a9bfe72c2a615194 : SUCCESS in 9m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3cbe27c68f74094831714e90a9ae7c5 : SUCCESS in 5m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1e0d26a74b9f49bc9b80d1a6d0e8bdc7 : SUCCESS in 10m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1aa0a3d21c94878ab65f90cc742105f : SUCCESS in 8m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d9a41d1a39f94854ba15a7ca056f2630 : SUCCESS in 9m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad7417abf8c540a69c66d44392d048ff : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e90dd0216f254cf1a175342995d4e6ce : SUCCESS in 25m 35s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/4c593767d5a1413c8b91b8b88564aac1 : SUCCESS in 6m 01s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/f6cdc19bf0c74a619f81341c84afadfa : SUCCESS in 1h 02m 35s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/ef4e1fb088ee42edb15a9c8cdcf979fe : SUCCESS in 32m 51s\n- horizon-dsvm-tempest-plugin-py27 https://zuul.opendev.org/t/openstack/build/1b051573a73141409f0df61d3110ced8 : SUCCESS in 37m 01s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a1ab73d3021045318bf2c9f3548ed6cf : SUCCESS in 4m 24s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/2ec94edde7a241a2bc06e16312d53a71 : SUCCESS in 41m 25s","accounts_in_message":[],"_revision_number":3},{"id":"0b6f88d3de72a7a7c4ab75588baf6fb662b0de27","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2020-03-24 10:26:12.000000000","message":"Patch Set 3: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"0ad33c504ead8a433bdae0a1f64dcce293d9d94b","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-03-26 15:23:12.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"342c8ecf9ee8be06e7ad765af9fa4dc76b8d0a3e","revisions":{"ff5ec954a99d7029b37d6c3b3cc2ef25f6918c79":{"kind":"REWORK","_number":1,"created":"2020-03-13 10:29:20.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/02/712902/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/02/712902/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/02/712902/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/02/712902/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/02/712902/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/02/712902/1"}}},"commit":{"parents":[{"commit":"e058e381607e93439a59f47c36e6a9b7cf61cc71","subject":"Allow to evacuate without specifying a target host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/e058e381607e93439a59f47c36e6a9b7cf61cc71"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-13 10:28:57.000000000","tz":120},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-13 10:28:57.000000000","tz":120},"subject":"DNR. Test CI","message":"DNR. Test CI\n\nChange-Id: I45e7cb5711b422cd9584b671f4a5549d5eca39ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/ff5ec954a99d7029b37d6c3b3cc2ef25f6918c79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/ff5ec954a99d7029b37d6c3b3cc2ef25f6918c79"}]},"branch":"refs/heads/stable/train"},"17f9a312a0b108ede969dc287d7f83f7d1f08ce0":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-03-21 18:45:25.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/02/712902/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/02/712902/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/02/712902/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/02/712902/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/02/712902/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/02/712902/2"}}},"commit":{"parents":[{"commit":"e058e381607e93439a59f47c36e6a9b7cf61cc71","subject":"Allow to evacuate without specifying a target host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/e058e381607e93439a59f47c36e6a9b7cf61cc71"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-13 10:28:57.000000000","tz":120},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-21 18:45:23.000000000","tz":0},"subject":"DNR. Test CI","message":"DNR. Test CI\n\nDepends-On: https://review.opendev.org/714275\nChange-Id: I45e7cb5711b422cd9584b671f4a5549d5eca39ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/17f9a312a0b108ede969dc287d7f83f7d1f08ce0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/17f9a312a0b108ede969dc287d7f83f7d1f08ce0"}]},"branch":"refs/heads/stable/train"},"342c8ecf9ee8be06e7ad765af9fa4dc76b8d0a3e":{"kind":"REWORK","_number":3,"created":"2020-03-21 19:05:45.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/02/712902/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/02/712902/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/02/712902/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/02/712902/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/02/712902/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/02/712902/3"}}},"commit":{"parents":[{"commit":"e058e381607e93439a59f47c36e6a9b7cf61cc71","subject":"Allow to evacuate without specifying a target host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/e058e381607e93439a59f47c36e6a9b7cf61cc71"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-13 10:28:57.000000000","tz":120},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-21 19:05:25.000000000","tz":120},"subject":"DNR. Test CI","message":"DNR. Test CI\n\nDepends-On: https://review.opendev.org/714275\nChange-Id: I45e7cb5711b422cd9584b671f4a5549d5eca39ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/342c8ecf9ee8be06e7ad765af9fa4dc76b8d0a3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/342c8ecf9ee8be06e7ad765af9fa4dc76b8d0a3e"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
