)]}'
{"id":"openstack%2Fhorizon~713640","triplet_id":"openstack%2Fhorizon~stable%2Ftrain~I38a1c2c3455a10b4852d25f6c39b3c21f5c450fc","project":"openstack/horizon","branch":"stable/train","hashtags":[],"change_id":"I38a1c2c3455a10b4852d25f6c39b3c21f5c450fc","subject":"Use pyScss2 and django-pyscss2","status":"ABANDONED","created":"2020-03-18 12:36:47.000000000","updated":"2020-03-24 04:20:12.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"61627182909ed6c5df7519f8d6dc8a8c4ace1f8d","_number":713640,"virtual_id_number":713640,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"date":"2020-03-18 13:17:06.000000000","_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-03-18 14:27:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":-1,"date":"2020-03-18 13:19:09.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":-1,"date":"2020-03-24 00:02:06.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-18 13:17:06.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-03-18 14:27:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-24 00:02:06.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"}],"messages":[{"id":"4aee6f2654078c734c58664e0d98506389569601","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-03-18 12:36:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e7fa3d144c16cbb4bcd2b0e4f20010b800764b56","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-03-18 12:43:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"36d3212b10f2b90672d1bad5233d53209cd5f72f","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-03-18 13:17:06.000000000","message":"Patch Set 2:\n\nDo we need a cherry-pick to requirements first?","accounts_in_message":[],"_revision_number":2},{"id":"2a5b6ce407e87dcec1cb242bf247626db58ff65d","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-03-18 13:18:08.000000000","message":"Patch Set 2:\n\n\u003e Do we need a cherry-pick to requirements first?\n\nyeah, I think so.","accounts_in_message":[],"_revision_number":2},{"id":"4d1302414c8281294174d1b0186c7bca6bc0a6ec","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-03-18 13:19:09.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"cc510f37bec26eb499e76030b64a17ee69c29aaa","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-03-18 13:23:10.000000000","message":"Patch Set 2:\n\nThis changes the requirements, so this needs to be careful. Perhaps the mailing list discussion would be fine. I know tox installs the latest setuptools, so tox jobs in stable branches fail, but I am not sure whether we should change our dependencies in released versions....","accounts_in_message":[],"_revision_number":2},{"id":"ceaab4df9cb81e31158f0a3530c4d7647b0ce25b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 14:27:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0663d5fdd640431980229a0b867392e1 : FAILURE in 3m 03s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/a31d3c4be3434f439753cc4820725f5d : SUCCESS in 7m 45s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/48d5a5f1f0614e7eb22920216ce6bc1c : SUCCESS in 8m 08s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/bb5a18a49d254e4291584486a05dac6a : SUCCESS in 8m 49s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/2dad367823974e51bad9f1df927e8a75 : SUCCESS in 9m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b810481386d45068874c1068b6f10d0 : SUCCESS in 11m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40bac3715317467aa3e57579064eaea1 : SUCCESS in 7m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c02bdc70729442dc881f593eeecd0a16 : SUCCESS in 11m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4689fcfa57045478904241e0004dcbf : SUCCESS in 9m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d086d2d3dc03414dbb6b73ae6f1b3c13 : SUCCESS in 8m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49a4f881ea7145d6943c98d68092847f : SUCCESS in 7m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9254d0815429462fb6a167cbc74657ef : SUCCESS in 25m 11s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/bfcbeab9127c40e2b381acc9f2b258c6 : SUCCESS in 8m 34s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/fcdccb0fe7834d339956f0aaf2cafe44 : RETRY_LIMIT in 29m 27s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/8130e64d8f5244cb95ac6a66987c520d : FAILURE in 33m 51s\n- horizon-dsvm-tempest-plugin-py27 https://zuul.opendev.org/t/openstack/build/61149e2d8e30496492559393da97d69f : FAILURE in 33m 13s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/81504519e7d9433883e517a4badffb06 : SUCCESS in 5m 15s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/e13ed47fc27b4fa3a44c7a272a5b5743 : FAILURE in 38m 00s","accounts_in_message":[],"_revision_number":2},{"id":"13f89d81a59ebae69eccf3a118fd8be405a51aa0","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-03-24 00:02:06.000000000","message":"Patch Set 2: Workflow-1\n\nWe are thinking to switch back to pyscss as e0ne is now a part of the maintainers of pyscss.","accounts_in_message":[],"_revision_number":2},{"id":"61627182909ed6c5df7519f8d6dc8a8c4ace1f8d","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-03-24 04:20:12.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7e4eaba9d0abcb78034b119f12cb75e8cf7e2c5f","revisions":{"5e21f414ffaeb6752b36ad8afaf29a2b9a466d23":{"kind":"REWORK","_number":1,"created":"2020-03-18 12:36:47.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/40/713640/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/40/713640/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/40/713640/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/40/713640/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/40/713640/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/40/713640/1"}}},"commit":{"parents":[{"commit":"e058e381607e93439a59f47c36e6a9b7cf61cc71","subject":"Allow to evacuate without specifying a target host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/e058e381607e93439a59f47c36e6a9b7cf61cc71"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-13 15:35:51.000000000","tz":120},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-03-18 12:35:06.000000000","tz":0},"subject":"Use pyScss2 and django-pyscss2","message":"Use pyScss2 and django-pyscss2\n\npyScss and django-pyscss packages are not maintained for a long time. We\ncan\u0027t use the pyScss with teh setuptools\u003e\u003d46.0.0, so I forked these\nproject and switch horizon to use them.\n\nChange-Id: I38a1c2c3455a10b4852d25f6c39b3c21f5c450fc\nDepends-On: https://review.opendev.org/712976\n(cherry picked from commit 11d4954cc3e4296921d1791610f7df24d46dea4d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/5e21f414ffaeb6752b36ad8afaf29a2b9a466d23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/5e21f414ffaeb6752b36ad8afaf29a2b9a466d23"}]},"branch":"refs/heads/stable/train"},"7e4eaba9d0abcb78034b119f12cb75e8cf7e2c5f":{"kind":"REWORK","_number":2,"created":"2020-03-18 12:43:33.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/40/713640/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/40/713640/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/40/713640/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/40/713640/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/40/713640/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/40/713640/2"}}},"commit":{"parents":[{"commit":"e058e381607e93439a59f47c36e6a9b7cf61cc71","subject":"Allow to evacuate without specifying a target host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/e058e381607e93439a59f47c36e6a9b7cf61cc71"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-03-13 15:35:51.000000000","tz":120},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-03-18 12:42:37.000000000","tz":0},"subject":"Use pyScss2 and django-pyscss2","message":"Use pyScss2 and django-pyscss2\n\npyScss and django-pyscss packages are not maintained for a long time. We\ncan\u0027t use the pyScss with teh setuptools\u003e\u003d46.0.0, so I forked these\nproject and switch horizon to use them.\n\nChange-Id: I38a1c2c3455a10b4852d25f6c39b3c21f5c450fc\nDepends-On: https://review.opendev.org/712976\n(cherry picked from commit 11d4954cc3e4296921d1791610f7df24d46dea4d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/7e4eaba9d0abcb78034b119f12cb75e8cf7e2c5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/7e4eaba9d0abcb78034b119f12cb75e8cf7e2c5f"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
