)]}'
{"id":"openstack%2Fhorizon~717547","triplet_id":"openstack%2Fhorizon~master~Ic12c84ad13a879f1ddca6d35c5bd3ceb24cec721","project":"openstack/horizon","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"Ic12c84ad13a879f1ddca6d35c5bd3ceb24cec721","subject":"Imported Translations from Zanata","status":"MERGED","created":"2020-04-05 07:32:22.000000000","updated":"2020-04-08 11:17:13.000000000","submitted":"2020-04-08 11:13:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"717547-1586344398420-13a268f8","meta_rev_id":"a7a34e98389dbb5922ec54c244f176c9676514eb","_number":717547,"virtual_id_number":717547,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"},{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-08 11:13:18.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"recommended":{"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"},"all":[{"value":1,"date":"2020-04-07 09:10:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"},{"value":2,"date":"2020-04-08 10:48:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":0,"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"},{"value":1,"date":"2020-04-08 10:48:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"},{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-07 09:10:57.000000000","updated_by":{"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"},"reviewer":{"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"},"state":"REVIEWER"},{"updated":"2020-04-08 10:48:33.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2020-04-08 11:13:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"70cd3b834bb6446f06f3d37f1f132d0362555b5e","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-05 07:32:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"978eabbbad4b6152f53f671d461ae6f1a3f9e12a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-05 07:57:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/2823e5e5e9a24f4f8862b6c9d85cb496 : SUCCESS in 5m 57s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/329c1ac7f1f243bf8baa10358d773f97 : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b53f20d3d7448f79c82eaff22ae2d75 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/503bddd7ac8147d8bb04c84090514626 : SUCCESS in 8m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3201d88d915f457ba30b73571468b897 : SUCCESS in 19m 48s","accounts_in_message":[],"_revision_number":1},{"id":"849af5f063f8c964121f0ac4c75ec5582f9e4e26","author":{"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"},"date":"2020-04-07 09:10:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"98497af826c2b772281d57ccf51c6a72429235f3","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-04-08 10:48:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"91ba9a362f7a633c2db51c92473745d7f1154f2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 10:48:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e7b1737084f0e9a80464b7f3f718f0f641777b82","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 11:13:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"dd3b7d7f39261a2be0b06ff052fb9bd05757dce4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 11:13:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/99e85d319d264a5bab124cb05f38b366 : SUCCESS in 8m 31s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/29111a16c30444c8acf6c30d90be8021 : SUCCESS in 11m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ddce39e210847c5b699085a565a051e : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a98250cfd25d49eea6c2544f75e9629f : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4efc8e4d4daa4197ab47dad34f2b6dd8 : SUCCESS in 23m 12s","accounts_in_message":[],"_revision_number":1},{"id":"3cd6bccd70d7fbcea3a9e33ad962c1ccf4303750","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 11:17:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b22c41ed751d47728dfb97a961d5db42 : SUCCESS in 2m 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/594d6b1cfb43478da31062c32aa57af6 : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a04b26e96c32fe1b50015b8322ea7d88b5c2bf92","revisions":{"a04b26e96c32fe1b50015b8322ea7d88b5c2bf92":{"kind":"REWORK","_number":1,"created":"2020-04-05 07:32:22.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/47/717547/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/47/717547/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/47/717547/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/47/717547/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/47/717547/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/47/717547/1"}}},"commit":{"parents":[{"commit":"63cf09e07ead9f6810f2a850f29cf47da3ffae2e","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/63cf09e07ead9f6810f2a850f29cf47da3ffae2e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-05 07:32:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-05 07:32:17.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Ic12c84ad13a879f1ddca6d35c5bd3ceb24cec721\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/a04b26e96c32fe1b50015b8322ea7d88b5c2bf92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/a04b26e96c32fe1b50015b8322ea7d88b5c2bf92"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
