)]}'
{"id":"openstack%2Fhorizon~723328","triplet_id":"openstack%2Fhorizon~stable%2Fussuri~Ia3869b75fa2e8bde7829475a00f63d7bec0f97eb","project":"openstack/horizon","branch":"stable/ussuri","topic":"zanata/translations","hashtags":[],"change_id":"Ia3869b75fa2e8bde7829475a00f63d7bec0f97eb","subject":"Imported Translations from Zanata","status":"MERGED","created":"2020-04-27 07:06:55.000000000","updated":"2020-04-27 09:26:37.000000000","submitted":"2020-04-27 09:23:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723328-1587979405190-54f1cd87","meta_rev_id":"d1867b7fe8218a303835b7a4fdf1ccc4757eed11","_number":723328,"virtual_id_number":723328,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-27 09:23:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":2,"date":"2020-04-27 09:04:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":1,"date":"2020-04-27 09:04:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-27 09:04:30.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-04-27 09:23:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"30303afaee58e5c0d72208da6ffb6f090e2d6bed","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-27 07:06:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3e32466bb4a9c38f0d084a1284ce0f6cb76e8fb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 07:23:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/59ca5df9330448bc9e2393ddc18e57f2 : SUCCESS in 6m 55s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/cc66e3d849eb4ee789fce5c1ba1aa948 : SUCCESS in 11m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/033daa9eb36e40b3899728d1931fba25 : SUCCESS in 12m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5982e75aa5d143aba5b0836730b99124 : SUCCESS in 10m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4858e22f2c8e4b7f975d65da028906a4 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e679069c6ce4f05b814450213a33216 : SUCCESS in 11m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d26bd3d19a3c4a1ea0846a6b5d2a07f2 : SUCCESS in 9m 24s","accounts_in_message":[],"_revision_number":1},{"id":"5b5cff3739613e47fe29236e9d27c5c4c3584955","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-04-27 09:04:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d270d3e693bb83f6ad6c988e902e5ca14b5edd12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 09:04:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"25ec082126c4142deffde8c8833bc3dbbfffd9a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 09:23:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/e595b0d70b93481abafa3aacc60be222 : SUCCESS in 7m 17s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/ca69edd93b224d18aa621a7140eb119f : SUCCESS in 7m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ea129251e4a742eca33c7bd27f4e78f5 : SUCCESS in 11m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/38153d7ef3e74250812cbaa08d1f74ff : SUCCESS in 13m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e65c1838707b494eba5b5c2415658e34 : SUCCESS in 7m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11934ab9206241f793169c9acfca59f0 : SUCCESS in 14m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aab7ec29375343c18c4b29641e7492c1 : SUCCESS in 8m 01s","accounts_in_message":[],"_revision_number":1},{"id":"94326a00f58a3ac9839e3240259472b3e90f068c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 09:23:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"217828667f12c7dd47d1120d7723ed335ebdd8ea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 09:26:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee0923ce292c4ba3ba9f4e5cc77c9916 : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6968bc09a7e40566a18f34b98aea94fd4417b50e","revisions":{"6968bc09a7e40566a18f34b98aea94fd4417b50e":{"kind":"REWORK","_number":1,"created":"2020-04-27 07:06:55.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/28/723328/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/28/723328/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/28/723328/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/28/723328/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/28/723328/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/28/723328/1"}}},"commit":{"parents":[{"commit":"1b9ccdbab22c0b6b91a292a154b24d0d18b495cb","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/1b9ccdbab22c0b6b91a292a154b24d0d18b495cb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-27 07:06:52.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-27 07:06:52.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Ia3869b75fa2e8bde7829475a00f63d7bec0f97eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/6968bc09a7e40566a18f34b98aea94fd4417b50e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/6968bc09a7e40566a18f34b98aea94fd4417b50e"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
