)]}'
{"id":"openstack%2Fhorizon~724949","triplet_id":"openstack%2Fhorizon~stable%2Fussuri~I970b07e52ed7f2f4440b47800fcb3b6528dce22d","project":"openstack/horizon","branch":"stable/ussuri","topic":"fix-keystoneauth-cliexc-arg-stable/ussuri","hashtags":[],"change_id":"I970b07e52ed7f2f4440b47800fcb3b6528dce22d","subject":"Give keystoneauth1.ClientException only 1 argument (message)","status":"MERGED","created":"2020-05-02 00:43:20.000000000","updated":"2020-06-23 18:56:03.000000000","submitted":"2020-06-23 18:53:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"724949-1592938427919-48842415","meta_rev_id":"866a69e6e1c52c77c0c23f5700c7e729dd9cbeea","_number":724949,"virtual_id_number":724949,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"date":"2020-05-02 00:49:19.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":31241,"name":"wdoekes","email":"walter+ubuntu@wjd.nu","username":"wdoekes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-23 18:53:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"recommended":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":1,"date":"2020-05-03 10:34:06.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2020-05-06 09:12:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":1,"date":"2020-05-03 08:25:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31241,"name":"wdoekes","email":"walter+ubuntu@wjd.nu","username":"wdoekes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":1,"date":"2020-06-19 15:43:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":31241,"name":"wdoekes","email":"walter+ubuntu@wjd.nu","username":"wdoekes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":31241,"name":"wdoekes","email":"walter+ubuntu@wjd.nu","username":"wdoekes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":31241,"name":"wdoekes","email":"walter+ubuntu@wjd.nu","username":"wdoekes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-03 08:25:16.000000000","updated_by":{"_account_id":31241,"name":"wdoekes","email":"walter+ubuntu@wjd.nu","username":"wdoekes"},"reviewer":{"_account_id":31241,"name":"wdoekes","email":"walter+ubuntu@wjd.nu","username":"wdoekes"},"state":"REVIEWER"},{"updated":"2020-05-03 10:34:06.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2020-06-19 15:43:48.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-06-23 18:53:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8acb92fedd048b115aeab80eb3f646f0563b21d2","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-05-02 00:43:20.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"53089b70d4e2f65ef269e3db6c5b4befabb55a93","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-05-02 00:49:19.000000000","message":"Patch Set 1:\n\nThis is a trivial fix but a backport to stable/train was already proposed and I see no reason to reject it, so I proposed this backport to stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"9933da2e51086f1fa23669f9d003d300b9e59ebd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-02 02:12:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/0dbe5412f94e4930949f77ee43ce7fa9 : SUCCESS in 5m 14s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/b7dbe0a4f59b4bc797c5fa607b75dcf1 : SUCCESS in 6m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a635b8146d5c49689aad2a401571bd79 : SUCCESS in 11m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba87d05f1bde430795caf7fdff414fe2 : SUCCESS in 11m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87998cf0a6a44e7caf3f63c010c577d1 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65ff2475917741a9a11e0d5428722470 : SUCCESS in 11m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0734ddd404a44febfe293879029ece5 : SUCCESS in 5m 57s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/7b67c8234ccc4eaeae86b4214edd7ddb : SUCCESS in 6m 43s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/5a0436fa28834d0f912b7fddf728340e : SUCCESS in 1h 23m 23s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/1a6d78ab5a844101a0a520c11f7adb7c : SUCCESS in 32m 04s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f4a905b13db14b6c8210322cc7a4e30e : SUCCESS in 4m 45s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/dd4668e2792a4bc69155bc562944d745 : SUCCESS in 32m 17s","accounts_in_message":[],"_revision_number":1},{"id":"aa3d3ffd9c4554a9f6b32958e9865286e96fe9f6","author":{"_account_id":31241,"name":"wdoekes","email":"walter+ubuntu@wjd.nu","username":"wdoekes"},"date":"2020-05-03 08:25:16.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"900d1a693d5accf36467400242099589fe73bcbf","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-05-03 10:34:06.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4c7586c990a9d2844df8f254977aa8797b92395a","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-05-06 09:12:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3564c9f30b5babad5da9b46eec1fd456af2b964a","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-06-19 15:43:48.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8b1666f6b39258ca7cff360f9410b04b5646c87e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 15:44:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a890fd8036fe0b518660560f72c8c94433676280","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 17:18:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/c40596ccbce3419aac38debaddb7914e : FAILURE in 7m 06s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/c8e2f5c8a4ab46da960cad5c64a13939 : FAILURE in 8m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/19aea05724e94a83aee8547229422dc2 : SUCCESS in 11m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3334fa90310f4091a313a2e71b07858a : SUCCESS in 9m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83cdf730ccba4fb6868caa49b234b126 : SUCCESS in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8944cc05e7034be99112c038e6bcfc21 : SUCCESS in 10m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63f6f1eb48a446f9b2fd3595034e9112 : SUCCESS in 8m 50s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/97000d343aa44528a5eff73ce18c353c : SUCCESS in 41m 19s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/ac87bbb544654a049daf187c913ea736 : SUCCESS in 10m 13s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/840639add6ef487d9468c48fd7b15747 : SUCCESS in 1h 26m 44s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/f257875b12ab46d6afa1ed6483024526 : SUCCESS in 34m 58s","accounts_in_message":[],"_revision_number":1},{"id":"9d063dc64bab5e4ebe335b9344a03b8748619a84","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-06-23 12:21:29.000000000","message":"Patch Set 1:\n\nrecheck.","accounts_in_message":[],"_revision_number":1},{"id":"8506e2b588576a7bca80e5e964ac1d36618e732a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 14:20:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/3d1ea49a30ea4568915197c9293e71d1 : POST_FAILURE in 8m 31s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/3adf7295e2254d1794b930709e28f40d : POST_FAILURE in 9m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fca34800616a4a8db88841cce9ba5c93 : SUCCESS in 12m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a7856696049441ebd4621bfb5c89dd8 : SUCCESS in 14m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d67628ef2349432eb5f0fb77f32f8ed9 : SUCCESS in 7m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7fe90861d9ea4d8f97ae489d5785bfbb : SUCCESS in 15m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/187105234012413489d2aba320f71881 : SUCCESS in 8m 15s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/4136f35067874d4682fa45276e8dce61 : SUCCESS in 9m 12s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/684b0deb1ff4423086f67f289eb6189e : POST_FAILURE in 1h 54m 47s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/204ec91e3f3c4811b7d97888e0efebaf : POST_FAILURE in 35m 18s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d137e1cfc47e4c6a9e4238148c9c5d3d : POST_FAILURE in 6m 40s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/3c6adb5774af4707ac446b26c9f948b1 : SUCCESS in 40m 08s","accounts_in_message":[],"_revision_number":1},{"id":"e6d1a913ade272639ec688c8099db1b556baea54","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-06-23 15:48:07.000000000","message":"Patch Set 1:\n\nrecheck.","accounts_in_message":[],"_revision_number":1},{"id":"5582dc843b51e87326a16e30f7d4a4d503d41efc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 17:15:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/11b8a8b21f064233baf4faec643e8047 : SUCCESS in 7m 26s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/62f8393427d046a9b6e6bc708f8f9503 : SUCCESS in 8m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bbbcbe8156704edca3a15bac6a0bfa67 : SUCCESS in 12m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ec012b630e04f3a82d9980ef8d20804 : SUCCESS in 11m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1a0135d30684e08a216012d405a1ad4 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/033d0dbaff1f47cf9bbca083cae8426c : SUCCESS in 8m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f585cacbf8e84027ba13997ea807ef4b : SUCCESS in 10m 00s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/4be6539f4933454189d997b17261bc5e : SUCCESS in 8m 59s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/9d198ae646ff41ed9e4bd7489868b24b : SUCCESS in 1h 18m 36s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/71d3f43573bb4955ba49c4a8f7cd40d3 : SUCCESS in 29m 07s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f2c1945fd84b4631913d863eac5456e3 : SUCCESS in 6m 01s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/693b052ee6684cccac97ea0935cb6903 : SUCCESS in 50m 19s","accounts_in_message":[],"_revision_number":1},{"id":"f7f9f149fd663e3ef3c4e50e6b05851eaa01eb18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 17:16:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"76a06acf861d22127b1e3e4eee83dc5ad968cf28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 18:53:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/7ac6683113974cd6a2e9c624224c536a : SUCCESS in 7m 20s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/11eec23053a446a084d568d16fe7ff9b : SUCCESS in 14m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0bb06213941a4fd9a2ad05b56135eff3 : SUCCESS in 12m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d06d17f31954726aa5f2bd4a2d67f66 : SUCCESS in 11m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b37a060789334fba8a254583d524baff : SUCCESS in 12m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2fa04b6a51bb4e28bd71c8eba5c475d3 : SUCCESS in 12m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/848c0b66cc20407fb965964e9efd6c69 : SUCCESS in 8m 48s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/79d70c485bd74d86b1582a8680db439a : SUCCESS in 39m 42s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/d4550b7979c049fea37dcce98aed0b86 : SUCCESS in 9m 47s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/ecdedbfa7893451d8974e6807cf45f1c : SUCCESS in 1h 36m 26s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/49613ca39272454389cdd3090f355595 : SUCCESS in 38m 28s","accounts_in_message":[],"_revision_number":1},{"id":"4a7c275f058c240c03baa2fb9d96f33d13dd01db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 18:53:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9a70bd1488b2b60483a7450747952979120ecf67","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 18:56:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a7d9fbdb4214f62813c7629b08d226a : SUCCESS in 1m 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"11cce6e71be994683b442e02d00e194edb1fa65f","revisions":{"11cce6e71be994683b442e02d00e194edb1fa65f":{"kind":"REWORK","_number":1,"created":"2020-05-02 00:43:20.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/49/724949/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/49/724949/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/49/724949/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/49/724949/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/49/724949/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/49/724949/1"}}},"commit":{"parents":[{"commit":"510757b6b30aad01b8fbb4f2929c4370e64b81bd","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/510757b6b30aad01b8fbb4f2929c4370e64b81bd"}]}],"author":{"name":"Walter Doekes","email":"walter+github@wjd.nu","date":"2019-11-01 14:34:41.000000000","tz":60},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-05-02 00:43:20.000000000","tz":0},"subject":"Give keystoneauth1.ClientException only 1 argument (message)","message":"Give keystoneauth1.ClientException only 1 argument (message)\n\nChange-Id: I970b07e52ed7f2f4440b47800fcb3b6528dce22d\n(cherry picked from commit 941f64915e825dedc2c58621e8390953bfcc4b7b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/11cce6e71be994683b442e02d00e194edb1fa65f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/11cce6e71be994683b442e02d00e194edb1fa65f"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
