)]}'
{"id":"openstack%2Fhorizon~726713","triplet_id":"openstack%2Fhorizon~stable%2Fqueens~I3dcf37a3bf9c6e95c21ff3fb99e6a221049f1ef8","project":"openstack/horizon","branch":"stable/queens","topic":"health-check","hashtags":[],"change_id":"I3dcf37a3bf9c6e95c21ff3fb99e6a221049f1ef8","subject":"DNM: gate health check","status":"ABANDONED","created":"2020-05-11 00:58:50.000000000","updated":"2020-05-16 08:26:02.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9129ac549bf8044641b5a9d4e0104b611db48b5a","_number":726713,"virtual_id_number":726713,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-05-15 19:49:15.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-05-11 01:35:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-11 01:35:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9e8a562f1957315165fbd6fd1cd12d78575c4312","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-05-11 00:58:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43d05d688ca3255a80556b69c8f61d79f51186f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 01:35:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/903c3c2bfbfd4de2b3992b6d2aa50ac4 : SUCCESS in 4m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7c7341870eb64c3189886398c3732576 : SUCCESS in 10m 54s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6ca756b0d8e9493786babc87199b3b4a : SUCCESS in 10m 30s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/c806c1fa89bf4f91a2d7a9cd3ad15fbd : SUCCESS in 4m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e5e1ea5a931453da4078506bddbce81 : SUCCESS in 31m 43s\n- nodejs-npm-run-lint https://zuul.opendev.org/t/openstack/build/a421ca4f302d4e69adec610cc2e4693b : RETRY_LIMIT in 4m 00s\n- nodejs-npm-run-test https://zuul.opendev.org/t/openstack/build/17c603097ded49ef9187474cc6b5ad1a : RETRY_LIMIT in 2m 23s\n- horizon-openstack-tox-py27dj110 https://zuul.opendev.org/t/openstack/build/65acb8761cb64842a8104a3e79627752 : SUCCESS in 10m 41s\n- horizon-openstack-tox-py27dj19 https://zuul.opendev.org/t/openstack/build/fd8a85f82bd348d5a04004054bfde70b : SUCCESS in 10m 22s\n- horizon-openstack-tox-py27dj18 https://zuul.opendev.org/t/openstack/build/6ed36686b9734f1487d6b27dc9423eac : SUCCESS in 11m 42s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/cd0d58d458314312932d9adff63ba72a : SUCCESS in 4m 24s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/57b9e28d75e54f109eb32312151e9cfa : FAILURE in 30m 23s","accounts_in_message":[],"_revision_number":1},{"id":"d35f6d2fcef150f5c2d3606b7c4f046e5d8eb2f1","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-05-15 19:49:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f72c38214c40a5e35cf8030f01610f3711a62a66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 20:37:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0413bd25147a4fe39a66bc410cdb6551 : SUCCESS in 5m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/092979889ad94a84bbecddcf094e05fc : SUCCESS in 10m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2e0e05bf74544dd8a53a5df7b94a6abd : SUCCESS in 10m 11s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/6b62a51314e74d3d91215888a273321a : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5de75d03fba94798b26a393e4706dd32 : SUCCESS in 32m 41s\n- nodejs-run-lint https://zuul.opendev.org/t/openstack/build/cadb6005b2764d9292813c2fa87ce424 : SUCCESS in 11m 25s\n- nodejs-run-test-browser https://zuul.opendev.org/t/openstack/build/83a1ea3354cb49bf9ee979eb8d6c5f03 : SUCCESS in 13m 22s\n- horizon-openstack-tox-py27dj110 https://zuul.opendev.org/t/openstack/build/9035c4ad6c264768a8ac6c84b1b8ddf9 : SUCCESS in 9m 25s\n- horizon-openstack-tox-py27dj19 https://zuul.opendev.org/t/openstack/build/5ed37331a4cc4e3cb7989571eaeaca56 : SUCCESS in 11m 05s\n- horizon-openstack-tox-py27dj18 https://zuul.opendev.org/t/openstack/build/4f0691e04c3f430b9e3ea55814c3101c : SUCCESS in 10m 23s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/03f0df6228a54bdeba15093e73f72fb0 : SUCCESS in 4m 34s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/a0f4477aac784774905826c9d3f67d1b : FAILURE in 32m 45s","accounts_in_message":[],"_revision_number":1},{"id":"4bb5bb2d424180ccefa4fa502701dc02bbfec96d","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-05-16 08:26:02.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0585ca449b2e28c9501ac269104d8bf539ad0542","revisions":{"0585ca449b2e28c9501ac269104d8bf539ad0542":{"kind":"REWORK","_number":1,"created":"2020-05-11 00:58:50.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/13/726713/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/13/726713/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/726713/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/726713/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/726713/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/13/726713/1"}}},"commit":{"parents":[{"commit":"560ffbbb74ef8a2b6acc6af7ca6e0549e822fac4","subject":"Merge \"Avoid forced logout when 403 error encountered\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/560ffbbb74ef8a2b6acc6af7ca6e0549e822fac4"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-05-11 00:58:29.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-05-11 00:58:29.000000000","tz":540},"subject":"DNM: gate health check","message":"DNM: gate health check\n\nChange-Id: I3dcf37a3bf9c6e95c21ff3fb99e6a221049f1ef8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/0585ca449b2e28c9501ac269104d8bf539ad0542"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/0585ca449b2e28c9501ac269104d8bf539ad0542"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
