)]}'
{"id":"openstack%2Fhorizon~726940","triplet_id":"openstack%2Fhorizon~stable%2Fstein~Ifcfb2bdf7480ecc94b90a7428e1029bb8f6c2b8b","project":"openstack/horizon","branch":"stable/stein","hashtags":[],"change_id":"Ifcfb2bdf7480ecc94b90a7428e1029bb8f6c2b8b","subject":"DNM: Test new javascript jobs","status":"ABANDONED","created":"2020-05-11 19:45:25.000000000","updated":"2020-05-12 16:11:54.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c92dda934e0f358abb9fa0481f6b5ecf7c0e08d4","_number":726940,"virtual_id_number":726940,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-05-12 06:57:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-12 06:57:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4352cdbd45790602d7c26b7ba6636203020b321a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-11 19:45:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c5f705d25ee4e4b7f2eebbab222fd4eb86368c8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 19:56:20.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"e7bedfc0f6328881acfd745f88f01cb8dee07425","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-11 19:56:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ef4a563353b986dfefd8add39d029869b14f9229","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 20:05:47.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"87d02f872ced155cca012ba1ce5829f62ba6e5e6","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-11 20:06:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6986e8874e4b1355731e241a8a9e6ef78125b9e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 20:19:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- nodejs4-npm-run-lint https://zuul.opendev.org/t/openstack/build/0d921be460994baaa0d4e64fc1477a04 : FAILURE in 10m 17s\n- nodejs4-npm-run-test https://zuul.opendev.org/t/openstack/build/58829dfde90a403daf97c57b19c0e28c : FAILURE in 11m 25s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/297678914e3742dabef1ed3cbb9fa520 : FAILURE in 7m 16s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/7bfaec3058eb4527b662ce25b0ed02a2 : FAILURE in 6m 14s","accounts_in_message":[],"_revision_number":1},{"id":"f9245d356e03e0e3f9663bdc6b032d1a638d69cf","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-12 05:38:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4c7edd7b42b75378f6797bccd44a497630ced684","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 05:51:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs4-npm-run-lint https://zuul.opendev.org/t/openstack/build/b4f97719c3a946e7a194a683c28a881f : SUCCESS in 6m 00s\n- nodejs4-npm-run-test https://zuul.opendev.org/t/openstack/build/f5c214b2dba14751a209777bee549807 : SUCCESS in 10m 34s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/47846e0958ee45c2a2d510c423070e7e : SUCCESS in 4m 22s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/2433dab14af14f318eed4064394afb7a : SUCCESS in 6m 49s","accounts_in_message":[],"_revision_number":1},{"id":"c7145f06d772bdfe4e8233f01c73855012b7bde3","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-12 05:57:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"93828c9cdc23f39fc7414103b6eec1a060f4150d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 06:16:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- nodejs4-npm-run-lint https://zuul.opendev.org/t/openstack/build/7950a879c84841428213f05e44814962 : SUCCESS in 14m 26s\n- nodejs4-npm-run-test https://zuul.opendev.org/t/openstack/build/74cd9be31fd94b29b0b6b970884f22c3 : SUCCESS in 17m 49s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/ce8018112fc74128a185a01abe317fa6 : SUCCESS in 5m 27s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/ec8c39babfff4ca6925d11859a4e7909 : SUCCESS in 7m 32s","accounts_in_message":[],"_revision_number":1},{"id":"6bfee7b8a9aa104504524174a5bfb5b5279f1cef","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-12 06:31:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b8721a73886b51a3050f06241c06120038288348","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 06:57:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- nodejs4-npm-run-lint https://zuul.opendev.org/t/openstack/build/4d7a383386bb48d9a4f46c98363da4d2 : SUCCESS in 8m 12s\n- nodejs4-npm-run-test https://zuul.opendev.org/t/openstack/build/b0a06bd55f8b4b9da7de43b050210b05 : RETRY_LIMIT in 5m 35s\n- nodejs4-run-lint https://zuul.opendev.org/t/openstack/build/2f6de003d6974225b6f7fa4d3400aca1 : SUCCESS in 7m 04s\n- nodejs4-run-test https://zuul.opendev.org/t/openstack/build/642757e538194574b4973ce82ccef6f0 : SUCCESS in 10m 58s\n- nodejs10-npm-run-lint https://zuul.opendev.org/t/openstack/build/6c28c3dead7a4af4ac6cc31ce1509a12 : SUCCESS in 7m 01s\n- nodejs10-npm-run-test https://zuul.opendev.org/t/openstack/build/d708b915c3954ee1ac3e5ce35ebc4eff : SUCCESS in 12m 55s\n- nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/487e4410a45d4cc0884289a752efa301 : SUCCESS in 4m 59s\n- nodejs10-run-test https://zuul.opendev.org/t/openstack/build/7b94f30d9f7742ed9dbeb0c90da3ea2f : SUCCESS in 8m 28s","accounts_in_message":[],"_revision_number":2},{"id":"c92dda934e0f358abb9fa0481f6b5ecf7c0e08d4","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-12 16:11:54.000000000","message":"Abandoned\n\ntesting done","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"184857a7fca5910296297b7433300e17e71294e4","revisions":{"735bc16a116ace98b862540c4bd94c428c356e2f":{"kind":"REWORK","_number":1,"created":"2020-05-11 19:45:25.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/40/726940/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/40/726940/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/40/726940/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/40/726940/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/40/726940/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/40/726940/1"}}},"commit":{"parents":[{"commit":"6871e03e065bc5debdd0c3d99b0547178b890531","subject":"DNM: Use nodejs10 for nodejs tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/6871e03e065bc5debdd0c3d99b0547178b890531"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-11 19:44:28.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-11 19:44:28.000000000","tz":120},"subject":"DNM: Test new javascript jobs","message":"DNM: Test new javascript jobs\n\nDepends-On: https://review.opendev.org/726900\nChange-Id: Ifcfb2bdf7480ecc94b90a7428e1029bb8f6c2b8b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/735bc16a116ace98b862540c4bd94c428c356e2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/735bc16a116ace98b862540c4bd94c428c356e2f"}]},"branch":"refs/heads/stable/stein"},"184857a7fca5910296297b7433300e17e71294e4":{"kind":"REWORK","_number":2,"created":"2020-05-12 06:31:38.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/40/726940/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/40/726940/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/40/726940/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/40/726940/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/40/726940/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/40/726940/2"}}},"commit":{"parents":[{"commit":"6871e03e065bc5debdd0c3d99b0547178b890531","subject":"DNM: Use nodejs10 for nodejs tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/6871e03e065bc5debdd0c3d99b0547178b890531"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-11 19:44:28.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-12 06:31:32.000000000","tz":120},"subject":"DNM: Test new javascript jobs","message":"DNM: Test new javascript jobs\n\nDepends-On: https://review.opendev.org/726900\nChange-Id: Ifcfb2bdf7480ecc94b90a7428e1029bb8f6c2b8b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/184857a7fca5910296297b7433300e17e71294e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/184857a7fca5910296297b7433300e17e71294e4"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
