)]}'
{"id":"openstack%2Fhorizon~741990","triplet_id":"openstack%2Fhorizon~master~Ib422331892f077d78e3f2efbdd88abafc4c52b9a","project":"openstack/horizon","branch":"master","hashtags":[],"change_id":"Ib422331892f077d78e3f2efbdd88abafc4c52b9a","subject":"Add allow_delete_snapshot_before_volume config option","status":"MERGED","created":"2020-07-20 17:01:58.000000000","updated":"2020-09-18 12:10:19.000000000","submitted":"2020-09-18 12:08:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":21,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741990-1600430897940-515a9e8b","meta_rev_id":"6482b5c0286c81053d538ea4f6b9893066d64be3","_number":741990,"virtual_id_number":741990,"owner":{"_account_id":28032,"name":"Oleksiy Petrenko","email":"opetrenko@mirantis.com","username":"enacero"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28032,"name":"Oleksiy Petrenko","email":"opetrenko@mirantis.com","username":"enacero"},{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-18 12:08:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":0,"_account_id":28032,"name":"Oleksiy Petrenko","email":"opetrenko@mirantis.com","username":"enacero"},{"value":2,"date":"2020-09-18 10:28:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":2,"date":"2020-09-18 02:00:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":0,"_account_id":28032,"name":"Oleksiy Petrenko","email":"opetrenko@mirantis.com","username":"enacero"},{"value":1,"date":"2020-09-18 10:28:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28032,"name":"Oleksiy Petrenko","email":"opetrenko@mirantis.com","username":"enacero"},{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28032,"name":"Oleksiy Petrenko","email":"opetrenko@mirantis.com","username":"enacero"},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-17 08:24:35.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-09-18 02:00:02.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-09-18 10:28:14.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2020-09-18 12:08:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c94a8391763eeccc7fc326a059a0f80df4f88776","author":{"_account_id":28032,"name":"Oleksiy Petrenko","email":"opetrenko@mirantis.com","username":"enacero"},"date":"2020-07-20 17:01:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5fdf0a67a0983f39ffe4db29709b8e7d539a4c4e","author":{"_account_id":28032,"name":"Oleksiy Petrenko","email":"opetrenko@mirantis.com","username":"enacero"},"date":"2020-07-20 17:11:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e8c056739a9c27bc37d9b9a485f192898b9755c9","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-07-20 17:17:56.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f40d5ec5a93bec2141b2571094a34fec03927a38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 18:54:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/38700b68f1d943609d5bde8084a421a5 : SUCCESS in 8m 10s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/e2bb481bcc65468897055641bbca5e83 : SUCCESS in 6m 29s (non-voting)\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/7cebc04484e147ab92ac7cad29345253 : SUCCESS in 7m 19s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/f912ffdb52dd41299c1b3524242b68ec : SUCCESS in 8m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b2ab0c3ff824e56b0489f3ad899d4ca : SUCCESS in 11m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/425364c0434e4388bdfab0ed411e8911 : SUCCESS in 11m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37223fd23b0c49d8ac441f8c8297be52 : SUCCESS in 7m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ec83f7d432844ecbf519683951adb5f : SUCCESS in 9m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07f0932a308c4d9eb1a7a73fe52cb729 : SUCCESS in 8m 10s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/46f9ee9e8d604ab183b17e7773ecafcc : SUCCESS in 7m 49s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/0a1a5cb234fe4893898f785e4ab18176 : SUCCESS in 1h 34m 53s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0cdf260490c0486295c593903932a182 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":2},{"id":"114068b6b24ed04f115ab5992aadcfd19ffe2f94","author":{"_account_id":28032,"name":"Oleksiy Petrenko","email":"opetrenko@mirantis.com","username":"enacero"},"date":"2020-07-21 08:25:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d7bf10c85a891bd8ed3cb2b0035e165048774fe1","author":{"_account_id":28032,"name":"Oleksiy Petrenko","email":"opetrenko@mirantis.com","username":"enacero"},"date":"2020-07-21 08:26:12.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"1376f6b2fab298e4442d1462b64a2e0031e0c459","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 10:01:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/f6d2f90704b14d59989c39a8f89e5802 : SUCCESS in 8m 44s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/2c5708311a214740b512c60529e9df8e : SUCCESS in 9m 05s (non-voting)\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/4006ebec332b4060b0582f22dc0401c7 : SUCCESS in 6m 06s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/bf2996d34f0546fa99465b653f19805f : SUCCESS in 10m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91c84eed001544378629f3b3907c19a7 : SUCCESS in 11m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/18fcee1f665d45d595b1bd28e5830df0 : SUCCESS in 11m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0882a0396f8a45c295d668390195ac04 : SUCCESS in 7m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df443bfd32354c779583366e01eb4402 : SUCCESS in 10m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bb9dfad2a464047815b05bc94bcf266 : SUCCESS in 9m 31s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/632f2913a63d43f583d1dccc3be3b9b2 : SUCCESS in 10m 37s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/d7e902244fbe4a3c9807a3cc870476d0 : SUCCESS in 1h 27m 07s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b873368d20f24e829d47a62e48a40e56 : SUCCESS in 6m 09s","accounts_in_message":[],"_revision_number":3},{"id":"59af06022bcdc51837ea585d568f6e5f5025d3e2","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-07-21 10:06:15.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"3e2c2a8d7f988b6a7037075f3e633608242aa50d","author":{"_account_id":28032,"name":"Oleksiy Petrenko","email":"opetrenko@mirantis.com","username":"enacero"},"date":"2020-07-21 13:26:52.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"83027660ad0c3ff6b3a5ed15995665c94c9d7753","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 15:04:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/3b6bb1fb02f942128c01d3fab5eb54d1 : SUCCESS in 6m 09s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/e9c8d3225fc54ca4a41dd750d68c9d2c : SUCCESS in 8m 13s (non-voting)\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/2345702bfc074ee789abf4c7c205b423 : SUCCESS in 6m 15s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/68fda59b7f164931926a46446c835db8 : SUCCESS in 7m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3985afd81ef14805a7fedba1a3c9d6d0 : SUCCESS in 7m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/660ad13a055843d59eb50f020fcac060 : SUCCESS in 11m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c53295a918d94c89bfca92be09538c5c : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70101cb492ab49b8a9d5849e3a6d0a48 : SUCCESS in 10m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a526bb8136094600ba52b780062688cb : SUCCESS in 8m 02s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/6f9fb1c13c884a5090955e3d4da45194 : SUCCESS in 8m 09s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/205b616a1bb44d27b521aa0e5d8afbd5 : SUCCESS in 1h 31m 26s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fe534f5c826c4951bd26cee8d67736b1 : SUCCESS in 7m 09s","accounts_in_message":[],"_revision_number":4},{"id":"98e7805ea9b721bd5d7c5e7bfeae9eb88e3e57ca","author":{"_account_id":28032,"name":"Oleksiy Petrenko","email":"opetrenko@mirantis.com","username":"enacero"},"date":"2020-08-05 09:42:33.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"4c7b23ea21009a95b39fbcf7513288af3d9ba008","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 11:24:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/d90c00f42a984ce4bf5a17e1ac71d34c : SUCCESS in 6m 15s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/7fc79d6e7a224a3d8cdc932983c1ad9f : SUCCESS in 7m 37s (non-voting)\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/a28335776b214397855c29d81bf6d2b0 : SUCCESS in 4m 36s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/48a59ee11c3b4fe1972bad9508d8e7e5 : SUCCESS in 7m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/10bf3df317ce44459dba746f4f362f81 : SUCCESS in 11m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7acecd9fa65448c48c6c85bcd2b8262a : SUCCESS in 12m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aac837a6ca44487eb2b545b34e5d0a9b : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3dabbf06d25d431caaf70f745f241962 : SUCCESS in 12m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91b3b6701aca43dd8a2ee700a76b2fc1 : SUCCESS in 7m 53s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/f8b7dd76f71c4055a7ecc1094da4e26d : SUCCESS in 11m 19s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/2c727546d4f64349ab87dbf5a0df8cd0 : SUCCESS in 1h 35m 44s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/75127b8a94e7465ea8be6615fdcf81c2 : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":5},{"id":"9bd58886d2a8741b42d293bffcda865d4bb6dc72","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-08-05 13:18:51.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)\n\nLGTM.","accounts_in_message":[],"_revision_number":5},{"id":"bfaf10f053c3bed06b20422e8faca7dfec96f7fc","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-08-29 22:02:00.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"7f588fcc9b84f402518b5c63f5d004fb33f0928b","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-09-08 20:03:53.000000000","message":"Patch Set 5: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"fb8599d07a6ae37253e75f20ce9a65a6de9dd1d2","author":{"_account_id":28032,"name":"Oleksiy Petrenko","email":"opetrenko@mirantis.com","username":"enacero"},"date":"2020-09-14 12:45:15.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"f97a979cc0537e4f5fee91bef8523556e32a0094","author":{"_account_id":28032,"name":"Oleksiy Petrenko","email":"opetrenko@mirantis.com","username":"enacero"},"date":"2020-09-14 12:48:32.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"22c97813dc9df854cba75371049650e305bb02c8","author":{"_account_id":28032,"name":"Oleksiy Petrenko","email":"opetrenko@mirantis.com","username":"enacero"},"date":"2020-09-14 12:50:38.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"318e36d0bcc40a67a6f8182aa9b7557ecec04a8c","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-09-14 13:14:17.000000000","message":"Patch Set 7: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"e7ff1f3de5d7baccf17eec1f8d7f351c03e43cc2","author":{"_account_id":28032,"name":"Oleksiy Petrenko","email":"opetrenko@mirantis.com","username":"enacero"},"date":"2020-09-14 13:38:54.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"7490a34c32eaaeba107b406161f2825ab69964c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 15:06:27.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/1836cf05b1d140d2b7b30736faf8e303 : SUCCESS in 7m 33s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/8efcae0538804b24a28fc2cc780a4c8f : SUCCESS in 7m 03s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/dd9b109583a243d9af11812040350feb : SUCCESS in 5m 24s (non-voting)\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/e2108375def449b0964d0eed921efa42 : SUCCESS in 6m 48s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/aac09ff416eb4492b7f8e4acfdd876ad : SUCCESS in 12m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6aa93071de7d4011977a90b0a6a04ff1 : SUCCESS in 13m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d98092326e7645349753695c12ee9829 : SUCCESS in 11m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1c8d8e74ae94af9a00ea6e440f2dd68 : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42bd10f917424394a919e5b6660168c3 : SUCCESS in 10m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1350ea172adb4c4baf7703fe7617acf2 : SUCCESS in 19m 00s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/de271392137640f599b956a42e310a2a : SUCCESS in 8m 02s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/f55b6f495c8f4124b6ec2e4656866bc0 : SUCCESS in 1h 22m 10s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fb0d943464e745aca15ab8c5abb9b4a4 : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":8},{"id":"a9d3cac2a60ac9076a5f493712f60112ecba3bba","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-09-14 21:48:40.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"4b426b6a6c1f78c77ecc3ae57980da8a7aac80f0","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-09-14 21:50:32.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"4509452070fadffb85e8b1d306ccd98beddeafda","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-09-14 21:50:50.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"e4a9c9f2ae893f18f70b2f5f3fadf79a114c3dba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 23:32:59.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/b437361de01045948d0ebe3508a63fff : SUCCESS in 13m 50s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/ec7ccce3c0974766ab04436d80a0cb8c : SUCCESS in 13m 45s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/d08d519011e04f3795eecd5f201686d2 : SUCCESS in 4m 22s (non-voting)\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/44153c75c01841d3908d1c3bdb00c7dc : SUCCESS in 6m 04s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/29230b15fe4a40aa848b7b23fc4f3132 : SUCCESS in 13m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5adfcf4f9cb64d3c96de97fd9f511865 : SUCCESS in 9m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a8361b07e504c9cb0c7c2b61ae66fb7 : SUCCESS in 11m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c60ea9375fcb41cba4dd08c88cc74817 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/231e91cee30a4b8db44f1a3623ae5134 : SUCCESS in 9m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c24d53fe716487fb5b961ef3d2290e7 : TIMED_OUT in 30m 57s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/de9c7b88e9c94b239f6f3f5356efc7af : SUCCESS in 8m 30s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/d23500f4ef144f768ba390cd3e7ad021 : SUCCESS in 1h 40m 26s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c2d5257c9f7d4644a7840700528a9d6e : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":9},{"id":"34a5300277aea7799aec05733a79f61ae30ec73c","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-09-15 00:35:31.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"a38ace063ba4d38a612259d338bac05e66b2aa9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 02:21:33.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/e7625a85d8ce4f77a52d1df6c2cce746 : SUCCESS in 5m 17s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/43bc7e509cac4e079483e4abe9a05d9d : SUCCESS in 6m 52s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/52fa41b6f246446ca4ce5cf85707d010 : FAILURE in 2m 21s (non-voting)\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/1dd0abacc0e14683b6a48c61d4155af3 : SUCCESS in 4m 48s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/19906aa9e831409c9dba90af119a4073 : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0dca930b89a49ad943b30ad6084d754 : FAILURE in 2m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9499c9affb8142a7bcda0b01c8329d97 : SUCCESS in 9m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af511ae9531248d384c4f5e85a4dd818 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7bb90fd7e23943c18d7149990f296f36 : FAILURE in 2m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e782d8a089fa41a88c6db5041b6c5c97 : SUCCESS in 14m 42s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/6439492fe2b14d0fb0ec37ab9f163ffa : SUCCESS in 6m 37s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/dd220d085cd447fbae0fa8062e850e95 : SUCCESS in 1h 25m 06s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6359d6eda4644e47abe76c93aa784245 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":9},{"id":"499a9c211dce2a8f852a30473f85ca2dbb94e54a","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-09-15 08:11:46.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"89b9ab180216ca849a0007074fd350875c7324e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 09:18:57.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/c2098570a1c14f7198c3e928224598db : FAILURE in 3m 50s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/af43c1c4941e432b9eac0ca8330310bc : FAILURE in 6m 16s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/eba4eee0724f4954b872d4c6e434334f : SUCCESS in 4m 14s (non-voting)\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/a43a349af38148e8baa56869b66e845c : FAILURE in 4m 39s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/42f344c6bfe0448d884ad850433568e8 : FAILURE in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/09eb381b12cd41909c0df0ac9bacf20a : FAILURE in 4m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21bc30af5f3d4315a8d39216b36b5a7c : SUCCESS in 10m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9914e075b02474cab392577e8083bab : FAILURE in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0dea33cc0daa49ba833254e5593699c4 : FAILURE in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84930d7695cb49ffb889b04a8187f6b1 : FAILURE in 14m 34s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/67b194802631419c8daaabc983091534 : FAILURE in 6m 06s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/6e4f99904c234e64a84e2264dd1785c8 : RETRY_LIMIT in 17m 05s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ece01ed5e4d54ea3a5f4d454ae004a01 : FAILURE in 4m 53s","accounts_in_message":[],"_revision_number":9},{"id":"46d6c59013ce2b57564ec3ffb3c83cc5b67ca7ab","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-09-15 11:03:32.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"df41ecc02c9affab024b26da0c8667b6743e91fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 11:53:26.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/b6c7530a050548e2b21e0bebe3ffa9ab : FAILURE in 5m 08s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/52cd91ae25a2421cae83a1a81b61d901 : FAILURE in 7m 35s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/2b0bbc514a834849a58055be54c733e0 : SUCCESS in 3m 31s (non-voting)\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/dc9f640ad8444deea5d9c537f1e46f88 : FAILURE in 5m 00s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/60f1d5cda2ed4dbfaf7707b3fc196618 : SUCCESS in 7m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ee83fbab0344bafb57dd38f6f7ca7ee : FAILURE in 3m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b256ae70ddd405fbb7e24f3c0c6c549 : SUCCESS in 9m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9457bcdec1f244d180f0a57bd14e2fc0 : FAILURE in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5552bcbc3f1a4f6a941581d62cf08724 : SUCCESS in 9m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e29336ccdc34a2b8c532f7ae4365084 : FAILURE in 13m 14s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/0a8acc0d40004a13ae877145391686ce : FAILURE in 5m 04s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/ec06e79e87a94494be273e13ce4da981 : RETRY_LIMIT in 13m 24s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ff77a772ea1a48d9b069ea280954f07d : FAILURE in 3m 42s","accounts_in_message":[],"_revision_number":9},{"id":"88fedffeeaaa3432ab0ff2178ce89a5ff0627574","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-09-16 06:54:43.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"b290084c54a5a84521ac38131b0642a182bb4683","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 09:08:33.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/2cd3d13ae24f432e9d658b0f8d6bfa6c : SUCCESS in 9m 32s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/85a1e748e9054f2c9d409efa0868e2c6 : SUCCESS in 7m 27s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/beef30c6e646479a8738f4f3d56eb5da : SUCCESS in 5m 47s (non-voting)\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/afd433b8540b4638a2f10b3643c3648e : SUCCESS in 6m 45s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/399489926e084f9cad41b5718b78c951 : SUCCESS in 15m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2752d483423041ad8eb92ea068d595ba : SUCCESS in 10m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d879d7df801242e69d2a0c6996bedfbc : SUCCESS in 10m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/564d6c9839194a87aba0f5d904d94351 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dddefedf19b6488c91a25f623598a862 : SUCCESS in 11m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc67e08539e149dcac94ce9e156e65da : SUCCESS in 18m 21s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/7eba0410eb474e029480b033b32b6bb1 : SUCCESS in 11m 38s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/586e0425d3d14976b687982b16c2b3e8 : SUCCESS in 1h 30m 54s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7a848bdf8b184df293bbaeca0174dd6f : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":9},{"id":"e7f8c57801893df0243684a0eca7e01493cfdfd1","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-09-16 14:55:07.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"e5d4c397c896d046980e83fe6cd62edec76c060b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 14:55:21.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"bca869e40eb0ce5ad293cfe9e2f54512ed5069dc","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-09-16 14:57:52.000000000","message":"Patch Set 9: -Code-Review -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"bee2701771bc7cd21c41012946299ea77c9967fd","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-09-16 14:58:19.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"2441afeab3e3fde4eccf5a833fb3de5028be1149","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 16:35:11.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/0fa89e972bca4fc785d5e9add1d99042 : SUCCESS in 6m 53s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/16108128628541918ab6145477813502 : SUCCESS in 8m 56s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/7058511f55be4aea95a8dd2bd735994c : SUCCESS in 4m 31s (non-voting)\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/77f30c973b7545228e7e45ff54534c9e : SUCCESS in 7m 25s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/a9e7a12dfabb4684ba9ea47aff708b95 : SUCCESS in 9m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38db819c7d6a4d87865019bd9887ad5d : SUCCESS in 12m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a34ce2c63c6477c88222f44a08914df : SUCCESS in 11m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/193dc90647be4b1681eab017dc167945 : SUCCESS in 7m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b9d2f4f640f4ada9e55a956decc7dc3 : SUCCESS in 11m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed0316ed21ca4b2b96d3520eb2dc30c2 : SUCCESS in 17m 56s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/d068f2688c894e558137827773e5efa4 : SUCCESS in 7m 38s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/7aee02f053a145bdb5664185733e34b0 : SUCCESS in 1h 26m 48s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/03e22d4cee2442e9b9d3f224742ac159 : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":10},{"id":"93ddcd4474aaf1377854d32262aae7f4d22c7c05","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-09-17 07:33:09.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"d88ce2675f61c9ea64cfdaa55fd6ef6243fb9486","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 07:33:22.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"ee4218a13d837b759d70d3c7bb59e359bf3a9a6a","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-09-17 07:33:55.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":10},{"id":"6019ade4b2edd5fd74fd20a49f3a4add3fc23efe","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-09-17 07:34:02.000000000","message":"Restored","accounts_in_message":[],"_revision_number":10},{"id":"949ec5017eeb2e570c1aaa72a0b332a2d022db0e","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-09-17 07:34:06.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"e0dda53c342acc5f1054d64f1dd7a8998599a86d","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-09-17 08:24:35.000000000","message":"Patch Set 11: -Code-Review","accounts_in_message":[],"_revision_number":11},{"id":"f4e55210f628faec099e8e3d544663b0c0ee2e34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 09:29:00.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/4d42a2505a13446aa89b7db175dd55db : SUCCESS in 9m 57s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/35e5ba01acfd4045a6f7feba746e4153 : SUCCESS in 9m 14s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/c56cd8c0165c4a8d853d3a3d07ed5a18 : SUCCESS in 4m 53s (non-voting)\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/5aaf6f2b4f394927842df9921d35e0fb : SUCCESS in 7m 01s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/bddd4f23fec543f3bb3efdf5531f3a22 : SUCCESS in 9m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da9486b1fa5d4259beabfdee9f7972f7 : SUCCESS in 12m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a1bd29e66ee14663ac370223bfce228d : SUCCESS in 11m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc595a85a8274bad896d0a1947f9dfcc : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/471c0f1ddd8345e6bfc909c9b8398780 : SUCCESS in 13m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f908f880aaa4cbaaa1480d2d0f54c81 : SUCCESS in 24m 18s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/2347ad9c0fe44e529a232735dbfb6c10 : SUCCESS in 9m 49s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/da1a4a0e71f846d8aaf98b4979ae7538 : SUCCESS in 1h 39m 13s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5da0ac2337eb4b7ca38372afab62c11f : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":11},{"id":"6d927883d1a3e0dfd158c786e538e01141028800","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-09-17 13:58:04.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"3909cd2401f3218a857f52cdb44777014c079e86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 15:38:24.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/878acb082d604f69a59ed9a0b18d7ba5 : SUCCESS in 7m 00s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/9a7088ae2efc4c7c89099cfb87daa2c0 : SUCCESS in 8m 30s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/12cb8578a2b84a5197bbc16edbfac719 : SUCCESS in 5m 19s (non-voting)\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/5395f318b22a455b9020b84ff68f8430 : SUCCESS in 5m 38s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/0420c72872d249feadaa9f24f00c8732 : SUCCESS in 10m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92115a87b9a344f8a265846bb5d840a2 : SUCCESS in 11m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f83ce8208d4438582753f94f168d10f : SUCCESS in 11m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e38ebe0fa024c75b4c20af2208e7ddc : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4816e6cefd044e08a8c91c623cb072d : SUCCESS in 11m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ac0fe04d3bc4460b3e9d23b8ee8d0ac : SUCCESS in 15m 39s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/9f0633af941c479ca394774de21426d7 : SUCCESS in 8m 01s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/881194cc02ae46a0839c6b42e0853472 : SUCCESS in 1h 28m 55s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c09c84c156984a80b321b49075b27d63 : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":12},{"id":"77458f0d36ab8f5cc8de38f837f838cdf53c90a6","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-09-18 02:00:02.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"d6717355faca423ff70c9999775e18c4c021c592","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-09-18 10:28:14.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"6a1e908d36e5a5dfb4ddea9c89e049e82dc06cfc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 10:28:33.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"1b577700f2ea17b50b39e27f4c5ec708e68ed04f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 12:08:17.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/ce102f0df83d45a59461940f1d72e142 : SUCCESS in 6m 17s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/f7cc6b7a331346babb5b297202f82629 : SUCCESS in 8m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73326d20f92e4935bd17fafda4d2c06e : SUCCESS in 9m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/38f1756080ed4631b91226ae450348f3 : SUCCESS in 10m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/739681e7c23249158dcb567dec4114be : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80d65c42dcd34652a3a8af7d329356cf : SUCCESS in 9m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66ae1eafb3494393a626e1e5cd93e5bc : SUCCESS in 12m 59s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/1db195039523457cb1937fc367e15d97 : SUCCESS in 6m 33s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/b3c6e37f070a46bd82ae2264fd54c85f : SUCCESS in 1h 32m 08s","accounts_in_message":[],"_revision_number":12},{"id":"993b1692ee2dd75b632df90253ea6b503822eb5f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 12:08:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12},{"id":"6482b5c0286c81053d538ea4f6b9893066d64be3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 12:10:19.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acf0537b0bba4a468a5e1108a98d08fe : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"97797197af1ebbee6d3ec3e1ea6dc1044aef6516","revisions":{"31e3e392ee67e357a2d0e83a861a56f129a2c534":{"kind":"REWORK","_number":1,"created":"2020-07-20 17:01:58.000000000","uploader":{"_account_id":28032,"name":"Oleksiy Petrenko","email":"opetrenko@mirantis.com","username":"enacero"},"ref":"refs/changes/90/741990/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/90/741990/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/90/741990/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/90/741990/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/90/741990/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/90/741990/1"}}},"commit":{"parents":[{"commit":"f04ffda41776cc02c255bde87da73794bec56f6b","subject":"Fix instance integration tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/f04ffda41776cc02c255bde87da73794bec56f6b"}]}],"author":{"name":"Oleksii Petrenko","email":"opetrenko@mirantis.com","date":"2020-07-20 16:57:18.000000000","tz":180},"committer":{"name":"Oleksii Petrenko","email":"opetrenko@mirantis.com","date":"2020-07-20 17:01:30.000000000","tz":180},"subject":"Adds flatten_volume_from_snapshot config option","message":"Adds flatten_volume_from_snapshot config option\n\nAdds flatten_volume_from_snapshot option, so that colume snapshots can\nbe run on environment with ceph\n\nChange-Id: Ib422331892f077d78e3f2efbdd88abafc4c52b9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/31e3e392ee67e357a2d0e83a861a56f129a2c534"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/31e3e392ee67e357a2d0e83a861a56f129a2c534"}]},"branch":"refs/heads/master"},"8badc718d0d66b3a57657165d76ce6b0dfdc12d9":{"kind":"REWORK","_number":2,"created":"2020-07-20 17:11:15.000000000","uploader":{"_account_id":28032,"name":"Oleksiy Petrenko","email":"opetrenko@mirantis.com","username":"enacero"},"ref":"refs/changes/90/741990/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/90/741990/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/90/741990/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/90/741990/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/90/741990/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/90/741990/2"}}},"commit":{"parents":[{"commit":"f04ffda41776cc02c255bde87da73794bec56f6b","subject":"Fix instance integration tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/f04ffda41776cc02c255bde87da73794bec56f6b"}]}],"author":{"name":"Oleksii Petrenko","email":"opetrenko@mirantis.com","date":"2020-07-20 16:57:18.000000000","tz":180},"committer":{"name":"Oleksii Petrenko","email":"opetrenko@mirantis.com","date":"2020-07-20 17:10:53.000000000","tz":180},"subject":"Adds flatten_volume_from_snapshot config option","message":"Adds flatten_volume_from_snapshot config option\n\nAdds flatten_volume_from_snapshot option, so that colume snapshots can\nbe run on environment with ceph\n\nChange-Id: Ib422331892f077d78e3f2efbdd88abafc4c52b9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/8badc718d0d66b3a57657165d76ce6b0dfdc12d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/8badc718d0d66b3a57657165d76ce6b0dfdc12d9"}]},"branch":"refs/heads/master"},"7241804d7ddd5346742219c416047075edcc1b0e":{"kind":"REWORK","_number":3,"created":"2020-07-21 08:25:40.000000000","uploader":{"_account_id":28032,"name":"Oleksiy Petrenko","email":"opetrenko@mirantis.com","username":"enacero"},"ref":"refs/changes/90/741990/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/90/741990/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/90/741990/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/90/741990/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/90/741990/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/90/741990/3"}}},"commit":{"parents":[{"commit":"f04ffda41776cc02c255bde87da73794bec56f6b","subject":"Fix instance integration tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/f04ffda41776cc02c255bde87da73794bec56f6b"}]}],"author":{"name":"Oleksii Petrenko","email":"opetrenko@mirantis.com","date":"2020-07-20 16:57:18.000000000","tz":180},"committer":{"name":"Oleksii Petrenko","email":"opetrenko@mirantis.com","date":"2020-07-21 08:24:34.000000000","tz":180},"subject":"Adds flatten_volume_from_snapshot config option","message":"Adds flatten_volume_from_snapshot config option\n\nAdds flatten_volume_from_snapshot option, so that colume snapshots can\nbe run on environment with ceph Ceph when you can\u0027t delete a parent\nsnapshot for volume. For more\ndetails, please see Cinder documentation [1].\n\n[1] https://docs.openstack.org/cinder/ussuri/drivers.html#rbddriver\n\nChange-Id: Ib422331892f077d78e3f2efbdd88abafc4c52b9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/7241804d7ddd5346742219c416047075edcc1b0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/7241804d7ddd5346742219c416047075edcc1b0e"}]},"branch":"refs/heads/master"},"0ac6e046289e71d39a66304549d1b1f457da7f80":{"kind":"REWORK","_number":4,"created":"2020-07-21 13:26:52.000000000","uploader":{"_account_id":28032,"name":"Oleksiy Petrenko","email":"opetrenko@mirantis.com","username":"enacero"},"ref":"refs/changes/90/741990/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/90/741990/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/90/741990/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/90/741990/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/90/741990/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/90/741990/4"}}},"commit":{"parents":[{"commit":"f04ffda41776cc02c255bde87da73794bec56f6b","subject":"Fix instance integration tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/f04ffda41776cc02c255bde87da73794bec56f6b"}]}],"author":{"name":"Oleksii Petrenko","email":"opetrenko@mirantis.com","date":"2020-07-20 16:57:18.000000000","tz":180},"committer":{"name":"Oleksiy Petrenko","email":"opetrenko@mirantis.com","date":"2020-07-21 13:26:49.000000000","tz":0},"subject":"Adds flatten_volume_from_snapshot config option","message":"Adds flatten_volume_from_snapshot config option\n\nAdds flatten_volume_from_snapshot option, so that colume snapshots can\nbe run on environment with ceph Ceph when you can\u0027t delete a parent\nsnapshot for volume. For more\ndetails, please see Cinder documentation [1].\n\n[1] https://docs.openstack.org/cinder/ussuri/drivers.html#rbddriver\n\nChange-Id: Ib422331892f077d78e3f2efbdd88abafc4c52b9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/0ac6e046289e71d39a66304549d1b1f457da7f80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/0ac6e046289e71d39a66304549d1b1f457da7f80"}]},"branch":"refs/heads/master"},"416a687f45bd1cdfdfc01d317b8554e5aca622ef":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2020-08-05 09:42:33.000000000","uploader":{"_account_id":28032,"name":"Oleksiy Petrenko","email":"opetrenko@mirantis.com","username":"enacero"},"ref":"refs/changes/90/741990/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/90/741990/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/90/741990/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/90/741990/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/90/741990/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/90/741990/5"}}},"commit":{"parents":[{"commit":"f04ffda41776cc02c255bde87da73794bec56f6b","subject":"Fix instance integration tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/f04ffda41776cc02c255bde87da73794bec56f6b"}]}],"author":{"name":"Oleksii Petrenko","email":"opetrenko@mirantis.com","date":"2020-07-20 16:57:18.000000000","tz":180},"committer":{"name":"Oleksiy Petrenko","email":"opetrenko@mirantis.com","date":"2020-08-05 09:42:30.000000000","tz":0},"subject":"Add flatten_volume_from_snapshot config option","message":"Add flatten_volume_from_snapshot config option\n\nAdds flatten_volume_from_snapshot config option, that controls whether\nto test only situation where first test deletes volume and then snapshot\nor also second situation, where first test deletes snapshot and then \nvolume. Second situation is impossible with Ceph [1].\n\n[1] https://docs.openstack.org/cinder/ussuri/drivers.html#rbddriver\n\nChange-Id: Ib422331892f077d78e3f2efbdd88abafc4c52b9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/416a687f45bd1cdfdfc01d317b8554e5aca622ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/416a687f45bd1cdfdfc01d317b8554e5aca622ef"}]},"branch":"refs/heads/master"},"72e45ea95960a215826b3e3a74acde9b23e24935":{"kind":"REWORK","_number":6,"created":"2020-09-14 12:48:32.000000000","uploader":{"_account_id":28032,"name":"Oleksiy Petrenko","email":"opetrenko@mirantis.com","username":"enacero"},"ref":"refs/changes/90/741990/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/90/741990/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/90/741990/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/90/741990/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/90/741990/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/90/741990/6"}}},"commit":{"parents":[{"commit":"f8e5fc3f9ea96708fba1b3bd1d53c5f5a988444e","subject":"Merge \"doc: Update default value of REST_API_REQUIRED_SETTINGS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/f8e5fc3f9ea96708fba1b3bd1d53c5f5a988444e"}]}],"author":{"name":"Oleksii Petrenko","email":"opetrenko@mirantis.com","date":"2020-07-20 16:57:18.000000000","tz":180},"committer":{"name":"Oleksii Petrenko","email":"opetrenko@mirantis.com","date":"2020-09-14 12:47:09.000000000","tz":180},"subject":"Add allow_delete_snapshot_before_volume config option","message":"Add allow_delete_snapshot_before_volume config option\n\nAdds allow_delete_snapshot_before_volume config option, that controls whether\nto test only situation where first test deletes volume and then snapshot\nor also second situation, where first test deletes snapshot and then\nvolume. Second situation is impossible with Ceph [1].\n\n[1] https://docs.openstack.org/cinder/ussuri/drivers.html#rbddriver\n\nChange-Id: Ib422331892f077d78e3f2efbdd88abafc4c52b9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/72e45ea95960a215826b3e3a74acde9b23e24935"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/72e45ea95960a215826b3e3a74acde9b23e24935"}]},"branch":"refs/heads/master"},"b7978b316aab6012404e3505d0766c7f8d4efcdd":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2020-09-14 12:50:38.000000000","uploader":{"_account_id":28032,"name":"Oleksiy Petrenko","email":"opetrenko@mirantis.com","username":"enacero"},"ref":"refs/changes/90/741990/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/90/741990/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/90/741990/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/90/741990/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/90/741990/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/90/741990/7"}}},"commit":{"parents":[{"commit":"f8e5fc3f9ea96708fba1b3bd1d53c5f5a988444e","subject":"Merge \"doc: Update default value of REST_API_REQUIRED_SETTINGS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/f8e5fc3f9ea96708fba1b3bd1d53c5f5a988444e"}]}],"author":{"name":"Oleksii Petrenko","email":"opetrenko@mirantis.com","date":"2020-07-20 16:57:18.000000000","tz":180},"committer":{"name":"Oleksii Petrenko","email":"opetrenko@mirantis.com","date":"2020-09-14 12:49:52.000000000","tz":180},"subject":"Add allow_delete_snapshot_before_volume config option","message":"Add allow_delete_snapshot_before_volume config option\n\nAdds allow_delete_snapshot_before_volume config option, that controls\nwhether to test only situation where first test deletes volume and\nthen snapshot or also second situation, where first test deletes\nsnapshot and then volume. Second situation is impossible with Ceph [1].\n\n[1] https://docs.openstack.org/cinder/ussuri/drivers.html#rbddriver\n\nChange-Id: Ib422331892f077d78e3f2efbdd88abafc4c52b9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/b7978b316aab6012404e3505d0766c7f8d4efcdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/b7978b316aab6012404e3505d0766c7f8d4efcdd"}]},"branch":"refs/heads/master"},"185580179a28576e358dd2044b02f3baaf34957e":{"kind":"REWORK","_number":8,"created":"2020-09-14 13:38:54.000000000","uploader":{"_account_id":28032,"name":"Oleksiy Petrenko","email":"opetrenko@mirantis.com","username":"enacero"},"ref":"refs/changes/90/741990/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/90/741990/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/90/741990/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/90/741990/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/90/741990/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/90/741990/8"}}},"commit":{"parents":[{"commit":"f8e5fc3f9ea96708fba1b3bd1d53c5f5a988444e","subject":"Merge \"doc: Update default value of REST_API_REQUIRED_SETTINGS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/f8e5fc3f9ea96708fba1b3bd1d53c5f5a988444e"}]}],"author":{"name":"Oleksii Petrenko","email":"opetrenko@mirantis.com","date":"2020-07-20 16:57:18.000000000","tz":180},"committer":{"name":"Oleksii Petrenko","email":"opetrenko@mirantis.com","date":"2020-09-14 13:38:28.000000000","tz":180},"subject":"Add allow_delete_snapshot_before_volume config option","message":"Add allow_delete_snapshot_before_volume config option\n\nAdds allow_delete_snapshot_before_volume config option, that controls\nwhether to test only situation where first test deletes volume and\nthen snapshot or also second situation, where first test deletes\nsnapshot and then volume. Second situation is impossible with Ceph [1].\n\n[1] https://docs.openstack.org/cinder/ussuri/drivers.html#rbddriver\n\nChange-Id: Ib422331892f077d78e3f2efbdd88abafc4c52b9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/185580179a28576e358dd2044b02f3baaf34957e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/185580179a28576e358dd2044b02f3baaf34957e"}]},"branch":"refs/heads/master"},"a8783d2c9caac9c699d8bd3051497d0c44c1a03d":{"kind":"REWORK","_number":9,"created":"2020-09-14 21:50:32.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/90/741990/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/90/741990/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/90/741990/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/90/741990/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/90/741990/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/90/741990/9"}}},"commit":{"parents":[{"commit":"f8e5fc3f9ea96708fba1b3bd1d53c5f5a988444e","subject":"Merge \"doc: Update default value of REST_API_REQUIRED_SETTINGS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/f8e5fc3f9ea96708fba1b3bd1d53c5f5a988444e"}]}],"author":{"name":"Oleksii Petrenko","email":"opetrenko@mirantis.com","date":"2020-07-20 16:57:18.000000000","tz":180},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-09-14 21:50:27.000000000","tz":0},"subject":"Add allow_delete_snapshot_before_volume config option","message":"Add allow_delete_snapshot_before_volume config option\n\nAdds allow_delete_snapshot_before_volume config option, that controls\nwhether to test only situation where first test deletes volume and\nthen snapshot or also second situation, where first test deletes\nsnapshot and then volume. Second situation is impossible with Ceph [1].\n\n[1] https://docs.openstack.org/cinder/ussuri/drivers.html#rbddriver\n\nChange-Id: Ib422331892f077d78e3f2efbdd88abafc4c52b9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/a8783d2c9caac9c699d8bd3051497d0c44c1a03d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/a8783d2c9caac9c699d8bd3051497d0c44c1a03d"}]},"branch":"refs/heads/master"},"b3dfd93a122fb92a76e73e335ad26bfd87cb8f03":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2020-09-16 14:58:19.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/90/741990/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/90/741990/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/90/741990/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/90/741990/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/90/741990/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/90/741990/10"}}},"commit":{"parents":[{"commit":"5056b9bbe652258c5584319d089afb8afbe629a9","subject":"Merge \"Make getPorts and setBreadcrumb tests work\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/5056b9bbe652258c5584319d089afb8afbe629a9"}]}],"author":{"name":"Oleksii Petrenko","email":"opetrenko@mirantis.com","date":"2020-07-20 16:57:18.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-09-16 14:58:19.000000000","tz":0},"subject":"Add allow_delete_snapshot_before_volume config option","message":"Add allow_delete_snapshot_before_volume config option\n\nAdds allow_delete_snapshot_before_volume config option, that controls\nwhether to test only situation where first test deletes volume and\nthen snapshot or also second situation, where first test deletes\nsnapshot and then volume. Second situation is impossible with Ceph [1].\n\n[1] https://docs.openstack.org/cinder/ussuri/drivers.html#rbddriver\n\nChange-Id: Ib422331892f077d78e3f2efbdd88abafc4c52b9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/b3dfd93a122fb92a76e73e335ad26bfd87cb8f03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/b3dfd93a122fb92a76e73e335ad26bfd87cb8f03"}]},"branch":"refs/heads/master"},"8ce3f38e65f349e246326c7dc2440c8d52e28e64":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2020-09-17 07:34:06.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/90/741990/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/90/741990/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/90/741990/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/90/741990/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/90/741990/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/90/741990/11"}}},"commit":{"parents":[{"commit":"d9bf46371eb411f227ef4bbfd9e002ba7e5d2a4b","subject":"Merge \"Refactor error messages.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/d9bf46371eb411f227ef4bbfd9e002ba7e5d2a4b"}]}],"author":{"name":"Oleksii Petrenko","email":"opetrenko@mirantis.com","date":"2020-07-20 16:57:18.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-09-17 07:34:06.000000000","tz":0},"subject":"Add allow_delete_snapshot_before_volume config option","message":"Add allow_delete_snapshot_before_volume config option\n\nAdds allow_delete_snapshot_before_volume config option, that controls\nwhether to test only situation where first test deletes volume and\nthen snapshot or also second situation, where first test deletes\nsnapshot and then volume. Second situation is impossible with Ceph [1].\n\n[1] https://docs.openstack.org/cinder/ussuri/drivers.html#rbddriver\n\nChange-Id: Ib422331892f077d78e3f2efbdd88abafc4c52b9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/8ce3f38e65f349e246326c7dc2440c8d52e28e64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/8ce3f38e65f349e246326c7dc2440c8d52e28e64"}]},"branch":"refs/heads/master"},"97797197af1ebbee6d3ec3e1ea6dc1044aef6516":{"kind":"REWORK","_number":12,"created":"2020-09-17 13:58:04.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/90/741990/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/90/741990/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/90/741990/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/90/741990/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/90/741990/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/90/741990/12"}}},"commit":{"parents":[{"commit":"d9bf46371eb411f227ef4bbfd9e002ba7e5d2a4b","subject":"Merge \"Refactor error messages.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/d9bf46371eb411f227ef4bbfd9e002ba7e5d2a4b"}]}],"author":{"name":"Oleksii Petrenko","email":"opetrenko@mirantis.com","date":"2020-07-20 16:57:18.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-09-17 13:57:26.000000000","tz":180},"subject":"Add allow_delete_snapshot_before_volume config option","message":"Add allow_delete_snapshot_before_volume config option\n\nAdds allow_delete_snapshot_before_volume config option, that controls\nwhether to test only situation where first test deletes volume and\nthen snapshot or also second situation, where first test deletes\nsnapshot and then volume. Second situation is impossible with Ceph [1].\n\n[1] https://docs.openstack.org/cinder/ussuri/drivers.html#rbddriver\n\nCo-Authored-By: Ivan Kolodyazhny \u003ce0ne@e0ne.info\u003e\nChange-Id: Ib422331892f077d78e3f2efbdd88abafc4c52b9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/97797197af1ebbee6d3ec3e1ea6dc1044aef6516"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/97797197af1ebbee6d3ec3e1ea6dc1044aef6516"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
