)]}'
{"id":"openstack%2Fhorizon~749593","triplet_id":"openstack%2Fhorizon~stable%2Ftrain~Ieb2fe4d2c4453e572c9251dc03c34d08bed62a0f","project":"openstack/horizon","branch":"stable/train","topic":"periodic-jobs-stable/train","hashtags":[],"change_id":"Ieb2fe4d2c4453e572c9251dc03c34d08bed62a0f","subject":"Define periodic jobs","status":"MERGED","created":"2020-09-02 20:52:39.000000000","updated":"2020-12-31 14:25:20.000000000","submitted":"2020-12-31 14:22:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749593-periodic-jobs-stable/train","meta_rev_id":"3a9391ff0a51836d9d3c31e79acf9d72ea75c73e","_number":749593,"virtual_id_number":749593,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-31 14:22:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"recommended":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":1,"date":"2020-09-22 11:57:56.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2020-10-02 20:34:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-12-23 01:16:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":1,"date":"2020-12-23 00:55:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-14 11:27:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-22 11:57:56.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2020-10-02 20:34:29.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-12-23 01:16:05.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"4ff158805fd5da94738914ff408f37cdfe21222b","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-09-02 20:52:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f318be922f59c7a631034ee98b7422ebe32d82f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 22:24:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/70d619f296814a849a27f8790f34e50e : SUCCESS in 5m 53s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/02e7c2a0f18d47d7a49cff4f28893972 : SUCCESS in 7m 18s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/493916d0bd3447fba109efea7621c899 : SUCCESS in 9m 14s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/ea5734e27c5a412384cae2b0b6441592 : SUCCESS in 13m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/33b4cde8925c44e49e0d7ee6f5dad69d : SUCCESS in 10m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/709ac6e61b9c483ca86612537fb65f99 : SUCCESS in 5m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3b3086431f344677931b4d65d9112214 : SUCCESS in 10m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/766e108c8c1349a88a78b964746223c2 : SUCCESS in 9m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ceee4a48552649bda86d3608914c864c : SUCCESS in 9m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b991cd22a0d4adeba5f92418079e8ee : SUCCESS in 6m 47s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/739fd11ed6b54160b85994814b8f49de : SUCCESS in 6m 26s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/e16435a390b648efb1bdda80fc03be81 : FAILURE in 1h 20m 45s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/fb2fc878009a4904a2d453f036f73115 : SUCCESS in 40m 27s\n- horizon-dsvm-tempest-plugin-py27 https://zuul.opendev.org/t/openstack/build/3c5f9b1a3c524a998ab4da02a42338f5 : SUCCESS in 38m 33s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7006adb2d5b2489c8ff46e2b08add4c1 : SUCCESS in 4m 45s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/1254c0757a984d2bb5b2d78a6e0e83a2 : SUCCESS in 36m 41s","accounts_in_message":[],"_revision_number":1},{"id":"7973348200eb2cce6c9e8bb97b1ed72b47379ac2","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-09-14 10:08:46.000000000","message":"Patch Set 1:\n\nrecheck.","accounts_in_message":[],"_revision_number":1},{"id":"df0aec452536c2d9221c528cee6428592e36e9d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 11:27:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/1cf9269482d74e1089be9b6ca1115e71 : SUCCESS in 5m 17s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/2bd9dfe934d940088ac01e0c34aff6f6 : SUCCESS in 6m 48s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/bf6b4672933f4c82bc9efc1139fdb93f : SUCCESS in 8m 04s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/6322e31f2f754a218a89e00b0ec503bf : SUCCESS in 9m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/94b2c671411d4f878cdbab9cdbf4d0e0 : SUCCESS in 8m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/559933bb7c174b3c8340b2e567aabcef : SUCCESS in 7m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a355c87adadc432ab1548a887df4e7f9 : SUCCESS in 8m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbe6ffb3809b451aa6ef6be5ee716132 : SUCCESS in 9m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ac4d9b17b4774316bec3fa75f1dd7474 : SUCCESS in 10m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52e2d7808a11467fb5335a49e070ef7b : SUCCESS in 4m 57s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/8aafce5364dd42b887af22bb825d5643 : SUCCESS in 6m 31s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/daf1b33773354a449f6581f6e6968304 : SUCCESS in 1h 11m 23s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/a94f11f5294b4ea4a26ef47718b60a80 : SUCCESS in 24m 48s\n- horizon-dsvm-tempest-plugin-py27 https://zuul.opendev.org/t/openstack/build/003f87b1cf5e4bc98cd782dd10a68624 : SUCCESS in 43m 29s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/204c04cdc2084c939e4c479847c74c24 : SUCCESS in 4m 54s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/a5823acdaea74a659cde9c575d210e62 : SUCCESS in 41m 22s","accounts_in_message":[],"_revision_number":1},{"id":"04c24c3f3c23747177443b837d6639c6d36385d8","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-09-22 11:57:56.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"517a9d1c79b6cbf9797ce5d78b0f78492868b6df","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-10-02 20:34:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"44be99c2e09c26f80bf572ea10616840e7ecfd16","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-12-23 00:55:04.000000000","message":"Patch Set 1: Workflow+1\n\nLet me merge this. It does not affect the actual behavior of horizon.","accounts_in_message":[],"_revision_number":1},{"id":"2dcd199b34dc6bca7cbe45d024424781bc12a42d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 00:55:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"73eaa49b1784606ff081e9f210877cbf9532add6","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-12-23 01:16:05.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"db6019ee77595ef735250dbf644e91d769685ca0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 03:04:35.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/d6ae1eb0d55e4016ab7e8bd941a0e8b3 : SUCCESS in 3m 57s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/b634f0a579b342b2b6cda7ce3be96c27 : SUCCESS in 6m 00s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/5eb3fa9cf1c34220be321383e3585926 : SUCCESS in 7m 51s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/77cd9dd9694a44ac974813829713376f : SUCCESS in 8m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4da1b065468148398937541ffaef3b9c : FAILURE in 2m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ebd30c1585247c3bc9819dae8e0c4d9 : SUCCESS in 5m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dc30c8693a4a4dbeb75c1e285973361d : FAILURE in 3m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27b19e48caa44b9cb362e3856d602cb6 : SUCCESS in 7m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2f253812239b4ece903170a38ff5aadf : SUCCESS in 6m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/007b930d09804f49bc46196805b045b8 : SUCCESS in 6m 16s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/05c9bbd434bb4da4a88d8d415aadf85d : SUCCESS in 33m 26s\n- horizon-dsvm-tempest-plugin-py27 https://zuul.opendev.org/t/openstack/build/2520d0fb924a4116aec672a7bee6dea4 : SUCCESS in 27m 48s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/e062ab0f22ec4c3097adbd5daeadfab6 : SUCCESS in 5m 50s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/7191ad11f8ea41c6a9d036a9c9d001dc : SUCCESS in 50m 04s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/1743f833f1a04974adb2cccabcef6416 : SUCCESS in 25m 04s","accounts_in_message":[],"_revision_number":1},{"id":"4b866f66ab262abf97b00e5427d92992f10d43f3","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-12-29 09:46:15.000000000","message":"Patch Set 1:\n\nrecheck.","accounts_in_message":[],"_revision_number":1},{"id":"55f18e3c7290923e1927d5bef899673eb8649320","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-29 10:45:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/e5f6cd334ecb4b34b38809f79d00686b : SUCCESS in 4m 19s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/7cc3b7b49e104cb5a85767f248edca3c : SUCCESS in 6m 55s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/341892d2b7f2492b936759aa210ad024 : SUCCESS in 6m 59s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/1792769e238245e5984f5215203e98ff : SUCCESS in 7m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f646fe5258c548d8a5fa9f1125592ca7 : SUCCESS in 9m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/290f00ebba93439b9e9d6605a56dca9b : SUCCESS in 5m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bee4acf00ca84d51ae573e0c90382739 : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dab24e80d22c4345816d3c941bff7661 : SUCCESS in 6m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fb7c1020754a4cdfbbce9d4bd2136925 : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1473f224731e4eca976f5155514d17e0 : SUCCESS in 7m 50s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/1069002e50a149759caa09340b134ae1 : SUCCESS in 6m 42s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/51c1923d6ab34b499e9a276fd4f7ac05 : FAILURE in 55m 25s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/7a674e2eeeef45ee993ba3dedbd4560a : SUCCESS in 57m 31s\n- horizon-dsvm-tempest-plugin-py27 https://zuul.opendev.org/t/openstack/build/b40e5fa2c6334dd5ac7791ec553d57e5 : SUCCESS in 26m 21s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/988f3d2f5fb242c1bc387fa9149b5ba9 : SUCCESS in 6m 00s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/cdcd9f52caae4ce3b3cfc018ba785330 : SUCCESS in 56m 38s","accounts_in_message":[],"_revision_number":1},{"id":"92d151e8e2107f249765b338775b314a2e99695f","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-12-31 12:25:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e7cb9629f0fb0550a42a35c0c394dab2ad300805","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-31 13:30:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/3799fe55b5d64b5dbb98cf9b1e0b7fa7 : SUCCESS in 4m 04s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/110f9a9972c249fc807ebb8abc42e2e4 : SUCCESS in 6m 59s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/59603c64647b453bbb772f2c1ae264dd : SUCCESS in 6m 13s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/af95827c6e284d3bb8e0808a23e28a1a : SUCCESS in 6m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a47ffbcef9f4beb950dc653885efdd7 : SUCCESS in 7m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2527f4c934af49b484098d10ddbc681d : SUCCESS in 4m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bd5a041837514d8d8a9c82bc787e5967 : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4dea76ea55054cb6a1dbc03b18be5093 : SUCCESS in 6m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0a160948dc354c47b91723cab785fa87 : SUCCESS in 8m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef6832492db848198edc31a20b0c22ed : SUCCESS in 6m 15s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/7345e59caa4a4725bc5a054ad9ecaa07 : SUCCESS in 6m 15s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/095897cc4e2247518c8b8ce100d245e7 : SUCCESS in 59m 39s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/84f895a2fc92495296376f08b27de6f1 : SUCCESS in 24m 58s\n- horizon-dsvm-tempest-plugin-py27 https://zuul.opendev.org/t/openstack/build/648f76117e4d470d8cb648357b4aa8e2 : SUCCESS in 33m 00s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e83eb3ebafa04386b84a40877c2144c8 : SUCCESS in 3m 42s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/dca49dd73d8f4c9fb6396ac97bb0c8c6 : SUCCESS in 34m 22s","accounts_in_message":[],"_revision_number":1},{"id":"3e8f116c2b9278b45f5a3b6058a8e30e66c36385","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-31 13:30:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8ad6f1a3a01d091ae20bb0d9dfbc1fa425ab57f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-31 14:22:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/5b2a45d04a014650b404f2f419acf9c2 : SUCCESS in 4m 05s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/efa534b5a99b4fcd90983db229af9c5a : SUCCESS in 5m 23s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/52ebb0bcf32c4a94bbe046c27f192831 : SUCCESS in 7m 05s\n- horizon-openstack-tox-python3-django22 https://zuul.opendev.org/t/openstack/build/8eddd601b99c48f584ea07fe6846a637 : SUCCESS in 6m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e89ecf2474d048799620b8356060c94b : SUCCESS in 8m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb137ce98d4048999ec907e58404427f : SUCCESS in 5m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b69a663f285f4a2db368b17f018a8a21 : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5d64c62ff9f4a378b86448de20a00b6 : SUCCESS in 7m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/54f6344f0ce24d6aa56cdac490e02484 : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94e18c3f6a8045b08ed5568cc91d72dd : SUCCESS in 7m 29s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/1f4c3b121bc54c57b22ee5119e9bbe30 : SUCCESS in 25m 18s\n- horizon-dsvm-tempest-plugin-py27 https://zuul.opendev.org/t/openstack/build/a0b59b98efb54878973ba02f2b9a9edc : SUCCESS in 26m 25s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/590973b10c504adc81a1eaac8b4b1332 : SUCCESS in 6m 34s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/2d5dc1bf2c8f4190b3773a897d4e7384 : SUCCESS in 50m 13s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/f7c295b1f7474ea9bcb093c79cd27c55 : SUCCESS in 23m 30s","accounts_in_message":[],"_revision_number":1},{"id":"bbd9201fee04ddf19eea3d85f083fa4a713f1418","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-31 14:22:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3a9391ff0a51836d9d3c31e79acf9d72ea75c73e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-31 14:25:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d4081d5080c490a85ba79f3584d1b2e : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7d4f76bc65a9180ade63e2d8be485d6cee7293c6","revisions":{"7d4f76bc65a9180ade63e2d8be485d6cee7293c6":{"kind":"REWORK","_number":1,"created":"2020-09-02 20:52:39.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/93/749593/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/93/749593/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/93/749593/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/93/749593/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/93/749593/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/93/749593/1"}}},"commit":{"parents":[{"commit":"d83d58af1e247ca39f90fdb22d0005698540f520","subject":"Fix horizon-nodejs10-run jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/d83d58af1e247ca39f90fdb22d0005698540f520"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-06-24 06:17:04.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-09-02 20:52:11.000000000","tz":540},"subject":"Define periodic jobs","message":"Define periodic jobs\n\nWe sometimes create gate check changes but periodic jobs can do\nmost of them, particularly in stable branhces.\nThis commit is planed to backport to stable branches.\nNote that we already have periodic-stable-jobs (for pyXX and docs jobs)\nbut it is nice to have npm and integration jobs in this coverage.\nOnce it is available in stable branches, I will configure the Grafana\ndashboard to monitor healthiness in stable branches.\n\nNote on stable backport:\nIn stable branches, we have a pipeline named \u0027periodic-stable\u0027\nand it is used in periodic-stable-jobs template.\nIt would be nice to use the same pipeline for all periodic jobs.\n\nChange-Id: Ieb2fe4d2c4453e572c9251dc03c34d08bed62a0f\n(cherry picked from commit 392bc2482eff851504b245c2e28de00694dd9eb8)\n(cherry picked from commit d96d6d60f73864c35ac985cc99f2d266a4223f0f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/7d4f76bc65a9180ade63e2d8be485d6cee7293c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/7d4f76bc65a9180ade63e2d8be485d6cee7293c6"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
