)]}'
{"id":"openstack%2Fhorizon~752016","triplet_id":"openstack%2Fhorizon~master~Ia9ff81694fc7a84417ea2c998721475332a7e772","project":"openstack/horizon","branch":"master","topic":"bug/1895570","hashtags":[],"change_id":"Ia9ff81694fc7a84417ea2c998721475332a7e772","subject":"Set unique identifier for Django messages","status":"MERGED","created":"2020-09-15 11:17:55.000000000","updated":"2020-09-24 02:59:51.000000000","submitted":"2020-09-24 02:58:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752016-1600916307846-5dd22bf2","meta_rev_id":"927291a8dc14a33f2cdc666c0d29a18135f0c8f6","_number":752016,"virtual_id_number":752016,"owner":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-24 02:58:27.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":2,"date":"2020-09-23 17:05:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":1,"date":"2020-09-23 17:10:27.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":2,"date":"2020-09-24 01:36:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":1,"date":"2020-09-24 01:36:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-16 15:15:14.000000000","updated_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"REVIEWER"},{"updated":"2020-09-23 17:05:43.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2020-09-23 17:10:27.000000000","updated_by":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"reviewer":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"state":"REVIEWER"},{"updated":"2020-09-24 01:36:44.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-09-24 02:58:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0424d18e5494a594f39eb8dd88c8e103ce2c0627","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-09-15 11:17:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"78ad56e484f7b9668e779e8d046a573530291a1e","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-09-15 11:22:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"234df66c61a6bf9989806753d9bbe1312383bef8","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-09-15 11:27:25.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7c1b08ca4839dcca91f1c72423f7ebbd72ae8a48","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-09-15 11:29:02.000000000","message":"Removed Code-Review+2 by Vishal Manchanda \u003cmanchandavishal143@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"4b078b8b6c2c7d59cf12106eed06d59e09b7ee8d","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-09-15 11:29:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"559b8b998d3fcd06db384617c6a9315cf4e9ea1e","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-09-15 11:42:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f440bded8ee9944bf3015dbf5db94ed8d7572333","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-09-15 11:43:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a4dd8add13acf1a95588065f0c2f5600a596eb0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 13:17:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/c5f23521b92f43f7b95c020d55330571 : FAILURE in 3m 30s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/aadf91a21c984cf28ae9784c99186e92 : SUCCESS in 6m 42s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/d7c887141c964e4da215f64ec9c713e4 : SUCCESS in 3m 33s (non-voting)\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/8dbf8d4d388d42d59f8d1589e7b5dc08 : SUCCESS in 5m 02s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/abfbe740f2f14cf094a5396d4e084c77 : FAILURE in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f1084393d40461d88d8cc1eb79fcd46 : FAILURE in 2m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ea980cdd98a421b8d3eddf12bcf67cb : POST_FAILURE in 10m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6c757ce1e7046fea0adcb2276964c69 : FAILURE in 2m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45996a3c99514cd38067e7d4cd180609 : FAILURE in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/038463ee40cf406b8399c030389af9df : POST_FAILURE in 14m 26s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/1b78761b7faf45b5a18a43fe677aa9a6 : SUCCESS in 6m 48s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/f2666c8fffde48daae3107ae1814bc7d : FAILURE in 1h 15m 09s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/2d4464b3d30c459d983afd9d1e1321ef : POST_FAILURE in 15m 47s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/60c672a6b1694960b9e3a6b5b1263fd9 : SUCCESS in 4m 48s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/4f6b1b087eb74b638afe88e252940d20 : FAILURE in 12m 48s","accounts_in_message":[],"_revision_number":3},{"id":"fa9b808d666305b9b38e175d2769009c96ac1974","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-09-15 13:22:44.000000000","message":"Patch Set 3:\n\nrecheck.","accounts_in_message":[],"_revision_number":3},{"id":"fa44597e5d1dea53cf5020dfa08a987c11c3047f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 14:56:08.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/f9c3aa4e34de444ba84a702f59cf0bf9 : SUCCESS in 7m 57s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/c20236f9af87449fa947b236d61f5876 : SUCCESS in 7m 41s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/6bcece446ef445199ed279e09eddfa2b : SUCCESS in 5m 56s (non-voting)\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/4c97938b4a9c45a692e0a4d6aaf2dd12 : SUCCESS in 6m 41s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/d6fd1cd5a5cf4d9291c54f9d32672159 : SUCCESS in 7m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c412b18a130f40c5872bf389bd61dddf : SUCCESS in 9m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3f87d9f98b34c208236318a538e9703 : SUCCESS in 11m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f8865cd3ed84c2d8a90b09df4807ea2 : POST_FAILURE in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff8aecc550bd47d789cabbd1117e43ce : SUCCESS in 10m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/041390f52d80499fbd203224fbc11506 : SUCCESS in 14m 57s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/a3f04bc684fe47e2879deb4cca358dc2 : POST_FAILURE in 9m 11s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/792a1fc18ed64bed81b416baa2a5abc6 : SUCCESS in 1h 24m 28s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/a280ab7305884e25b9b5f7d8060e3797 : FAILURE in 15m 38s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3558170b3365425894ae96f646ca330f : SUCCESS in 5m 53s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/6bfa874014774e2c8f56cdf9c6abda73 : FAILURE in 22m 59s","accounts_in_message":[],"_revision_number":3},{"id":"ee8fed3f281dc4fddaa4a1da2f8376d21543c682","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-09-16 06:09:43.000000000","message":"Patch Set 3:\n\nrecheck.","accounts_in_message":[],"_revision_number":3},{"id":"422d27b8d6f266c4ad3d7089d061de7dd46ca5e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 08:08:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/aa77e3e151554a94820ad2a294addf50 : SUCCESS in 10m 02s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/0109a83454cf4e16b638ff8c2edb6f32 : SUCCESS in 10m 23s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/1decc786a4c0460d91032cc36ea1e7da : SUCCESS in 7m 08s (non-voting)\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/e253e890ffd54371a0514916b0b76e82 : SUCCESS in 8m 18s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/e1530d59827544f087cbd348957f7eaf : SUCCESS in 10m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18457045794b4f35970d26ddf6125a90 : SUCCESS in 13m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a98b04ea5cbd4ee3a1fbf8416858e3b2 : SUCCESS in 12m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7e2cd01d9044f5cac22dc8d1882559a : SUCCESS in 9m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ccfdda685f474214ba4256232751893d : SUCCESS in 10m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4395b58c0a294e05ba7aca6ad37fbf22 : SUCCESS in 20m 05s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/643192b1efc4414da5b5e85a9f127cbe : SUCCESS in 9m 21s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/7511775d055b4aed9fc3364035d78e93 : SUCCESS in 1h 35m 06s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/f8faadb7e7c84c6380bb2082577dbdd7 : SUCCESS in 28m 02s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fb02b940b1b545e7a9b7321aa2047e91 : SUCCESS in 5m 54s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/6569e06453bf4bddb8f45ec14cdb7a32 : SUCCESS in 27m 09s","accounts_in_message":[],"_revision_number":3},{"id":"8b16a2605eb2c9718e7fe544301a36e5c5b17bcf","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-09-16 09:43:20.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ad0a480a8d8e6a75b30950e58c13a230179a3cf9","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2020-09-16 13:08:48.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"5ca4ffe1d069db2bdb59bee25e7a74e549f23ac4","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2020-09-16 15:15:14.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"fd19c1c0adc0baac2f0f19b958ff509933556ea1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 17:08:55.000000000","message":"Patch Set 3: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"7f7f4c9d40d877dcdab14d2e0021e7b092203265","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-09-17 07:19:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5c0b6568797001c005289f47594e77b62fa369d4","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-09-17 07:48:37.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"1130c1774e009131b4fbd7ff487a89b5ca2cd199","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 09:18:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/61478725009c4849a61cc6d603e35fdb : SUCCESS in 8m 17s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/f36410a35b2347108befb1620b9052ca : SUCCESS in 9m 49s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/55f95556334f4662aaf661b228fdcd7e : SUCCESS in 5m 57s (non-voting)\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/18e1b491bb0e4f779ea716b10f532be7 : SUCCESS in 7m 43s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/73974f418e4d48da8e52857367cdf28f : SUCCESS in 10m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f892fa2ccfe4167a1c96ca1893029f8 : SUCCESS in 11m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ac3c239656848dbae932c894ed61cc1 : SUCCESS in 10m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/809ffe8c63d740d3820b6c615e65e217 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df8c88370cbe4f8db81d3a20022eb56a : SUCCESS in 12m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41f1dd064e884c23a902fdbc0486b82e : SUCCESS in 23m 01s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/902e13f12a0a46bba4d725754fb33ab5 : SUCCESS in 11m 04s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/6f1bc8b285df4c11949c60a9743fcb56 : SUCCESS in 1h 29m 01s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/8fb137a4c44a4954bde762ae0b610780 : SUCCESS in 43m 23s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2b251247605c4b8882491cd4a6203059 : SUCCESS in 5m 39s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/2259abf0819a4ec286d30f101ce583e2 : SUCCESS in 37m 15s","accounts_in_message":[],"_revision_number":4},{"id":"f262a93cc719acdc4687f845eb769b342db886fc","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2020-09-17 14:10:37.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"a22d18f92652ec7062d73dc68781c52d9b67e3d7","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-09-23 05:36:47.000000000","message":"Patch Set 4: Code-Review+1\n\nLooks good, but I wonder what is an easy way to test the client-side message horizon/templates/horizon/client_side/_alert_message.html....","accounts_in_message":[],"_revision_number":4},{"id":"1926a75569ff891d0e1568771c1929cbb53e527c","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-09-23 07:48:00.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"34006a98dcdb4862306206f23ac94d08e172ccba","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-09-23 07:52:06.000000000","message":"Patch Set 4: Code-Review+1\n\n\u003e (1 comment)\n\n+1. Let\u0027s try with UUID as the difference in message IDs is very less .","accounts_in_message":[],"_revision_number":4},{"id":"b3fbc2804054be91df637d731170bd33ee1b45ca","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-09-23 11:21:56.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c4a5cd15bd06741bf77a58a4f15138ba160949c6","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-09-23 12:28:31.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"fb68bc08aa87d085257499d44f39607aab807994","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-09-23 12:34:52.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"8a2466b93c571a5cde8c1f2b806052c174d559cc","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-09-23 13:01:04.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8b933a99dbcd05c170f19093c8f964ecc377412e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 14:31:20.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/4955972e8324494785bda078c734d3f5 : SUCCESS in 7m 19s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/8e1f8cfeb157420a88958bdc2272bae5 : SUCCESS in 10m 38s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/ca4b0d5d3dc74b4a96cc07fdf581fff6 : SUCCESS in 6m 49s (non-voting)\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/d47aff308a8b4c83b5df0a32df5caa4c : SUCCESS in 7m 59s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/915d8211dfca4ffc8806a64efc179cdc : SUCCESS in 10m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37037b8220724154abae9b2163ed0293 : SUCCESS in 11m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/260cc5f50b7346239c07abc8e5d0903e : SUCCESS in 12m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/674cdb089ce7416c9df1fba6083c72d2 : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3eb4ca777ca049359b9d6d2f12714418 : SUCCESS in 11m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e06e45d6b6f432baa34c7f90eec3d64 : SUCCESS in 8m 01s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/e1c3e106e1d743ddaadee4112d35bbd2 : SUCCESS in 7m 35s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/a8e29688e69d4b008672fc8078b5aa88 : SUCCESS in 1h 20m 37s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/416bbea0927343fd85bdd884118bdcd3 : SUCCESS in 26m 40s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8e170980207f4347bac4ab9793e4dd3f : SUCCESS in 5m 38s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/27546f7624f445849e1656192acdc87f : SUCCESS in 41m 58s","accounts_in_message":[],"_revision_number":7},{"id":"40f3fb03350ccf5b3863c9118f67cf61cef1e9cd","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-09-23 17:05:43.000000000","message":"Patch Set 7: Code-Review+2\n\nLGTM, tested with devstack. I have tested multiple-error messages scenarios for both client-side and non-client-side and this patch fixes both.","accounts_in_message":[],"_revision_number":7},{"id":"83c222700f2ff1e33de868262167e80da1287024","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2020-09-23 17:10:27.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"ac6793846503aa75c45e3bba48eb0dfa6f06bd31","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-09-24 01:36:44.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"9bc2a83a8019c1b4ec266a5fb12eb5aac13afaa8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 01:36:58.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"7490e596fc95a45e0853c50d22c35edcf87506c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 02:58:27.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/3e8ffab4ca88411d90b9ee9111f231ae : SUCCESS in 4m 48s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/2f9b69daa5a9443face6cf1e4da531ae : SUCCESS in 7m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1231b94d7e9246cea9a862131cdcaf59 : SUCCESS in 8m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c09851d3c4714d5e934cc3bf34b1b561 : SUCCESS in 8m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fc33dd95b404aa5a05ae3b9916801b5 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b91e3d71d1a4a7fb0cb81560260fbad : SUCCESS in 8m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d49548cf468430aa041c1b7d91ef1be : SUCCESS in 5m 59s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/32143835b51148c6993f7e8437c294c2 : SUCCESS in 27m 28s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/53bab2c886614472a035c5aa5b51f2b0 : SUCCESS in 6m 33s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/130958b77ef54167912b8f1bcf35ed15 : SUCCESS in 1h 19m 48s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/1b174a7a3702431ba2316ba136587f05 : SUCCESS in 27m 38s","accounts_in_message":[],"_revision_number":7},{"id":"5a233170da76af3a4dd750707bd07e266f4f4573","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 02:58:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"d30f1ef3716dde3068d4ab0307501a079fb37031","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 02:59:51.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba805804f486418c96a4c20bbe328791 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"1b8627c0988d18eaad0d5c4285acab5fa7b8dad4","revisions":{"3a94efffb4b532ddcfd2e639837e6568b1f91059":{"kind":"REWORK","_number":1,"created":"2020-09-15 11:17:55.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/16/752016/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/16/752016/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/16/752016/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/16/752016/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/16/752016/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/16/752016/1"}}},"commit":{"parents":[{"commit":"f8e5fc3f9ea96708fba1b3bd1d53c5f5a988444e","subject":"Merge \"doc: Update default value of REST_API_REQUIRED_SETTINGS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/f8e5fc3f9ea96708fba1b3bd1d53c5f5a988444e"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-09-15 11:07:38.000000000","tz":0},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-09-15 11:17:53.000000000","tz":0},"subject":"Set unique identifier for Django messages","message":"Set unique identifier for Django messages\n\nMessages details template should have unique IDs to get\ncollapse/expand working.\n\nChange-Id: Ia9ff81694fc7a84417ea2c998721475332a7e772\nCloses: Bug #1895570\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/3a94efffb4b532ddcfd2e639837e6568b1f91059"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/3a94efffb4b532ddcfd2e639837e6568b1f91059"}]},"branch":"refs/heads/master"},"80d85b67f14a473b3554c1cd39c60e2d8a8a9251":{"kind":"REWORK","_number":2,"created":"2020-09-15 11:22:34.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/16/752016/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/16/752016/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/16/752016/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/16/752016/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/16/752016/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/16/752016/2"}}},"commit":{"parents":[{"commit":"f8e5fc3f9ea96708fba1b3bd1d53c5f5a988444e","subject":"Merge \"doc: Update default value of REST_API_REQUIRED_SETTINGS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/f8e5fc3f9ea96708fba1b3bd1d53c5f5a988444e"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-09-15 11:07:38.000000000","tz":0},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-09-15 11:22:24.000000000","tz":0},"subject":"Set unique identifier for Django messages","message":"Set unique identifier for Django messages\n\nMessages details template should have unique IDs to get\ncollapse/expand working.\n\nChange-Id: Ia9ff81694fc7a84417ea2c998721475332a7e772\nCloses: Bug #1895570\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/80d85b67f14a473b3554c1cd39c60e2d8a8a9251"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/80d85b67f14a473b3554c1cd39c60e2d8a8a9251"}]},"branch":"refs/heads/master"},"c42ca16eac9f04515e43c6633f7625910b85531c":{"kind":"REWORK","_number":3,"created":"2020-09-15 11:43:07.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/16/752016/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/16/752016/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/16/752016/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/16/752016/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/16/752016/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/16/752016/3"}}},"commit":{"parents":[{"commit":"f8e5fc3f9ea96708fba1b3bd1d53c5f5a988444e","subject":"Merge \"doc: Update default value of REST_API_REQUIRED_SETTINGS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/f8e5fc3f9ea96708fba1b3bd1d53c5f5a988444e"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-09-15 11:07:38.000000000","tz":0},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-09-15 11:42:54.000000000","tz":0},"subject":"Set unique identifier for Django messages","message":"Set unique identifier for Django messages\n\nMessages details template should have unique IDs to get\ncollapse/expand working.\n\nChange-Id: Ia9ff81694fc7a84417ea2c998721475332a7e772\nCloses: Bug #1895570\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/c42ca16eac9f04515e43c6633f7625910b85531c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/c42ca16eac9f04515e43c6633f7625910b85531c"}]},"branch":"refs/heads/master"},"54e3497e4358c76f5811d8b963b0f6000323a760":{"kind":"REWORK","_number":4,"created":"2020-09-17 07:19:18.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/16/752016/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/16/752016/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/16/752016/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/16/752016/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/16/752016/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/16/752016/4"}}},"commit":{"parents":[{"commit":"d9bf46371eb411f227ef4bbfd9e002ba7e5d2a4b","subject":"Merge \"Refactor error messages.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/d9bf46371eb411f227ef4bbfd9e002ba7e5d2a4b"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-09-15 11:07:38.000000000","tz":0},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-09-17 07:19:00.000000000","tz":180},"subject":"Set unique identifier for Django messages","message":"Set unique identifier for Django messages\n\nMessages details template should have unique IDs to get\ncollapse/expand working.\n\nChange-Id: Ia9ff81694fc7a84417ea2c998721475332a7e772\nCloses: Bug #1895570\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/54e3497e4358c76f5811d8b963b0f6000323a760"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/54e3497e4358c76f5811d8b963b0f6000323a760"}]},"branch":"refs/heads/master"},"edf46f4fd9c6c05b9ba4379f892f88880b577fb4":{"kind":"REWORK","_number":5,"created":"2020-09-23 12:28:31.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/16/752016/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/16/752016/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/16/752016/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/16/752016/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/16/752016/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/16/752016/5"}}},"commit":{"parents":[{"commit":"d9bf46371eb411f227ef4bbfd9e002ba7e5d2a4b","subject":"Merge \"Refactor error messages.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/d9bf46371eb411f227ef4bbfd9e002ba7e5d2a4b"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-09-15 11:07:38.000000000","tz":0},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-09-23 12:28:29.000000000","tz":0},"subject":"Set unique identifier for Django messages","message":"Set unique identifier for Django messages\n\nMessages details template should have unique IDs to get\ncollapse/expand working.\n\nChange-Id: Ia9ff81694fc7a84417ea2c998721475332a7e772\nCloses: Bug #1895570\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/edf46f4fd9c6c05b9ba4379f892f88880b577fb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/edf46f4fd9c6c05b9ba4379f892f88880b577fb4"}]},"branch":"refs/heads/master"},"2835aa519998ccf37997a40eb67d663d8007aefd":{"kind":"REWORK","_number":6,"created":"2020-09-23 12:34:52.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/16/752016/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/16/752016/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/16/752016/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/16/752016/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/16/752016/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/16/752016/6"}}},"commit":{"parents":[{"commit":"d9bf46371eb411f227ef4bbfd9e002ba7e5d2a4b","subject":"Merge \"Refactor error messages.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/d9bf46371eb411f227ef4bbfd9e002ba7e5d2a4b"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-09-15 11:07:38.000000000","tz":0},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-09-23 12:34:31.000000000","tz":0},"subject":"Set unique identifier for Django messages","message":"Set unique identifier for Django messages\n\nMessages details template should have unique IDs to get\ncollapse/expand working.\n\nChange-Id: Ia9ff81694fc7a84417ea2c998721475332a7e772\nCloses: Bug #1895570\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/2835aa519998ccf37997a40eb67d663d8007aefd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/2835aa519998ccf37997a40eb67d663d8007aefd"}]},"branch":"refs/heads/master"},"1b8627c0988d18eaad0d5c4285acab5fa7b8dad4":{"kind":"REWORK","_number":7,"created":"2020-09-23 13:01:04.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/16/752016/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/16/752016/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/16/752016/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/16/752016/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/16/752016/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/16/752016/7"}}},"commit":{"parents":[{"commit":"d9bf46371eb411f227ef4bbfd9e002ba7e5d2a4b","subject":"Merge \"Refactor error messages.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/d9bf46371eb411f227ef4bbfd9e002ba7e5d2a4b"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-09-15 11:07:38.000000000","tz":0},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-09-23 13:00:43.000000000","tz":180},"subject":"Set unique identifier for Django messages","message":"Set unique identifier for Django messages\n\nMessages details template should have unique IDs to get\ncollapse/expand working.\n\nChange-Id: Ia9ff81694fc7a84417ea2c998721475332a7e772\nCloses: Bug #1895570\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/1b8627c0988d18eaad0d5c4285acab5fa7b8dad4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/1b8627c0988d18eaad0d5c4285acab5fa7b8dad4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
