)]}'
{"id":"openstack%2Fhorizon~754367","triplet_id":"openstack%2Fhorizon~stable%2Fvictoria~If4399524c7e7687b146d324198aeb38e1660d6c2","project":"openstack/horizon","branch":"stable/victoria","topic":"create-victoria","hashtags":[],"change_id":"If4399524c7e7687b146d324198aeb38e1660d6c2","subject":"Update .gitreview for stable/victoria","status":"MERGED","created":"2020-09-25 12:27:59.000000000","updated":"2020-09-25 13:51:53.000000000","submitted":"2020-09-25 13:49:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"754367-1601041791187-e4987785","meta_rev_id":"a915ea7a9577e315ba90962543240da1009ddf5e","_number":754367,"virtual_id_number":754367,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-25 13:49:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":2,"date":"2020-09-25 13:35:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":1,"date":"2020-09-25 13:35:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-25 13:35:06.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-09-25 13:49:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b3886bd3d44c6bbd0191023aa11a056a5b34e036","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-09-25 12:27:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"91608a7897bbcb6d1196bde900e0bf7cfe6e0b18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 12:49:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/bcb3d59394df437da3fddd62af0783c4 : SUCCESS in 8m 19s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/3db756f2687749ff82e7e0029a85d563 : FAILURE in 8m 45s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/b2bf7f23dbe64156835b43bc80437c78 : SUCCESS in 5m 15s (non-voting)\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/a48b61a23c524639a3784c0265a29c75 : SUCCESS in 6m 49s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/d99cb95d88474d3b9558fa9ec404d35f : SUCCESS in 11m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b97da0b14254b72b1747a6a613d642d : SUCCESS in 10m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80f3d0ef8a6e4d3da7c89e71489970da : SUCCESS in 11m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67bee7238c964a72a07d3916bba7f68a : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a62a1564023d45edb17d0d76f1d9196a : SUCCESS in 10m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf4504b92acf4680b47c9b124bba5db5 : SUCCESS in 7m 26s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/3a0a391a3efa4b1d97190b0fae1074b0 : SUCCESS in 9m 02s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b4f60578d9534b4e8b92ca3f71d1dab9 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":1},{"id":"eb34bbf793f7e78bdcad9e2343bc8b6d773cc5cf","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-09-25 13:35:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dfa26553a08686ad76b5e06613c32ecf4d3b6e19","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 13:35:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4da3717d29aa644c79917f7da4dbff7ec54b308a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 13:49:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/2dccac3e2604466e88e034fe0a2e0fcd : SUCCESS in 5m 06s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/fc442e6601d24e83b7772a75e461b8f5 : SUCCESS in 8m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0541a81247134fe2a86ceab26bb6250e : SUCCESS in 9m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d072728a89af42d6a7cfd836dba67ade : SUCCESS in 10m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21c96edd03784f3eb15f27510d64a0cc : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8f265f75f744860a63a94e75c9c2b38 : SUCCESS in 10m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/767fbbc5066147599222a169a6ef8b14 : SUCCESS in 7m 44s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/d714a7b4afb54a6b83cea8fb0d2632d9 : SUCCESS in 7m 58s","accounts_in_message":[],"_revision_number":1},{"id":"cc2f33a9582c224520e3e9932d4952990109ea16","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 13:49:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cccb625569aefde15f52b827a378b3c97758f367","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 13:51:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/976e2004771441689480fff5d515c1de : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c410e860b11615d29b66fb8a6b9c9c3a086506aa","revisions":{"c410e860b11615d29b66fb8a6b9c9c3a086506aa":{"kind":"REWORK","_number":1,"created":"2020-09-25 12:27:59.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/67/754367/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/67/754367/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/67/754367/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/67/754367/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/67/754367/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/67/754367/1"}}},"commit":{"parents":[{"commit":"c44e651589f8589452ab6e436d3428a8d56fa9cf","subject":"Merge \"Add Unit Test for cinder-user-facing messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/c44e651589f8589452ab6e436d3428a8d56fa9cf"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-25 12:27:57.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-25 12:27:57.000000000","tz":0},"subject":"Update .gitreview for stable/victoria","message":"Update .gitreview for stable/victoria\n\nChange-Id: If4399524c7e7687b146d324198aeb38e1660d6c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/c410e860b11615d29b66fb8a6b9c9c3a086506aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/c410e860b11615d29b66fb8a6b9c9c3a086506aa"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
