)]}'
{"id":"openstack%2Fhorizon~754658","triplet_id":"openstack%2Fhorizon~master~Id43adbd44778d3375c0d49da3f7530cbb99e16fe","project":"openstack/horizon","branch":"master","topic":"bug/1888490","hashtags":[],"change_id":"Id43adbd44778d3375c0d49da3f7530cbb99e16fe","subject":"Non-api filters not working with admin instance tab","status":"MERGED","created":"2020-09-28 08:59:22.000000000","updated":"2020-11-18 13:35:02.000000000","submitted":"2020-11-18 13:33:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"754658-1605706386522-422172d5","meta_rev_id":"1f34b9cd2552be02ece4d3e84e631d7111892de0","_number":754658,"virtual_id_number":754658,"owner":{"_account_id":31916,"name":"vinay harsha mitta","email":"vinay.harsha@tcs.com","username":"vinay_m"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"date":"2020-10-28 21:08:11.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-18 13:33:06.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31916,"name":"vinay harsha mitta","email":"vinay.harsha@tcs.com","username":"vinay_m"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":2,"date":"2020-10-29 10:40:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":1,"date":"2020-11-04 19:59:18.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":2,"date":"2020-11-18 10:05:22.000000000","_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31916,"name":"vinay harsha mitta","email":"vinay.harsha@tcs.com","username":"vinay_m"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":1,"date":"2020-11-18 10:05:22.000000000","_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31916,"name":"vinay harsha mitta","email":"vinay.harsha@tcs.com","username":"vinay_m"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31916,"name":"vinay harsha mitta","email":"vinay.harsha@tcs.com","username":"vinay_m"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":31916,"name":"vinay harsha mitta","email":"vinay.harsha@tcs.com","username":"vinay_m"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-28 21:08:11.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-10-29 10:40:51.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2020-11-04 19:59:18.000000000","updated_by":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"reviewer":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"state":"REVIEWER"},{"updated":"2020-11-18 10:05:22.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-11-18 13:33:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"629235ed816670b4009d7a4c4a76f00b77fc4271","author":{"_account_id":31916,"name":"vinay harsha mitta","email":"vinay.harsha@tcs.com","username":"vinay_m"},"date":"2020-09-28 08:59:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9bc645e1da16ac44771ea77a97d1dd2a9536f91b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 10:45:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/5b805e4adc8e47f288091536794ccd71 : SUCCESS in 7m 09s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/e93bc4f174d14a6694a140e2c460090b : FAILURE in 9m 01s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/d7d382854310494c82e188b89790ad6c : SUCCESS in 5m 17s (non-voting)\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/9c93510cbefb4f81a34efef63ef2054f : SUCCESS in 7m 22s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/fa518df80e28451f8e39154c01f6c03a : SUCCESS in 7m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca756b3f919f4b30be803f601f1ddead : FAILURE in 11m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da7f21ffe66141b78e75ba4b4daa4ec5 : FAILURE in 12m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5a5db4ce70a412b811546591b484c78 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ea42bb898404a1dbbf3f7b0670aeaa5 : FAILURE in 10m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/088a3ce7be504cb7b902742c44defea9 : SUCCESS in 7m 25s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/089db93ef2064daa98fad83d81310297 : SUCCESS in 11m 38s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/ff20451cbd3d499ca46e29c6aee20058 : SUCCESS in 1h 35m 30s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/f3feb0126d4c4d44a3943c4b1ba15a3a : SUCCESS in 36m 50s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e60f70d817d14938b47d4d6b6a66a9ba : SUCCESS in 5m 04s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/42cee7936df143b28e4dbe7c33548888 : SUCCESS in 40m 47s","accounts_in_message":[],"_revision_number":1},{"id":"f99db5a01626a6cd21ca3f6749b98c2a41a2fb16","author":{"_account_id":31916,"name":"vinay harsha mitta","email":"vinay.harsha@tcs.com","username":"vinay_m"},"date":"2020-09-28 11:32:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d348184c74acfd701967bdf55e82fa5c5365aa4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 13:31:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/a62db7d68093487d847d7ef754e0db2a : SUCCESS in 8m 18s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/7732999f473c46efb34114ac17e2950b : FAILURE in 7m 10s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/cfc2800efe6a486a915d38f85a523f11 : SUCCESS in 4m 59s (non-voting)\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/d8d7b7a314af4f1982e7d848b9ea6cda : SUCCESS in 7m 28s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/ccadac719f29460f91e552a0c90594b0 : SUCCESS in 7m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4387b9c28e69406bbbb9e1ba30dbb017 : FAILURE in 10m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/116559d218134abe85443b01d74fe373 : FAILURE in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd686581cab549b8aec1b0417451c69f : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/715285fff5a74d9abb12552df35375b8 : FAILURE in 12m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80e8d4f493784f5098a1e856d73c1eb2 : SUCCESS in 5m 56s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/816fec44813548a283301c68ee9611cd : SUCCESS in 10m 51s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/d7efd1f9a8b54b29bfb193f95e2250c8 : SUCCESS in 1h 53m 17s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/ca59d13b59694781a05af7db6cf2842a : SUCCESS in 38m 11s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e6c4a9205a804371a650a12ff7fb2a43 : SUCCESS in 4m 35s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/f5edf7c44512440bb0591bd139553c32 : SUCCESS in 31m 40s","accounts_in_message":[],"_revision_number":1},{"id":"41f7a096f918454eefca507256dd78cbe6706e24","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-10-05 07:39:12.000000000","message":"Patch Set 1: Code-Review-1\n\nhi, thanks for the patch. Could you fix the failed Unit tests [1]?\n\n[1] https://zuul.opendev.org/t/openstack/build/715285fff5a74d9abb12552df35375b8/log/job-output.txt#3034-3041","accounts_in_message":[],"_revision_number":1},{"id":"aa04893b763ed35b412b6741e08c89ee87d67a58","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-10-05 07:41:38.000000000","message":"Patch Set 1:\n\nI have fix one unit test in paste [2]. Similarly, you can fix other tests.\n\n[2] http://paste.openstack.org/show/798687/","accounts_in_message":[],"_revision_number":1},{"id":"dae14a7493c0d73dcc85ddbda1499155ed6f4ae9","author":{"_account_id":31916,"name":"vinay harsha mitta","email":"vinay.harsha@tcs.com","username":"vinay_m"},"date":"2020-10-05 11:22:02.000000000","message":"Patch Set 1:\n\n\u003e I have fix one unit test in paste [2]. Similarly, you can fix other\n \u003e tests.\n \u003e \n \u003e [2] http://paste.openstack.org/show/798687/\n\nHi vishal , thanks for the paste and will look into other test\u0027s.\nAnd need a suggestion , thinking to add a condition that checks for non-api filters in search_opts dict and then a api call will be happening.","accounts_in_message":[],"_revision_number":1},{"id":"44e3d6f36812b888e91e7ad5e24ac5f79400f1b5","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-10-05 15:25:37.000000000","message":"Patch Set 1:\n\n\u003e \u003e I have fix one unit test in paste [2]. Similarly, you can fix\n \u003e other\n \u003e \u003e tests.\n \u003e \u003e\n \u003e \u003e [2] http://paste.openstack.org/show/798687/\n \u003e \n \u003e Hi vishal , thanks for the paste and will look into other test\u0027s.\n \u003e And need a suggestion , thinking to add a condition that checks for\n \u003e non-api filters in search_opts dict and then a api call will be\n \u003e happening.\n\nWhat do you mean, Isn\u0027t a condition is already there on Line 166?","accounts_in_message":[],"_revision_number":1},{"id":"7f91fe7eeb48a91bdc7650935e55946831a88e68","author":{"_account_id":31916,"name":"vinay harsha mitta","email":"vinay.harsha@tcs.com","username":"vinay_m"},"date":"2020-10-06 14:01:25.000000000","message":"Patch Set 1:\n\n\u003e \u003e \u003e I have fix one unit test in paste [2]. Similarly, you can fix\n \u003e \u003e other\n \u003e \u003e \u003e tests.\n \u003e \u003e \u003e\n \u003e \u003e \u003e [2] http://paste.openstack.org/show/798687/\n \u003e \u003e\n \u003e \u003e Hi vishal , thanks for the paste and will look into other test\u0027s.\n \u003e \u003e And need a suggestion , thinking to add a condition that checks\n \u003e for\n \u003e \u003e non-api filters in search_opts dict and then a api call will be\n \u003e \u003e happening.\n \u003e \n \u003e What do you mean, Isn\u0027t a condition is already there on Line 166?\n\ncorrect if i\u0027m wrong, the condition check over the line 166 does returns true when ever any mapping happens ,i.e if api contains non-api filters right?","accounts_in_message":[],"_revision_number":1},{"id":"12e8a93d262ba6ea52b6f9c51871ebcbc9b2c5df","author":{"_account_id":31916,"name":"vinay harsha mitta","email":"vinay.harsha@tcs.com","username":"vinay_m"},"date":"2020-10-08 15:53:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f90398ea09d1959627e4fdbb9ad56050a70b16f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 17:40:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/aadce6a5fb6047b7b5fb7e9b8692d414 : SUCCESS in 5m 59s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/a151979c879844fe986c20e14ca0e64b : SUCCESS in 6m 20s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/6f108ee3485041d5871e5c30f53b887b : SUCCESS in 5m 07s (non-voting)\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/8d17112b1d0f4c06bee4870d86064486 : SUCCESS in 9m 54s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/9b585148758b4da3be3ee1faee7f4326 : SUCCESS in 7m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e51fa15f4e63497dbe0731f6a0a4e771 : FAILURE in 10m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83884e12b3e0403fb70740d3fefc4196 : FAILURE in 10m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fceedf1f9fff4431b277179c239851ea : SUCCESS in 7m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/afbb9febfc7d423bb5a2752a01f6c732 : FAILURE in 10m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cae4a5cc53154b0696c04ec944d523f9 : SUCCESS in 6m 01s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/cc6daf2e19f24372a1363aac34601c96 : SUCCESS in 7m 02s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/d89b9117503c4e678728de6d4b953886 : SUCCESS in 1h 35m 32s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/1c093e6f4e0d448bb1b2b9793c597f75 : SUCCESS in 45m 30s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bf31e8cf72dd4d3ba5a4dfbe99a14974 : SUCCESS in 5m 24s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/2f414c1ea0884ebd8ed327ead499b914 : SUCCESS in 34m 13s","accounts_in_message":[],"_revision_number":2},{"id":"f84c458ba8e0b81ee384c3ff4db8ec9c8c7aa0eb","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-10-09 06:06:19.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a1517b1f4471875e3604f11b5b6e000901a536a6","author":{"_account_id":31916,"name":"vinay harsha mitta","email":"vinay.harsha@tcs.com","username":"vinay_m"},"date":"2020-10-09 06:48:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a465be016430cd2f4fc43025717c22b21907571e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 08:45:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/989bca22ea2b49f5976a0ac3f93f59e6 : SUCCESS in 8m 18s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/f86a1dac2410445e9fc261b57e00e3ed : SUCCESS in 8m 47s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/5643fc286b7646e082a8e0a2520d0a64 : SUCCESS in 5m 39s (non-voting)\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/b2611e348dc94a1e97b2bdb325bce406 : SUCCESS in 5m 37s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/54f6d8143d394aee892a5001ac0cc524 : SUCCESS in 8m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f23b1796cb5472fa422943609602597 : SUCCESS in 9m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78f1be32a8884df2b3999af933233541 : SUCCESS in 10m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f68eec1e267e49e6b2dda66c80971382 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c84fc25186b44b495200737b9d4500d : SUCCESS in 10m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d71b1fc166344b8a3cbc5b629a52d67 : SUCCESS in 6m 49s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/336713eb820644d29b7fe316e76dfe77 : SUCCESS in 9m 30s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/8d9512e2f67848cb9f0b4bf106666c14 : SUCCESS in 1h 34m 04s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/93be30b18d9941b9931263022369a285 : SUCCESS in 37m 56s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7b49abf251b14b8abc2b8ffb593e3ca6 : SUCCESS in 4m 22s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/49a2b7d160754346ab8f868d0b6814f5 : SUCCESS in 37m 33s","accounts_in_message":[],"_revision_number":3},{"id":"3c4eabbaf576025aef29d887f998ba50ad5f61cf","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-10-09 18:50:38.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM, I have tested the case mentioned in the bug summary and this patch fixes that issue.","accounts_in_message":[],"_revision_number":3},{"id":"7c258a65bd65b362647870d0a48e30da1a57ca8d","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2020-10-09 20:44:18.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"ee96efea1d60e6b27a0b7ed4de5a83f2a78409b3","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-10-12 09:48:45.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ca13a39f9f4fb35bf141e33d49e193d7d605aa57","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-10-12 09:54:10.000000000","message":"Patch Set 3:\n\nwe\u0027ve done various performance improvements including reducing the number of API calls, minimizing the number of responses as much as possible, parallelizing API calls and so on. Doubling one of the heavy APIs calls will break these effort so far. Thus, I am against this nasty workaround.","accounts_in_message":[],"_revision_number":3},{"id":"de078cca3b7238ab535df1de6bfc9962be9a45e8","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-10-12 10:17:43.000000000","message":"Patch Set 3: -Code-Review\n\nremoving my vote due to amotoki comments.","accounts_in_message":[],"_revision_number":3},{"id":"89405c9c8b05ce00a9afad0dff03241e96857bf7","author":{"_account_id":31916,"name":"vinay harsha mitta","email":"vinay.harsha@tcs.com","username":"vinay_m"},"date":"2020-10-12 10:27:27.000000000","message":"Patch Set 3:\n\n\u003e (1 comment)\n\nyes regarding this double api call only , i have doubt with performance and  tried to convey about it in #openstack-horizon channel many a times. The first api call is required for to retrieve the image ids mapped to of those instances. My doubt was what if here implemented same as done in projects tab.","accounts_in_message":[],"_revision_number":3},{"id":"d865d1b4de1ba587999112b18bc7e4a2e3481aee","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-10-12 10:44:55.000000000","message":"Patch Set 3:\n\n(Is it okay to continue the discussion in the cover comment rather than the inline comment? You used it, so it seems you prefer to it.)\n\nRegarding the performance, it is part of the admin API and the number of instances can be larger compared to the project side.\n\nPerhaps this regression happened when we switched to specify related image IDs when calling the image list API. I apparently agree with you that we need to call the server list API after resolving the non-API filters. The question now is whether we need the first call. Perhaps we need to move the call of the image list after calling the server list when \u0027image\u0027 filter is not specified.","accounts_in_message":[],"_revision_number":3},{"id":"10340ba91292ff038b5b2908a43d32711c87523f","author":{"_account_id":31916,"name":"vinay harsha mitta","email":"vinay.harsha@tcs.com","username":"vinay_m"},"date":"2020-10-13 14:34:32.000000000","message":"Patch Set 3:\n\nTBF sir, i\u0027m not clear of why the need of instances for retrieving images in admin/instances/views.py , where as in project/instances/views.py there is no such implementation.\n\nCorrect me if i\u0027m wrong sir, the main aspect of that get_images_by_ids in admin/instances/views was to get only the images which are used to spwan the instances, beacuase to simply get all images under admin contains all tenents images . so inoder to overcome, there was a first api call to get instances which used to get images.","accounts_in_message":[],"_revision_number":3},{"id":"4367bc186c3922027cedb74907465f0a2e0f919e","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2020-10-13 22:13:53.000000000","message":"Patch Set 3: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c5831a514c82dfdffffc044279986b22164529e1","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-10-14 04:37:17.000000000","message":"Patch Set 3:\n\n\u003e TBF sir, i\u0027m not clear of why the need of instances for retrieving images in admin/instances/views.py , where as in project/instances/views.py there is no such implementation.\n\u003e Correct me if i\u0027m wrong sir, the main aspect of that get_images_by_ids in admin/instances/views was to get only the images which are used to spwan the instances, beacuase to simply get all images under admin contains all tenents images . so inoder to overcome, there was a first api call to get instances which used to get images.\n\nIt looks like you just consider small and/or simple deployments. You need to imagine more complex deployments.\nFor example, you can imagine a deployment with hundreds of tenants and several (or more) tenants registers over one hundred images per their needs. It might not be cloud-native but it is the real and I know such deployments. In such case the admin context can retrieve all images of all tenants and it leads to large number easily. It is not true for the project panel as it just retrieves images from a single tenant. This is the biggest difference between the project and admin panels.\n\nI hope this answers your question. That\u0027s the background of my -1.","accounts_in_message":[],"_revision_number":3},{"id":"0efce06e6856614dee588664b27ba92a90816e11","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-10-14 04:38:04.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"63ca5ab18670a7a060a0d4e004a03ceabbb8876f","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-10-14 04:42:34.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nI pushed patch set 2 to show what you can handle the situation with non-API filters.\n\nI reverted your changes in UT, but a new test case needs to be added to cover this change. (Note that I didn\u0027t check patch set 2 passes UT or not as it is to show a way to handle the situation.)","accounts_in_message":[],"_revision_number":4},{"id":"496b7d8c637119ee1bf93f48409f8884d128eaf4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 06:06:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/a75fe58ba714430399b9a12f852f4134 : SUCCESS in 5m 05s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/4f86bf97463b4422a89847fd525b0a02 : SUCCESS in 7m 08s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/94a2b456575e42f788a1ed01338cf4af : SUCCESS in 3m 40s (non-voting)\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/29f3cac933134bbeaac4a2edaf2c9895 : SUCCESS in 5m 27s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/b129cf19adec4116a09104e8fc4509a9 : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95129a93c7f540c8aa7c7915c19f858b : SUCCESS in 8m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8a15a0037584ff988355faf9abe549d : SUCCESS in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1190ec9f8d0e4063a5e10866bde539c3 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8213c69d3ec4fa5b80c3828ada79ea8 : SUCCESS in 6m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6e4e75493c94169b1a98ac7bdc6570e : SUCCESS in 7m 34s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/c4b618eaa4f747358c20926dc73c5913 : SUCCESS in 9m 08s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/6e5cc41dabfe4f548aebc866282b3e9f : SUCCESS in 1h 27m 00s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/a0daef415b324c9e84d41739ba1bf29c : SUCCESS in 36m 30s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/10cd440060944606b34049a4dddf737b : SUCCESS in 4m 38s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/c55eb6d6935940948b9f43042f25b1c7 : SUCCESS in 36m 47s","accounts_in_message":[],"_revision_number":4},{"id":"5f3e9979bfa5e26067ef9b4dea65fa828f30a680","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-10-27 08:47:36.000000000","message":"Patch Set 4: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":4},{"id":"2726ae8e930cb7e20d2baa97e0b82095e1f57eb2","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-10-27 11:07:33.000000000","message":"Patch Set 4: Workflow-1\n\nThe commit message needs to be updated.","accounts_in_message":[],"_revision_number":4},{"id":"d0dccae91145f4c4b4aa701fccd9de7bcf934872","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-10-27 11:25:56.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"a2712e3ab1352604d63977c8165f9a7f5b713ff1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 13:14:06.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/0f0fe9d9c6774b7b8769609216eb8014 : SUCCESS in 6m 07s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/65d04e01b3e043579dcf89423267aed7 : SUCCESS in 15m 56s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/afe045faeaac4363a30aa3321c43d170 : SUCCESS in 4m 29s (non-voting)\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/6b1fc0dc37484c819fd0a6ce3305bebf : SUCCESS in 5m 38s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/6912a3aa035f46a39586a381273ee9a4 : SUCCESS in 6m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0b96f838eeb45599b829da5d5377a0c : SUCCESS in 7m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e36b74b67a1d40a8ae80ac90da87e9ff : SUCCESS in 11m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/602c21f5112b4c31888c62b66b9374fc : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/128f1c8a84e44506852faa7ced0794dd : SUCCESS in 7m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d4aa232e61140c882e1456d69e5e1e7 : SUCCESS in 6m 12s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/de62612f5c664f1eb1f325bc266bcd3f : SUCCESS in 7m 09s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/39391f2e1aa34b59b7c29f9c0ba9d67b : FAILURE in 1h 33m 50s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/4c6f894fc3c24f04950fe94d87caf2b6 : SUCCESS in 35m 47s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f2b29460105d40b69252e3b057713868 : SUCCESS in 4m 33s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/2e6dc60ccb5440f48ea348d5bf39319a : SUCCESS in 29m 33s","accounts_in_message":[],"_revision_number":5},{"id":"d368121d07d4305a4beb8f5da2300c4e16b4cfae","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-10-28 21:08:11.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"df0f43548fcb5b81b83064132d633aa0c1cc73b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 22:53:39.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/92aadd78615d43febbca6ef83620b570 : SUCCESS in 5m 43s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/6c0debc23c82438389054e1b050433f2 : SUCCESS in 6m 48s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/c1adeccf40004d57ab464183c55ded2b : SUCCESS in 10m 26s (non-voting)\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/565e32e74c2c4e68859f8937412901fb : SUCCESS in 9m 17s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/4d231e05e9f44cedb0ef2d9340894072 : SUCCESS in 7m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf773e21891e424991f0c0aa09fa5cab : SUCCESS in 10m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20feddd8ce204a54a3aa492c3367a7bb : SUCCESS in 9m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e3bb9f1a1d845bb8d81a37528444e14 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f23b499d88c2447bbd00d9773898856f : SUCCESS in 10m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e9fcadfa41845a084a92b8a7d045260 : SUCCESS in 6m 01s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/2cb3373e10e74630b6753e24b9d54840 : SUCCESS in 7m 23s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/5a17796318d64c71ada7b50419f315af : SUCCESS in 1h 35m 56s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/2de1d884315e46038e15313ed37fa87a : SUCCESS in 46m 40s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/07db887991f34805b76911401561b73a : SUCCESS in 4m 30s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/2883ae4d115c43a0853b1b1a30a8006d : SUCCESS in 36m 18s","accounts_in_message":[],"_revision_number":5},{"id":"8fb4905796b2dd8ad4f62996f70f62aca335f088","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-10-29 10:40:51.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"1c49434dd5b7e588dfa4d2d26c7af39498d99687","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2020-11-04 19:59:18.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"91ea8dcb6e1e877cd173d4a1e440f9f08df7c540","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-11-18 10:05:22.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"0104f00aa529fe223e53b56e9d40f852b9f63917","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 10:05:57.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a7bf195e2b6744fd8276a795622e439f67a20294","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 13:33:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"cf7b3b2c90436fca0c6ff46c0558cc7a6e2342b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 13:33:06.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/4035eb16b2674a5687c4f35245e62cbf : SUCCESS in 7m 24s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/55b4c2204a594d7caee86650a563160a : SUCCESS in 7m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2d2de59f4ea4f85b07e88d208effe1c : SUCCESS in 9m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e87a59dec7a483eac1b2764e20581b7 : SUCCESS in 9m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a85ad40622df4e138c06681180882cb6 : SUCCESS in 7m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a19a3d5ac5849f0be403be3df04ba9f : SUCCESS in 13m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7d7ed6e6060492ca762e20e191aa162 : SUCCESS in 14m 56s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/444bc7f60a5945f19e060abb95ff6c2d : SUCCESS in 42m 03s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/160d2cf79d6942138c4952e67e299f2d : SUCCESS in 13m 45s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/26dc9af6b7b241debb3ca985b1b76dbd : SUCCESS in 1h 36m 25s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/d37908c56dd843839803334ab864b7c5 : SUCCESS in 44m 16s","accounts_in_message":[],"_revision_number":5},{"id":"3e95842b8782154e808c95a03f52ab09e744ee53","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 13:35:02.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3fac872f6074f7fbc758970efdf8096 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"6b79cdcc1b92a968c967286cc267bac39803c85e","revisions":{"c0c0846235958963c94ceac3eb09f9c2c5d98c54":{"kind":"REWORK","_number":1,"created":"2020-09-28 08:59:22.000000000","uploader":{"_account_id":31916,"name":"vinay harsha mitta","email":"vinay.harsha@tcs.com","username":"vinay_m"},"ref":"refs/changes/58/754658/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/58/754658/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/58/754658/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/58/754658/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/58/754658/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/58/754658/1"}}},"commit":{"parents":[{"commit":"106e713ee553acce775120520ecaf70e4734981f","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/106e713ee553acce775120520ecaf70e4734981f"}]}],"author":{"name":"vinay_m","email":"vinay.harsha@tcs.com","date":"2020-09-28 08:52:18.000000000","tz":330},"committer":{"name":"vinay_m","email":"vinay.harsha@tcs.com","date":"2020-09-28 08:59:11.000000000","tz":330},"subject":"Non-api Filters not working with admin tab.","message":"Non-api Filters not working with admin tab.\n\nThe issue was with GET call for instances ,\ni.e the filtering opts like project_name, image_name,\nflavor_name are not supported by nova_api (non-api filters),\nso these get mapped to their id\u0027s first and then updated in\nsearch_opts dict which is passed for retrieving instances.\nFor the admin tab this call[1] was happening prior to\nnon-api filters mapping[2] , so in-order to work a second call\nwith updated search_opts dict should be called.\n\n[1] .https://opendev.org/openstack/horizon/src/branch/master/openstack_dashboard/dashboards/admin/instances/views.py#L154\n\n[2]. https://opendev.org/openstack/horizon/src/branch/master/openstack_dashboard/dashboards/admin/instances/views.py#L166\n\nChange-Id: Id43adbd44778d3375c0d49da3f7530cbb99e16fe\nCloses-Bug: #1888490\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/c0c0846235958963c94ceac3eb09f9c2c5d98c54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/c0c0846235958963c94ceac3eb09f9c2c5d98c54"}]},"branch":"refs/heads/master"},"d3e2dd79de2d5942b31b619a7c9ddb82be53b12c":{"kind":"REWORK","_number":2,"created":"2020-10-08 15:53:12.000000000","uploader":{"_account_id":31916,"name":"vinay harsha mitta","email":"vinay.harsha@tcs.com","username":"vinay_m"},"ref":"refs/changes/58/754658/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/58/754658/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/58/754658/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/58/754658/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/58/754658/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/58/754658/2"}}},"commit":{"parents":[{"commit":"106e713ee553acce775120520ecaf70e4734981f","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/106e713ee553acce775120520ecaf70e4734981f"}]}],"author":{"name":"vinay_m","email":"vinay.harsha@tcs.com","date":"2020-09-28 08:52:18.000000000","tz":330},"committer":{"name":"vinay_m","email":"vinay.harsha@tcs.com","date":"2020-10-08 15:51:56.000000000","tz":330},"subject":"Non-api Filters not working with admin tab.","message":"Non-api Filters not working with admin tab.\n\nThe issue was with GET call for instances ,\ni.e the filtering opts like project_name, image_name,\nflavor_name are not supported by nova_api (non-api filters),\nso these get mapped to their id\u0027s first and then updated in\nsearch_opts dict which is passed for retrieving instances.\nFor the admin tab this call[1] was happening prior to\nnon-api filters mapping[2] , so in-order to work a second call\nwith updated search_opts dict should be called.\n\n[1] .https://opendev.org/openstack/horizon/src/branch/master/openstack_dashboard/dashboards/admin/instances/views.py#L154\n\n[2]. https://opendev.org/openstack/horizon/src/branch/master/openstack_dashboard/dashboards/admin/instances/views.py#L166\n\nChange-Id: Id43adbd44778d3375c0d49da3f7530cbb99e16fe\nCloses-Bug: #1888490\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/d3e2dd79de2d5942b31b619a7c9ddb82be53b12c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/d3e2dd79de2d5942b31b619a7c9ddb82be53b12c"}]},"branch":"refs/heads/master"},"5c56c1120389992430c0b4a487d1d2ee7a27b7aa":{"kind":"REWORK","_number":3,"created":"2020-10-09 06:48:15.000000000","uploader":{"_account_id":31916,"name":"vinay harsha mitta","email":"vinay.harsha@tcs.com","username":"vinay_m"},"ref":"refs/changes/58/754658/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/58/754658/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/58/754658/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/58/754658/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/58/754658/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/58/754658/3"}}},"commit":{"parents":[{"commit":"106e713ee553acce775120520ecaf70e4734981f","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/106e713ee553acce775120520ecaf70e4734981f"}]}],"author":{"name":"vinay_m","email":"vinay.harsha@tcs.com","date":"2020-09-28 08:52:18.000000000","tz":330},"committer":{"name":"vinay_m","email":"vinay.harsha@tcs.com","date":"2020-10-09 06:46:47.000000000","tz":330},"subject":"Non-api Filters not working with admin tab.","message":"Non-api Filters not working with admin tab.\n\nThe issue was with GET call for instances ,\ni.e the filtering opts like project_name, image_name,\nflavor_name are not supported by nova_api (non-api filters),\nso these get mapped to their id\u0027s first and then updated in\nsearch_opts dict which is passed for retrieving instances.\nFor the admin tab this call[1] was happening prior to\nnon-api filters mapping[2] , so in-order to work a second call\nwith updated search_opts dict should be called.\n\n[1] .https://opendev.org/openstack/horizon/src/branch/master/openstack_dashboard/dashboards/admin/instances/views.py#L154\n\n[2]. https://opendev.org/openstack/horizon/src/branch/master/openstack_dashboard/dashboards/admin/instances/views.py#L166\n\nChange-Id: Id43adbd44778d3375c0d49da3f7530cbb99e16fe\nCloses-Bug: #1888490\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/5c56c1120389992430c0b4a487d1d2ee7a27b7aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/5c56c1120389992430c0b4a487d1d2ee7a27b7aa"}]},"branch":"refs/heads/master"},"480340cb0d70a46f359d01002494edf584cc460f":{"kind":"REWORK","_number":4,"created":"2020-10-14 04:38:04.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/58/754658/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/58/754658/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/58/754658/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/58/754658/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/58/754658/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/58/754658/4"}}},"commit":{"parents":[{"commit":"106e713ee553acce775120520ecaf70e4734981f","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/106e713ee553acce775120520ecaf70e4734981f"}]}],"author":{"name":"vinay_m","email":"vinay.harsha@tcs.com","date":"2020-09-28 08:52:18.000000000","tz":330},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-10-14 03:59:37.000000000","tz":540},"subject":"Non-api Filters not working with admin tab.","message":"Non-api Filters not working with admin tab.\n\nThe issue was with GET call for instances ,\ni.e the filtering opts like project_name, image_name,\nflavor_name are not supported by nova_api (non-api filters),\nso these get mapped to their id\u0027s first and then updated in\nsearch_opts dict which is passed for retrieving instances.\nFor the admin tab this call[1] was happening prior to\nnon-api filters mapping[2] , so in-order to work a second call\nwith updated search_opts dict should be called.\n\n[1] .https://opendev.org/openstack/horizon/src/branch/master/openstack_dashboard/dashboards/admin/instances/views.py#L154\n\n[2]. https://opendev.org/openstack/horizon/src/branch/master/openstack_dashboard/dashboards/admin/instances/views.py#L166\n\nChange-Id: Id43adbd44778d3375c0d49da3f7530cbb99e16fe\nCloses-Bug: #1888490\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/480340cb0d70a46f359d01002494edf584cc460f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/480340cb0d70a46f359d01002494edf584cc460f"}]},"branch":"refs/heads/master"},"6b79cdcc1b92a968c967286cc267bac39803c85e":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2020-10-27 11:25:56.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/58/754658/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/58/754658/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/58/754658/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/58/754658/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/58/754658/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/58/754658/5"}}},"commit":{"parents":[{"commit":"106e713ee553acce775120520ecaf70e4734981f","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/106e713ee553acce775120520ecaf70e4734981f"}]}],"author":{"name":"vinay_m","email":"vinay.harsha@tcs.com","date":"2020-09-28 08:52:18.000000000","tz":330},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-10-27 11:07:57.000000000","tz":540},"subject":"Non-api filters not working with admin instance tab","message":"Non-api filters not working with admin instance tab\n\nThe issue was that the filtering opts like project_name, image_name,\nflavor_name that are not supported by nova API (non-api filters) do\nnot work expectedly. These filters are mapped to their IDs [1], but\nan instance list retrieved before resolving non-API filters is used [2].\nThis commit changes the logic to resolve non-API filters first and\nthen retrieve instances using the updated search_opts.\n\nNote that the image list is handled a bit specially. If \u0027image_name\u0027\nis specified as a filter, we retrieve a corresponding image detail\nfirst before fetching instances to resolve the image name into its ID.\nOtherwise, we retrieve images only related to instances retrieved\nand this happens after retrieving instances.\n\n[1] https://opendev.org/openstack/horizon/src/commit/f90c3cd50174af4927737e29e2765cd2d7ca507f/openstack_dashboard/dashboards/admin/instances/views.py#L166\n[2] https://opendev.org/openstack/horizon/src/commit/f90c3cd50174af4927737e29e2765cd2d7ca507f/openstack_dashboard/dashboards/admin/instances/views.py#L154\n\nCloses-Bug: #1888490\nCo-Authored-By: Akihiro Motoki \u003camotoki@gmail.com\u003e\nChange-Id: Id43adbd44778d3375c0d49da3f7530cbb99e16fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/6b79cdcc1b92a968c967286cc267bac39803c85e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/6b79cdcc1b92a968c967286cc267bac39803c85e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
