)]}'
{"id":"openstack%2Fhorizon~759652","triplet_id":"openstack%2Fhorizon~master~Ia249e83a7d40af873d6509c932b11cb9bb7a9497","project":"openstack/horizon","branch":"master","topic":"drop-neutron-legacy-usage","hashtags":[],"change_id":"Ia249e83a7d40af873d6509c932b11cb9bb7a9497","subject":"Drop the legacy way to retrieve neutron quotas and usages","status":"MERGED","created":"2020-10-26 08:36:50.000000000","updated":"2020-11-18 18:10:44.000000000","submitted":"2020-11-18 18:08:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"759652-1605722886585-0535b842","meta_rev_id":"58a47235bd594f38ad801b924c5890d86e3eadd0","_number":759652,"virtual_id_number":759652,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-18 18:08:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":2,"date":"2020-11-06 06:12:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2020-11-18 09:53:59.000000000","_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":1,"date":"2020-11-18 09:53:59.000000000","_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-06 06:12:00.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2020-11-18 09:53:59.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-11-18 18:08:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"99986088943a36455c3cabbd424a4d83a1c2f7fa","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-10-26 08:36:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1988194ca87a0d8053672322a19d207467de4b8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 10:23:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/a09e0d7ac2304adba4f4b2a7be136e36 : SUCCESS in 5m 31s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/b3adf1c8fdd447b2a8ea750187f0533d : SUCCESS in 7m 10s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/dc4aed73c42e41e682bbc2e94a818371 : SUCCESS in 4m 43s (non-voting)\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/b6102c9541c54e7889bfd1f261d135a6 : SUCCESS in 4m 54s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/57d29a47a4354b3c8640356c1f051c53 : SUCCESS in 12m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db33401b8aa944cd9a4aca2fb8664a1d : SUCCESS in 10m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7288ac75d60a4acf95658499c4dcdc84 : SUCCESS in 9m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b95dd4033186431aaacfe6c2cfbb3ad6 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e9fa291092448ed9a3f2490c0ad270a : SUCCESS in 10m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e3e9ab08a3843758b7058ab8215c59c : SUCCESS in 6m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a8968a7446240d1b5f31d1d52fe4054 : SUCCESS in 10m 58s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/1bf309f495144c8b99a3d8de23bd74ca : SUCCESS in 11m 45s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/16eb427029564eb195bd9efdd967760b : SUCCESS in 1h 35m 39s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/b395dcd1985f41698efaa2f3f8e63017 : SUCCESS in 31m 29s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/290f966ea3dc45629f1abdc7969e6f9a : SUCCESS in 4m 06s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/b02786eef5b445da9ede26b2817f61d8 : SUCCESS in 25m 40s","accounts_in_message":[],"_revision_number":1},{"id":"da93f1ecb0d89f4b866d16572b12f028b909af9d","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-11-06 06:12:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2d093ce33e3c75eb890d5611b06622572bb40046","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-11-18 09:53:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"534e9b1504a02e25d5c33be7c4ad130a6c24dd48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 09:54:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"37b06e40f1c6c9c0283712b81c960bb403d2099d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 11:38:20.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/a68f48aafb2c40b0ac7b0e05a3517074 : SUCCESS in 6m 36s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/cb7e05aa82dc4d0e98387488810cac88 : SUCCESS in 7m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05caf80e032f4c17951c01b4e01a979d : SUCCESS in 11m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cbd9484fd82c4bb9a9708606a1d00083 : SUCCESS in 10m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c55e0f38c070497f8317640b4647b062 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f91e377903884e1c94667c4a1984f00e : SUCCESS in 9m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e379482b19ef459193dbe779da1c1a05 : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/440ca476f25e407aa7bad123c9d82d84 : SUCCESS in 10m 12s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/2dd5a8f16f2548409dd9036fc33ffe7a : SUCCESS in 38m 59s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/c6a7be2ac0ea44ffb19328269eadd7e3 : SUCCESS in 8m 39s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/fb2b6f009294420b981874ef5eca7220 : FAILURE in 1h 35m 26s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/dc923cb117974862a51d782770a75867 : SUCCESS in 33m 04s","accounts_in_message":[],"_revision_number":1},{"id":"7d2eb6c12e78f880fef873422d3f40fda149d310","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-11-18 14:52:31.000000000","message":"Patch Set 1:\n\nrecheck.","accounts_in_message":[],"_revision_number":1},{"id":"6ba79386b7c69755d364346eb06de93dccbcd5a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 16:31:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/d178d01e305d408cade02c3612d098f7 : SUCCESS in 5m 49s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/c8fc169c1b0d47bda37cb043abef5bd7 : SUCCESS in 6m 41s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/1b1936b2b3bc425ea1ae3f9c3bacb4bf : SUCCESS in 6m 30s (non-voting)\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/3f7821a55616466a9e984fc57870c674 : SUCCESS in 5m 41s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/69fb0a59e4bb43f694390992bba116c6 : SUCCESS in 8m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79811d5f01074a039f24c84b4bb2435a : SUCCESS in 8m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/38652a17561d4960b1c8739a331e53b7 : SUCCESS in 10m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d934f4123f5a4d53ba41e22ec4be3f94 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb71583061664307a4f8d19de8a04584 : SUCCESS in 10m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/83a07c72b53343de865e304ca632d062 : SUCCESS in 11m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5e0342eff824299bafc6e25ee8d56ca : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aafdb1d5340a4259a5cc6cf6097c5646 : SUCCESS in 8m 54s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/fb4e958f0ae9499194419a514962585c : SUCCESS in 7m 50s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/be8aff25f9e14c81acd46838aa72b982 : SUCCESS in 1h 30m 06s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/74981b18bb1745888c04d3f63fd9daaa : SUCCESS in 47m 05s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/eb84c45382dd4f2b88eae2b9df1705ec : SUCCESS in 4m 11s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/3bff8d549ee744cbbf6b5a276238789c : SUCCESS in 39m 01s","accounts_in_message":[],"_revision_number":1},{"id":"ae4d123a06fca2e968ee3cf4a1e2094aa1eaffb5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 16:31:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3f9bbe7cc1436090db86291974dcd65a177a7771","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 18:08:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"263d0a75bc29f38eb6bb8a0f0fefe2d19cc7d92c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 18:08:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/4e8978700761427681c4da20a78fb5e8 : SUCCESS in 5m 47s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/82edcc70961a437887a337d7743bdc10 : SUCCESS in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d736a897a0044807b87573380ad5982e : SUCCESS in 10m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56b60c921feb45ffac8aed033ae872f5 : SUCCESS in 11m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cacad23455f1437f8b9c1818e3b9364a : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8b964e65b954016b1dcebe51d4b0495 : SUCCESS in 10m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa845b680449496186166458b73904d1 : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf072abe04404c0da0d4d85bea7ff6a9 : SUCCESS in 11m 06s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/f06437a6abb7417ebb5ed42b8f1be511 : SUCCESS in 44m 53s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/cc510bbb26c844ca8b2d8fc666d46cdc : SUCCESS in 6m 47s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/48e50cad026849318e90264f11fd77ac : SUCCESS in 1h 29m 38s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/4f05d7b6ae054a65a2f3ba039b15c840 : SUCCESS in 39m 47s","accounts_in_message":[],"_revision_number":1},{"id":"6d0ebb28c5c2e6772a264ac4801bb79b95bf56de","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 18:10:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d3ddab016a2479ab505c48060638797 : SUCCESS in 1m 24s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c52f848753e546b8915ab643109bd8a2 : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"22741aaa3eb36a98e5c599833655e0274f58e544","revisions":{"22741aaa3eb36a98e5c599833655e0274f58e544":{"kind":"REWORK","_number":1,"created":"2020-10-26 08:36:50.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/52/759652/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/52/759652/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/52/759652/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/52/759652/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/52/759652/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/52/759652/1"}}},"commit":{"parents":[{"commit":"b89c7051649a72df898ff0f5396e1f32303da8ed","subject":"Refactor quota related unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/b89c7051649a72df898ff0f5396e1f32303da8ed"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-10-26 07:53:55.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-10-26 08:25:56.000000000","tz":540},"subject":"Drop the legacy way to retrieve neutron quotas and usages","message":"Drop the legacy way to retrieve neutron quotas and usages\n\nquota_details neutron API extension was implemented since Pike\nand the enough migration period has passed. We have not received\nany negative feedback on the removal of the legacy way since\nits deprecation in Ussuri release.\n\nChange-Id: Ia249e83a7d40af873d6509c932b11cb9bb7a9497\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/22741aaa3eb36a98e5c599833655e0274f58e544"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/22741aaa3eb36a98e5c599833655e0274f58e544"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
