)]}'
{"id":"openstack%2Fhorizon~763551","triplet_id":"openstack%2Fhorizon~master~I4c2942960259782d47c6777394dec6888959e26c","project":"openstack/horizon","branch":"master","topic":"bug/1905024","hashtags":[],"change_id":"I4c2942960259782d47c6777394dec6888959e26c","subject":"Fix AttributeError in the project instance detail view","status":"MERGED","created":"2020-11-20 12:43:42.000000000","updated":"2020-11-23 00:29:37.000000000","submitted":"2020-11-23 00:28:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763551-bug/1905024","meta_rev_id":"137b9df75b300e8e40e4e7a6e3f588da1674b0a5","_number":763551,"virtual_id_number":763551,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-23 00:28:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"recommended":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"all":[{"value":1,"date":"2020-11-20 13:07:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":2,"date":"2020-11-20 12:45:09.000000000","_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":2,"date":"2020-11-20 14:17:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"all":[{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":1,"date":"2020-11-20 14:17:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-20 12:45:09.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-11-20 13:07:46.000000000","updated_by":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"reviewer":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"state":"REVIEWER"},{"updated":"2020-11-20 14:11:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-20 14:17:30.000000000","updated_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"REVIEWER"}],"messages":[{"id":"810b930d19f14e8bc3a40387ba181862e115ac18","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-11-20 12:43:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c033a4c7aefe150288012f050bcf5c0ed4e349eb","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-11-20 12:45:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"96c5d5b5a603b80ddb6b1dea652c36c5cb2b174c","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2020-11-20 13:07:46.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"71b24c244c82a93d0f4f0c38447695ae7f7b0ecc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 14:11:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/892506479dae4fb7a7cf27de4317db44 : SUCCESS in 6m 16s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/f39be8bf45df41d0bd1e2c6102deeca8 : SUCCESS in 7m 26s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/37108ec694c94a51bb34d52f99f55f06 : SUCCESS in 3m 53s (non-voting)\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/cacd99b4863343d08ee11dc957567d5c : SUCCESS in 5m 08s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/7e8603b866cd4bdea2d8cafff54329f4 : SUCCESS in 7m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a305fec4b894613b150f1f7fdf76606 : SUCCESS in 9m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/efd37bbced844cb79019afafef8d060a : SUCCESS in 10m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/922d8252f43b4129b5af39ee7b515390 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d49433a9a4564c8ab7c46518fcd7f46d : SUCCESS in 10m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5757162e764644fa8de25e493109d01d : SUCCESS in 11m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0b45b3d39b148eea799e82d0281b37d : SUCCESS in 7m 02s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/ad39b65956314c98a54dca471c7e81dd : SUCCESS in 6m 37s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/f3512a3adfbb415cb04f517dee2ac679 : FAILURE in 1h 15m 00s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/0985d2228dce486097edda6e42cdabbb : SUCCESS in 31m 49s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/abf18b325ce546fb8e47fcb252795259 : SUCCESS in 5m 47s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/3850505566a54616a41e7ac68d65328d : SUCCESS in 34m 28s","accounts_in_message":[],"_revision_number":1},{"id":"b33f1c5e7e07f40436245127a4286bc8b2d37278","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2020-11-20 14:17:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"69e4d1914ed4373fee1a00a7649f4c06c9d72a70","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-11-22 21:44:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c7aeb6d45dcad76a2ddf8ed2ad6654f05617c623","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-22 23:17:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/514b2cec8ef14331883cc5826860b61a : SUCCESS in 5m 10s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/7dee8c31dfff4eb382d761a084496ccb : SUCCESS in 6m 48s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/288d99f4de834c61b0a1db244bdd21ab : SUCCESS in 3m 19s (non-voting)\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/82d74bffd24c4e819cfce9662e8fe4f7 : SUCCESS in 5m 38s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/36019bb027724af4a2dbae614729df28 : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fddc9615e8354b7086b50d8c9d9a79f4 : SUCCESS in 9m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/346aa4fc81214cebb87fee5ce74266a5 : SUCCESS in 10m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b77e8011eb44282be09741c0bcd667f : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b30ebbb06fae4f998ee5e002cd408457 : SUCCESS in 10m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89e0ad6442a44443a1a223cfb3b91fdc : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e1a126a29364cfe87b8ddeacff199a5 : SUCCESS in 5m 24s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/d369de8289d4467e9de1b9a67094c084 : SUCCESS in 6m 24s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/7caa0ffcb7e94328b5a26c19e769ab16 : SUCCESS in 1h 27m 55s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/81816c8031cd429ebb613b203dc95f82 : SUCCESS in 39m 04s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/483a03107a2e41678427d49237e17ec3 : SUCCESS in 3m 42s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/2822fa0c697c40cfa9d709d9534be75c : SUCCESS in 38m 59s","accounts_in_message":[],"_revision_number":1},{"id":"9f9ad984cb0cf7181653d07390a98acc45f9d516","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-22 23:18:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a6c34c81e2a816a375e8fe012f775b77e85b8f15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 00:28:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/3bc91b10c33f4e138799d23a94a30c8e : SUCCESS in 5m 32s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/54c0f8274c064ae78cab1be0c1f38431 : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f7f6fbf54484abd81566ef8d3465df4 : SUCCESS in 9m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e62222874554b17a5f168dce519789e : SUCCESS in 10m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/395f4937498e448ba748807fdf474d59 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17c3899cb51049b4a52a6719a1489c89 : SUCCESS in 9m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfdddfb6eedf46729fea28fe2f6d3827 : SUCCESS in 9m 44s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/5cf7f417773d4d198172b5719e14ea8f : SUCCESS in 34m 04s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/ad7c5e951c4a4c88b6d0842b63329007 : SUCCESS in 9m 37s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/06d708f4d62a414687e6254b300943e2 : SUCCESS in 1h 09m 14s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/0dfc328567eb4c07bf4377051be152a9 : SUCCESS in 24m 04s","accounts_in_message":[],"_revision_number":1},{"id":"4b7cda9c450b6533d289e269918ee7f1ffea7739","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 00:28:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"137b9df75b300e8e40e4e7a6e3f588da1674b0a5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 00:29:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb98d0f72f804b3e8aa49e9d3c2bcc36 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a20a275ef5cae1b38ead1c3191dd96a042eb2070","revisions":{"a20a275ef5cae1b38ead1c3191dd96a042eb2070":{"kind":"REWORK","_number":1,"created":"2020-11-20 12:43:42.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/51/763551/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/51/763551/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/51/763551/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/51/763551/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/51/763551/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/51/763551/1"}}},"commit":{"parents":[{"commit":"d6ab5431ba609194d449fb00da878d03a9065322","subject":"Merge \"Drop the legacy way to retrieve neutron quotas and usages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/d6ab5431ba609194d449fb00da878d03a9065322"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-11-20 11:57:28.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-11-20 11:57:28.000000000","tz":540},"subject":"Fix AttributeError in the project instance detail view","message":"Fix AttributeError in the project instance detail view\n\nThe previous code does not handle a case where admin-specific attributes\ndo not exist. This leads to AttributeError in the project instance detail\npage as these attributes are not included in the response from nova API.\n\nChange-Id: I4c2942960259782d47c6777394dec6888959e26c\nCloses-Bug: #1905024\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/a20a275ef5cae1b38ead1c3191dd96a042eb2070"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/a20a275ef5cae1b38ead1c3191dd96a042eb2070"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
