)]}'
{"id":"openstack%2Fhorizon~765142","triplet_id":"openstack%2Fhorizon~master~Ibf04a2f8ab899e957a1a28d4b8193c5c6086b247","project":"openstack/horizon","branch":"master","topic":"drop-django-wsgi","hashtags":[],"change_id":"Ibf04a2f8ab899e957a1a28d4b8193c5c6086b247","subject":"Remove depretaced django.wsgi","status":"MERGED","created":"2020-12-02 14:09:45.000000000","updated":"2020-12-03 16:25:59.000000000","submitted":"2020-12-03 16:21:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"765142-drop-django-wsgi","meta_rev_id":"7d71e7864ee8fc6745bcb49dff7c29e0146aecd9","_number":765142,"virtual_id_number":765142,"owner":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-03 16:21:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":2,"date":"2020-12-03 13:44:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2020-12-03 12:49:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":1,"date":"2020-12-03 13:44:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-02 15:43:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-03 08:51:09.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2020-12-03 12:40:48.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"CC"},{"updated":"2020-12-03 12:49:04.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"}],"messages":[{"id":"798e09458e813e7795fc4d196962307b4b42b07a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-12-02 14:09:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ed2fabd935ff09d4416b777c9dc5672a9d8e90b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 15:43:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/2c8511fb7700460b8624bd6e7b5cddb3 : SUCCESS in 5m 38s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/578dda31578b4bf5b869df0e01b527d1 : SUCCESS in 8m 51s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/77e299244349491e9cb5eafd0ed8acaf : SUCCESS in 4m 07s (non-voting)\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/c9711625e1f54863a3901f82c960887c : SUCCESS in 5m 14s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/1835d07df3ec4389b1b315de7b978fb0 : SUCCESS in 9m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7292c43ee7d24c47b3042050054c9bb6 : SUCCESS in 9m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3fa54081bba48aa98764752770f019a : SUCCESS in 11m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f4703bea0bc46d39285c3172b2e7685 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3fa25e578a3944b2a05a62d5674d586e : SUCCESS in 10m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0495d326268248cab23cf681e9935447 : SUCCESS in 10m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/724a7306957f4e9a96c60011b4532faf : SUCCESS in 6m 30s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/cc23bb726eed48d9a7a44b18a2ad080b : SUCCESS in 7m 58s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/95f25a7bcaef4222843cf763ee3e025c : FAILURE in 1h 27m 33s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/f51a3aee7bee4b04a5a4549662ab194d : SUCCESS in 29m 35s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/48b2fc0c3d5c4677b4da37a953f85b66 : SUCCESS in 4m 11s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/90203e9f09614fc296d28d82b2cb0cbf : SUCCESS in 33m 10s","accounts_in_message":[],"_revision_number":1},{"id":"cfe505f4dca043f13859ad23eaed11492eaca311","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-12-02 16:45:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a8783a27e0f90e423a11575f0accab2ebe22e6f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 18:19:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/b5d8fe9001ea45c98c48386c36f5c687 : SUCCESS in 7m 08s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/cfc9a2ff139b44489f11160d740a1403 : SUCCESS in 10m 34s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/aa1f5fee7b4b409d938f25c656d30064 : SUCCESS in 8m 26s (non-voting)\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/5749c252313e4cdcb9dabb1c38516986 : SUCCESS in 9m 23s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/fe3c9c5646f04d77824daa27c819d0f2 : SUCCESS in 16m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dddc788f5e0148dfa6b27d533c673981 : SUCCESS in 13m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8eb3bc739376458496f8bbf77892c813 : SUCCESS in 10m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6dc7c6ac9bfa416888225e9fe094de6d : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5da28c016734c43995a0a3f7be587cf : SUCCESS in 10m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d25a2c991a140b9bcfad4b976b70230 : SUCCESS in 10m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f657bb7700cf4a49b14908df3233342a : SUCCESS in 8m 42s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/5f0fa4051cd84185a47ed6fb918b008b : SUCCESS in 7m 10s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/697d05632bad4ccca35d87bcfbba7738 : SUCCESS in 1h 30m 09s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/7940df245b3d41a1a850af4c63f94608 : SUCCESS in 33m 53s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/aba5571b4d754bf2999625e73759aa3c : SUCCESS in 5m 49s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/e0636cdec3b14ccbbea1b021312599bb : SUCCESS in 1h 02m 37s","accounts_in_message":[],"_revision_number":1},{"id":"405c3cd02990a6187e2db7f550dbcde156b430a5","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-12-03 08:51:09.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nDo we need to add a release note for this change.","accounts_in_message":[],"_revision_number":1},{"id":"1a33d5e1fb21adbcad03b3044651db748d085d10","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-12-03 09:18:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"50accb95ad9f999732712496d61315e5ca2454fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-12-03 09:19:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"69732963fd5261b7dbf4ab3529e3b39be8c7b1fc","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-12-03 09:25:34.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"61ab8d504b3c87b4457b810e65c7f74be24f467c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 11:04:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/b6402d13c16a47ffb8b387d485e8eb91 : SUCCESS in 6m 08s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/88455fafa8f3446f9f5b7b2e611e10fa : SUCCESS in 9m 30s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/8ec75c4df1f14cf58b38e7d2d29d9dc4 : SUCCESS in 4m 17s (non-voting)\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/416bcfe79dd14f09aa7a6cecb52ce199 : SUCCESS in 6m 10s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/b4a923c7cef34e68b59a813f6bc82c29 : SUCCESS in 8m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/480d95a3fb7b4a94bfab32c37c879577 : SUCCESS in 7m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0dcc2921ccd4f9eb1251f69c707a76d : SUCCESS in 25m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82db40b7d7bd482895dc8841069ab0aa : SUCCESS in 20m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6fd86aff22c6404d89f204a97bac09e6 : SUCCESS in 12m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90b75bc6d1d243f2a93302cf16312705 : SUCCESS in 15m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32957c8674d44178b77508068b87ff86 : SUCCESS in 16m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbfacda8bd92437e8c2f7b9e9dc8be5b : SUCCESS in 9m 35s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/6183831553414a5da6b448c3bd4667ad : SUCCESS in 6m 55s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/2a6f4f68f09f4638ba21b61d47b8176f : SUCCESS in 1h 33m 31s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/dd65ae9adc014e7bb02a9879c8f4894b : SUCCESS in 38m 52s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9d25e85ed54d40aaac985300d6066ba3 : SUCCESS in 4m 13s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/5d0e9ddf081b477a8243b0415fee7ce8 : SUCCESS in 28m 24s","accounts_in_message":[],"_revision_number":2},{"id":"8b01948990fdd761bd3ef44bb77ea19658ddfff1","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-12-03 12:40:48.000000000","message":"Patch Set 2:\n\n(2 comments)\n\nminor suggestions on the release note","accounts_in_message":[],"_revision_number":2},{"id":"442b47e7ce805a41c749053793378ed59e730ef5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-12-03 12:43:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2554830871846c0d2021bb7558701ca4604129ec","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-12-03 12:49:04.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"bdf067857b21e4d490d52505cbb0e3088faaad1b","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-12-03 13:44:05.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6ed0bc0a3801a65a4cebe4493730241019a674a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 14:41:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/081af983017d49a6b955d792e03218ca : SUCCESS in 6m 21s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/1fd2b519cec143ed8284e43680a0c3d5 : SUCCESS in 8m 48s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/d290a5decbaa4e2a9151a7564ca6582f : SUCCESS in 5m 01s (non-voting)\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/f4ecc571d34e454f8d435c15e735e6c3 : SUCCESS in 4m 40s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/9c4ed12e33fb4051aeec19df531eb4ae : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a329e92d62ce470b9d8d6c9c13477b2e : SUCCESS in 10m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c2a3ea9d23a847f88192ae38c0e99d38 : SUCCESS in 10m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f022f5d9187a4b8991547f86f8b2e9fe : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea1cf11e36bd478686a1a9fb2c6bbdf9 : SUCCESS in 10m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4bd5885b2cb84f4db624ad128a08d8ec : SUCCESS in 11m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3aa0a9343c4547c08b8bf5f5c1d0748d : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d64e20bea9ae403cbaa3a1aa6a12eae5 : SUCCESS in 9m 29s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/8d97dbb801aa46929d72485d9c34c0ec : SUCCESS in 6m 24s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/24e500258f484a21b902d78777a37e76 : SUCCESS in 1h 50m 26s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/3f8be3d58e724db59e8b3ac9064f03f4 : SUCCESS in 51m 12s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0b178a1ea95d4f458eaf8dab943ba3e0 : SUCCESS in 3m 59s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/b74206f4f04647e793418aa0b791f762 : SUCCESS in 40m 42s","accounts_in_message":[],"_revision_number":3},{"id":"4ec2cf73291909e5f39717547e319a4d5f3b8097","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 14:42:00.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"442df437b36091564547744ba6de32b09b018899","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 16:21:25.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/4ed4995a279a46cc84279518574f9d10 : SUCCESS in 5m 57s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/46c2d519245449bebbdc6218b3d2b07d : SUCCESS in 8m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bbfb0d8fe3264777b188b30e21d20a25 : SUCCESS in 9m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9bb50d7669a444e4ad2b2648d5a98302 : SUCCESS in 10m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f391539820d846edab31b59d72146070 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09b72e78f93047ce8f0af488bfe15f32 : SUCCESS in 12m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3a90b356f9746fa890cd5c99276aa87 : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23c682bdf23e49c7b97e12abf29c8dfa : SUCCESS in 10m 51s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/c486b48ae8734512a422232eeacc4474 : SUCCESS in 1h 00m 36s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/23e2d24e723748b5b238b7de314bb7ef : SUCCESS in 7m 20s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/f87f2394a3c741bbaf1e8c0b6ee83a56 : SUCCESS in 1h 30m 37s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/3556af7753bc461396426503c1c6b57d : SUCCESS in 40m 08s","accounts_in_message":[],"_revision_number":3},{"id":"b633ca08393698e8221da26072cf149647feabab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 16:21:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"7d71e7864ee8fc6745bcb49dff7c29e0146aecd9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 16:25:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7de4cd23bc7243d8bca2a2dfaddf0d56 : SUCCESS in 1m 21s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78d26b52840f415199bad3539d866ba9 : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7e682a2d528bcff992334d8792e33abbd57f5802","revisions":{"06810b0b89b71f2e0d7315b25b84c863f80eda98":{"kind":"REWORK","_number":1,"created":"2020-12-02 14:09:45.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/42/765142/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/42/765142/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/42/765142/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/42/765142/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/42/765142/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/42/765142/1"}}},"commit":{"parents":[{"commit":"4821fd0e1bdc66e383494fe276d89e6b68d5ce3e","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/4821fd0e1bdc66e383494fe276d89e6b68d5ce3e"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-12-02 14:09:24.000000000","tz":120},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-12-02 14:09:24.000000000","tz":120},"subject":"Remove depretaced django.wsgi","message":"Remove depretaced django.wsgi\n\nChange-Id: Ibf04a2f8ab899e957a1a28d4b8193c5c6086b247\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/06810b0b89b71f2e0d7315b25b84c863f80eda98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/06810b0b89b71f2e0d7315b25b84c863f80eda98"}]},"branch":"refs/heads/master"},"0c2600da11db9cae81542b6e951869fb679eb2cf":{"kind":"REWORK","_number":2,"created":"2020-12-03 09:19:14.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/42/765142/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/42/765142/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/42/765142/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/42/765142/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/42/765142/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/42/765142/2"}}},"commit":{"parents":[{"commit":"4821fd0e1bdc66e383494fe276d89e6b68d5ce3e","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/4821fd0e1bdc66e383494fe276d89e6b68d5ce3e"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-12-02 14:09:24.000000000","tz":120},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-12-03 09:19:02.000000000","tz":120},"subject":"Remove depretaced django.wsgi","message":"Remove depretaced django.wsgi\n\nChange-Id: Ibf04a2f8ab899e957a1a28d4b8193c5c6086b247\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/0c2600da11db9cae81542b6e951869fb679eb2cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/0c2600da11db9cae81542b6e951869fb679eb2cf"}]},"branch":"refs/heads/master"},"7e682a2d528bcff992334d8792e33abbd57f5802":{"kind":"REWORK","_number":3,"created":"2020-12-03 12:43:39.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/42/765142/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/42/765142/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/42/765142/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/42/765142/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/42/765142/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/42/765142/3"}}},"commit":{"parents":[{"commit":"4821fd0e1bdc66e383494fe276d89e6b68d5ce3e","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/4821fd0e1bdc66e383494fe276d89e6b68d5ce3e"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-12-02 14:09:24.000000000","tz":120},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-12-03 12:43:28.000000000","tz":120},"subject":"Remove depretaced django.wsgi","message":"Remove depretaced django.wsgi\n\nChange-Id: Ibf04a2f8ab899e957a1a28d4b8193c5c6086b247\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/7e682a2d528bcff992334d8792e33abbd57f5802"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/7e682a2d528bcff992334d8792e33abbd57f5802"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
