)]}'
{"id":"openstack%2Fhorizon~788813","triplet_id":"openstack%2Fhorizon~master~Ifec3bca94845b5d19c37b4c2844f1a983ad9365b","project":"openstack/horizon","branch":"master","topic":"bug/1920004","attention_set":{},"removed_from_attention_set":{"6914":{"account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"last_update":"2022-02-15 11:07:32.000000000","reason":"Change was submitted"},"8648":{"account":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"last_update":"2022-02-15 11:07:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifec3bca94845b5d19c37b4c2844f1a983ad9365b","subject":"Mark \"Source\" tab in the launch instance form as required","status":"MERGED","created":"2021-04-29 17:32:26.000000000","updated":"2022-02-15 11:11:01.000000000","submitted":"2022-02-15 11:07:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"788813-bug/1920004","meta_rev_id":"3f0dae8bf7bf2bd4904dafe7810e0b93fdc75b72","_number":788813,"virtual_id_number":788813,"owner":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-15 11:07:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":2,"date":"2022-02-15 11:11:01.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":1,"date":"2022-02-15 11:11:01.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":2,"date":"2022-02-15 09:21:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":1,"date":"2022-02-15 09:21:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"}],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-29 18:55:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-03 08:53:53.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2021-09-14 20:25:06.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"REVIEWER"},{"updated":"2021-09-14 20:25:06.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2021-09-14 20:25:06.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"}],"messages":[{"id":"7db5764019210074575b59b6bf0296dbd45f033a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2021-04-29 17:32:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"51160d7e8cff387e98514771d27dc64bf7c892b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 18:55:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/d809440392084a238762eaaa2dbc4cbc : SUCCESS in 6m 36s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/e8b110a462ee48e5b0ef24a280aacf23 : SUCCESS in 7m 31s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/3c94b0d3d1d94bbb88578e9a7d5f0e4a : SUCCESS in 5m 14s (non-voting)\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/c5e836a48a6141e69b9223bc140a9623 : SUCCESS in 5m 34s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/b4bd73c574ea4aeeb512981317b3cd50 : SUCCESS in 9m 21s\n- horizon-nodejs14-run-lint https://zuul.opendev.org/t/openstack/build/63652a31596549d7a1bd2c71067dbfef : SUCCESS in 5m 06s\n- horizon-nodejs14-run-test https://zuul.opendev.org/t/openstack/build/125d55a9dc3f43f2b49a0906208ddc1e : SUCCESS in 9m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d34dad15fbbc4efb8d8f5c5577b19160 : SUCCESS in 8m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b1dc6cc993b0426dbccd81797f93d595 : SUCCESS in 9m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a1637ce704448a79568ee8f2cd99bf8 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48643f492bed462c8f573eb017bd5da0 : SUCCESS in 9m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97dc7d0a891443da9babe99740e4a2da : SUCCESS in 13m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d0ec7bd65ca4cafad8f0d5b707a3d80 : SUCCESS in 6m 26s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/f70e64b4fa2e4963a113268489a5d12e : SUCCESS in 7m 36s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/af9673173b9e4604a6a8ad21bf44bc9f : SUCCESS in 1h 19m 06s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/8a4fb5a363b54560bcd8a2e29954f23d : SUCCESS in 26m 56s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c2e81216628943f4a3f7f449535cd6c9 : SUCCESS in 5m 13s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/0b65438d0ba747bb971c0342f2b392a5 : SUCCESS in 26m 10s","accounts_in_message":[],"_revision_number":1},{"id":"a145ed8941b2635782ce0536884fda12c906cb6d","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-05-03 08:53:53.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks, LGTM.","accounts_in_message":[],"_revision_number":1},{"id":"1961fff78d79197c5f97bcd842998fa18c33222d","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2021-10-01 07:26:06.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5bbdec12440a8a89e97e56d246f4282096058d1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2021-10-19 13:26:16.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"4c70b26f45ad9431bc6d299944ff481cfdbf2def","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 13:56:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/5d80445d19b140879fc3f0ae0715d7ee : SUCCESS in 6m 55s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/a8bbeb10e54242c995bcfc07d0a1ee58 : SUCCESS in 5m 21s (non-voting)\n- horizon-nodejs14-run-lint https://zuul.opendev.org/t/openstack/build/9cd75104638f4a8791683ac737f128db : SUCCESS in 5m 58s\n- horizon-nodejs14-run-test https://zuul.opendev.org/t/openstack/build/763aaf229d1a41d6973a5f8c791c47a2 : SUCCESS in 8m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d761ba5b8664eb0ad7e6fcfd988259f : SUCCESS in 13m 50s\n- horizon-tox-python3-django32 https://zuul.opendev.org/t/openstack/build/c6045651b33a47dd8ab1396cf9089d2d : SUCCESS in 12m 44s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5423f43fa99547149a5ee6a6e7282906 : SUCCESS in 12m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0467695381274e62b24e44106ce8b314 : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61e53e4b5bfe4bbf87d77dc38562a69f : SUCCESS in 11m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f29e9b1bfc5a4184b47c2621eeed93ca : SUCCESS in 11m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8273a23f9b9e404996babd90b73dda95 : SUCCESS in 7m 01s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/96e575fe409d40f0a0a66573a3e1c80a : FAILURE in 5m 43s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/d70d7c8db06e46e9b8c9037e6bf8a64d : FAILURE in 25m 22s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/e2a3630afae742169230b540194bfced : SUCCESS in 23m 48s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/840b318a7d6e48f78fe63c0fcee52e78 : SUCCESS in 5m 19s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/cb3e5d29499c4422a491ab561d81f244 : SUCCESS in 26m 52s","accounts_in_message":[],"_revision_number":2},{"id":"80c5e809f38bb06a2bba6d101901b7d4a7417d28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2021-10-27 16:12:54.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"5c7e394c197cc737d66a9e757dd80f6b287162cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 17:43:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/d6bcd30959e34f928f53ff6735cd36c1 : SUCCESS in 7m 01s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/207b8f4c8e49414999d2ece188b9bbb9 : SUCCESS in 5m 59s (non-voting)\n- horizon-nodejs14-run-lint https://zuul.opendev.org/t/openstack/build/ebb73725c86a4cbebf4bcf7fc60c585a : SUCCESS in 6m 56s\n- horizon-nodejs14-run-test https://zuul.opendev.org/t/openstack/build/aa468c2e55d949b7b71c1d42bbbaba54 : SUCCESS in 10m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16310f098238413683f4b11f1d7ee93b : SUCCESS in 11m 59s\n- horizon-tox-python3-django32 https://zuul.opendev.org/t/openstack/build/1ace5bdc14334a70af7eb5ab8e247139 : SUCCESS in 12m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dfc76fd6c54342708fcb6c245711847b : SUCCESS in 11m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/860bdcd9b9614ef4a75836bfe89f05c4 : SUCCESS in 7m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4cc5c01270db4258946171a2260a8b1f : SUCCESS in 11m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f39da92020f4bf6b07a8debae085e1c : SUCCESS in 11m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a9b5ef57305496f859523580990eb4d : SUCCESS in 8m 43s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/6cb9f192f97d4573b077a1eaa8123066 : SUCCESS in 9m 43s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/8de15804c3ea428c94bbacbfa144c958 : SUCCESS in 1h 24m 26s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/ef1a6407a66e461988e57a2c39a6027e : SUCCESS in 33m 44s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/83e86b3bec68480382a7022d582e7983 : SUCCESS in 5m 54s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/bc87c7ed7a3f4915b10ff86979763d89 : SUCCESS in 25m 52s","accounts_in_message":[],"_revision_number":3},{"id":"544a0589513fabad82aca05b7ec0a7e510cf8c74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2022-02-03 16:55:12.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"c7805982f4b3aa91f7434ccb7fc95c10c9895ecb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 18:45:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/59a6dacb6f5a492eb4cdefb79988b377 : SUCCESS in 7m 18s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/46ddfb757b274e5f893b735d0618601e : FAILURE in 6m 09s (non-voting)\n- horizon-nodejs14-run-lint https://zuul.opendev.org/t/openstack/build/818ad9d92035466fac14385ad4bfe1f5 : SUCCESS in 8m 08s\n- horizon-nodejs14-run-test https://zuul.opendev.org/t/openstack/build/af46a1b40aca4e72848fa2aef0749a67 : SUCCESS in 9m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a74698c9e524dbe85052eed41531c1e : SUCCESS in 11m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c5f324812b14e71989aee2d9dbd33cc : SUCCESS in 11m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73a501f2a1e94e59b0041ec075d442bc : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f9cce81818a446c80a7e88e43d5c122 : SUCCESS in 9m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fde96a87dc34c8c95bc3e44f8e80b05 : SUCCESS in 7m 26s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/eeebafedbab546f39ddf111978d1caa5 : SUCCESS in 8m 45s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/fb1050bad24444febb74cf59a958e6cc : SUCCESS in 1h 43m 46s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/221f9616759045b0aa99bffd35de44ae : SUCCESS in 31m 14s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8fc0ea11924746358d4b69f2d5a37370 : SUCCESS in 6m 05s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/c20c4ad380334d7bbc97b01c600322f0 : SUCCESS in 31m 23s","accounts_in_message":[],"_revision_number":4},{"id":"1c58a0460aca96cb8a0c5ad72da93596a6227c1a","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2022-02-15 09:21:49.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"662a0a35b6a5b63539509799b9507686d4efe792","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 09:22:47.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9901256522abf6acd9ea9856351ce4f70ddeb24c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 11:07:32.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- horizon-nodejs14-run-lint https://zuul.opendev.org/t/openstack/build/cf80a233a87144bdb7b0d520c02c758d : SUCCESS in 6m 35s\n- horizon-nodejs14-run-test https://zuul.opendev.org/t/openstack/build/6d2d170c0c334a809c496c2b2be87722 : SUCCESS in 9m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85031e0ca7f6446cabe27dbfae213e1b : SUCCESS in 10m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ae25c8de0254de1a1aa933399de227d : SUCCESS in 11m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dd97c212cc348dcbe4e113e716bdc05 : SUCCESS in 7m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a56ebb7993d4dc58293d13b302f7edd : SUCCESS in 11m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90c71675b98d422bb5c0000ef7004882 : SUCCESS in 7m 15s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/ffe161d82de648e7b9922425741c32d0 : SUCCESS in 30m 09s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/398690f62ece48dea37d089471cf2e2c : SUCCESS in 7m 52s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/88e8ffac50ea493f9fe8f3e7fb7fbde1 : SUCCESS in 1h 41m 19s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/406702646bdf4aea81a450c977b92a0a : SUCCESS in 33m 38s","accounts_in_message":[],"_revision_number":4},{"id":"2261a419e3d2295b48a1c48feaf0091e71481462","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 11:07:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"da365062a0215e625b066cbd8d194615196a3eb6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 11:11:01.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72d2b9c76a474797b54798f39e33d825 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b8470120857284e425097e9e309611de837a46f6","revisions":{"7e4878a1300b770096dd40c32d631ab074376fb6":{"kind":"REWORK","_number":1,"created":"2021-04-29 17:32:26.000000000","uploader":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"ref":"refs/changes/13/788813/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/13/788813/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/788813/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/788813/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/788813/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/13/788813/1"}}},"commit":{"parents":[{"commit":"00c5825ca7c597aeaab0279684befb07f0a13dde","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/00c5825ca7c597aeaab0279684befb07f0a13dde"}]}],"author":{"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","date":"2021-04-29 17:26:43.000000000","tz":-300},"committer":{"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","date":"2021-04-29 17:26:48.000000000","tz":-300},"subject":"Mark \"Source\" tab in the launch instance form as required","message":"Mark \"Source\" tab in the launch instance form as required\n\nThe boot source must be specified to launch instance. This patch adds\na validation for the allocated sources table length and marks \"Source\"\ntab with asterisk.\n\nCloses-Bug: #1920004\n\nChange-Id: Ifec3bca94845b5d19c37b4c2844f1a983ad9365b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/7e4878a1300b770096dd40c32d631ab074376fb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/7e4878a1300b770096dd40c32d631ab074376fb6"}]},"branch":"refs/heads/master"},"2fd780d7c203229466865249bbf955fd01ba4a77":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-10-19 13:26:16.000000000","uploader":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"ref":"refs/changes/13/788813/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/13/788813/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/788813/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/788813/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/788813/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/13/788813/2"}}},"commit":{"parents":[{"commit":"5e405d71926764b8aa60c75794b62f668f4e8122","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/5e405d71926764b8aa60c75794b62f668f4e8122"}]}],"author":{"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","date":"2021-04-29 17:26:43.000000000","tz":-300},"committer":{"name":"Radomir Dopieralski","email":"openstack@sheep.art.pl","date":"2021-10-19 13:26:16.000000000","tz":0},"subject":"Mark \"Source\" tab in the launch instance form as required","message":"Mark \"Source\" tab in the launch instance form as required\n\nThe boot source must be specified to launch instance. This patch adds\na validation for the allocated sources table length and marks \"Source\"\ntab with asterisk.\n\nCloses-Bug: #1920004\n\nChange-Id: Ifec3bca94845b5d19c37b4c2844f1a983ad9365b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/2fd780d7c203229466865249bbf955fd01ba4a77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/2fd780d7c203229466865249bbf955fd01ba4a77"}]},"branch":"refs/heads/master","description":"Rebase"},"7edd51f33dd010b19911fd427e58b9ec0c986f4b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-10-27 16:12:54.000000000","uploader":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"ref":"refs/changes/13/788813/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/13/788813/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/788813/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/788813/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/788813/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/13/788813/3"}}},"commit":{"parents":[{"commit":"9d1bb3626bc1dbcf29a55aeb094f4350067317cd","subject":"Allow both Django 2.2 and 3.2 for smooth transition","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/9d1bb3626bc1dbcf29a55aeb094f4350067317cd"}]}],"author":{"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","date":"2021-04-29 17:26:43.000000000","tz":-300},"committer":{"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","date":"2021-10-27 16:12:54.000000000","tz":0},"subject":"Mark \"Source\" tab in the launch instance form as required","message":"Mark \"Source\" tab in the launch instance form as required\n\nThe boot source must be specified to launch instance. This patch adds\na validation for the allocated sources table length and marks \"Source\"\ntab with asterisk.\n\nCloses-Bug: #1920004\n\nChange-Id: Ifec3bca94845b5d19c37b4c2844f1a983ad9365b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/7edd51f33dd010b19911fd427e58b9ec0c986f4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/7edd51f33dd010b19911fd427e58b9ec0c986f4b"}]},"branch":"refs/heads/master","description":"Rebase"},"b8470120857284e425097e9e309611de837a46f6":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-02-03 16:55:12.000000000","uploader":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"ref":"refs/changes/13/788813/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/13/788813/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/788813/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/788813/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/788813/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/13/788813/4"}}},"commit":{"parents":[{"commit":"a629893a1038ed0ed5bd7d42f5b23cc7099c63b4","subject":"Merge \"Modified the image page to use CSS class\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/a629893a1038ed0ed5bd7d42f5b23cc7099c63b4"}]}],"author":{"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","date":"2021-04-29 17:26:43.000000000","tz":-300},"committer":{"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","date":"2022-02-03 16:55:12.000000000","tz":0},"subject":"Mark \"Source\" tab in the launch instance form as required","message":"Mark \"Source\" tab in the launch instance form as required\n\nThe boot source must be specified to launch instance. This patch adds\na validation for the allocated sources table length and marks \"Source\"\ntab with asterisk.\n\nCloses-Bug: #1920004\n\nChange-Id: Ifec3bca94845b5d19c37b4c2844f1a983ad9365b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/b8470120857284e425097e9e309611de837a46f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/b8470120857284e425097e9e309611de837a46f6"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
