)]}'
{"id":"openstack%2Fhorizon~802476","triplet_id":"openstack%2Fhorizon~stable%2Fwallaby~Ic8330b947b1a733f70c3bdad8b3493f20a2f26fb","project":"openstack/horizon","branch":"stable/wallaby","hashtags":[],"change_id":"Ic8330b947b1a733f70c3bdad8b3493f20a2f26fb","subject":"Fix Unable to use multiattach volume as boot for new server","status":"MERGED","created":"2021-07-28 17:30:33.000000000","updated":"2021-07-29 12:45:18.000000000","submitted":"2021-07-29 12:43:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802476","cherry_pick_of_change":798730,"cherry_pick_of_patch_set":2,"meta_rev_id":"a0c3600f33b5eb70bea7bfe079a9946097e90b67","_number":802476,"virtual_id_number":802476,"owner":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-29 12:43:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"recommended":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":1,"date":"2021-07-29 08:07:46.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2021-07-29 11:15:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":2,"date":"2021-07-29 08:15:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":1,"date":"2021-07-29 11:15:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"}],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-28 17:30:33.000000000","updated_by":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2021-07-28 17:32:50.000000000","updated_by":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"REVIEWER"},{"updated":"2021-07-28 17:32:50.000000000","updated_by":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2021-07-28 17:32:50.000000000","updated_by":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2021-07-28 19:07:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7c1661843fb24172d3050c1545a46664b9615281","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2021-07-28 17:30:33.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1f133b18cbf59b9ab50ea4530c7dac257625a4ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 19:07:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/270c4da0e4a34f11aad9d7728157bb39 : SUCCESS in 7m 11s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/5eca9c04fb524fa494cd208ba424f6ad : SUCCESS in 8m 32s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/9e4991f34fd84258bf25a6fe0b38e1fb : SUCCESS in 5m 05s (non-voting)\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/292b14fe41364f7281ab113567abc882 : SUCCESS in 5m 35s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/8df27d2e768546139614a983bff154d5 : SUCCESS in 9m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7904ef29cc9b49af9508766d5301896e : SUCCESS in 11m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9bdc9a7ff9f49bd8ef588baca0c52ed : SUCCESS in 12m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f68453c076ea4e0989722230b8cbbd30 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0fc9c4dd132445eacc703cddb46f4ab : SUCCESS in 12m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b4a1a2e04654ee6b3c772deffe15525 : SUCCESS in 11m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/863d70d50d074412b15a88100d33143e : SUCCESS in 7m 07s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/81e358a450d546d293bbe343bfb65779 : SUCCESS in 7m 10s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/9b3e8672feaf479ea39817a7513dbff1 : SUCCESS in 1h 30m 45s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/8d7f37b729df474ea5e284489ceed4a0 : SUCCESS in 36m 53s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a77bf7165b3b46448da5be3e0df4b7fe : SUCCESS in 5m 13s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/14739d4d153446628b8db5f3c0eb5659 : SUCCESS in 34m 38s","accounts_in_message":[],"_revision_number":1},{"id":"6d76eaf0db640e05cef5bb41dfaf3be2ac41b820","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-07-29 08:07:46.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bd96aff9d38dfbf9c8b4253d674063ba8e8e9898","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-07-29 08:15:07.000000000","message":"Patch Set 1: Code-Review+2\n\nThe compute API 2.60 is available since Queens and 2.60 is already a part of the microversions, so it looks good.","accounts_in_message":[],"_revision_number":1},{"id":"857c05cf4a661018b15e4c2d838b8b196e5a740e","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2021-07-29 11:15:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"92d9b196fc674c2a0cfce223270e27a9c45ad6f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 11:15:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b992bc110ef38002729221b56a846b662e60f9ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 12:43:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/35704bd42aff4e178649983c2806add7 : SUCCESS in 7m 14s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/fe716bb0f6b246f2825877b570317a75 : SUCCESS in 14m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cbc90841471f4406a1fa56c5f3cd0ce8 : SUCCESS in 10m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96e2c268196947d892e9df401885eefb : SUCCESS in 10m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b5474a75eea4a11b92857305994f454 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db001ca65e5d4033b9d4591330debe14 : SUCCESS in 10m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba47623927d048749967188d021f0f3d : SUCCESS in 8m 31s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/8d607995e26e468c90788f40fa6ad131 : SUCCESS in 34m 05s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/791235cec8aa4d08aa3d2c61c5ad6ae3 : SUCCESS in 9m 06s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/3c74566741744da692239acbf1ada6da : SUCCESS in 1h 26m 54s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/7b1b390dc7c34af380c4b5a20883c109 : SUCCESS in 34m 14s","accounts_in_message":[],"_revision_number":1},{"id":"342fedc91ed38b779b254bfef023a67e32edb7d5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 12:43:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a0c3600f33b5eb70bea7bfe079a9946097e90b67","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 12:45:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d03c6e46f644137a51b9a5541091402 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"217dcc4b9ca2416b84116dfab3279faa3e04fc36","revisions":{"217dcc4b9ca2416b84116dfab3279faa3e04fc36":{"kind":"REWORK","_number":1,"created":"2021-07-28 17:30:33.000000000","uploader":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"ref":"refs/changes/76/802476/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/76/802476/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/76/802476/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/76/802476/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/76/802476/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/76/802476/1"}}},"commit":{"parents":[{"commit":"f8d3abd1fc28bf288aabd6f998c4d7462179f26f","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/f8d3abd1fc28bf288aabd6f998c4d7462179f26f"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-06-29 18:08:41.000000000","tz":330},"committer":{"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","date":"2021-07-28 17:30:33.000000000","tz":0},"subject":"Fix Unable to use multiattach volume as boot for new server","message":"Fix Unable to use multiattach volume as boot for new server\n\nIf we try to create a new server from a bootable volume that\nsupports multiattach, it will fail to create with an error\nmessage that ``multiattach volumes are only supported starting\nwith compute API version 2.60``. This patch fixes the issue.\n\nCloses-Bug: #1931440\nChange-Id: Ic8330b947b1a733f70c3bdad8b3493f20a2f26fb\n(cherry picked from commit 64fe0abb653950c85d455dedd09ef42856c6b07b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/217dcc4b9ca2416b84116dfab3279faa3e04fc36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/217dcc4b9ca2416b84116dfab3279faa3e04fc36"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
